)]}'
{"id":"openstack%2Fpuppet-tripleo~704373","triplet_id":"openstack%2Fpuppet-tripleo~master~I41ab9b3593bf3d078c5bbd1826df8308e3f5e7af","project":"openstack/puppet-tripleo","branch":"master","topic":"glance-multistore","hashtags":[],"change_id":"I41ab9b3593bf3d078c5bbd1826df8308e3f5e7af","subject":"Add support for glance multistore","status":"MERGED","created":"2020-01-27 17:43:07.000000000","updated":"2020-02-26 22:31:45.000000000","submitted":"2020-02-26 22:30:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"704373-1582756207293-b16e9fe7","meta_rev_id":"2f7d656272449b316d32cd8b3566483c36c3e38d","_number":704373,"virtual_id_number":704373,"owner":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},{"value":0,"date":"2020-02-25 08:49:32.000000000","_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"date":"2020-02-03 13:44:03.000000000","_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"value":0,"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-26 22:30:07.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-02-01 19:51:23.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":2,"date":"2020-02-26 17:34:41.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"value":2,"date":"2020-02-26 14:24:33.000000000","_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"date":"2020-02-20 22:11:01.000000000","_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"date":"2020-02-25 13:53:00.000000000","_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":1,"date":"2020-02-26 17:34:41.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"date":"2020-02-12 15:54:38.000000000","_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"value":0,"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-27 20:49:11.000000000","updated_by":{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},"reviewer":{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},"state":"REVIEWER"},{"updated":"2020-02-01 19:51:23.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-02-03 13:44:03.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2020-02-12 15:54:38.000000000","updated_by":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"reviewer":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"state":"REVIEWER"},{"updated":"2020-02-20 22:11:01.000000000","updated_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"reviewer":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"state":"REVIEWER"},{"updated":"2020-02-25 08:49:32.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2020-02-26 14:24:33.000000000","updated_by":{"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},"reviewer":{"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},"state":"REVIEWER"},{"updated":"2020-02-26 17:34:41.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2020-02-26 22:30:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9ab6a8bf9751243299e64663ca3e02178309834d","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2020-01-27 17:43:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"19b7ba7e8dbe9ca7808f99cdb05946ee443bbfe4","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-01-27 18:57:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 https://review.rdoproject.org/zuul/build/ddc6a8fff43947cd8ef6dabcc85a06c8 : FAILURE in 33m 06s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7e4762dbe1ebe13d3f0d4bec1ce7e7039b1df336","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-27 21:01:17.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0a2f3b73f8a94a9ba98a26c48c8e5714 : FAILURE in 5m 35s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/53bfb19f87d34703be10c17c672a2e6a : SUCCESS in 5m 50s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d409ee40b1764012bb8f01920d8834f3 : SUCCESS in 6m 00s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/1c480446fa364239b437df1cfe7e3e27 : SUCCESS in 44m 35s\n- puppet-openstack-unit-6.0-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2dafeef392ff40f0beccc57182244d23 : SUCCESS in 24m 29s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2b461b8d8cea45d2aa77991b78a8e6f4 : SUCCESS in 28m 23s (non-voting)\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/fa14262f5d6f406f9a79b96c9df5478f : SUCCESS in 1h 55m 41s\n- tripleo-ci-centos-7-containers-undercloud-minion https://zuul.opendev.org/t/openstack/build/820801cc12504aad9f59f47d834f2697 : SUCCESS in 1h 40m 27s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/41d9155b7ca04a9b868212e0a5fb9362 : FAILURE in 1h 23m 04s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/ba7435ef523547ef90165b74be488637 : SUCCESS in 3h 15m 01s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/6f044ece2e494389a2c28202e3fa91cf : SUCCESS in 2h 28m 17s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/7d5245f347d242f981bf784b9f1fa075 : SUCCESS in 2h 12m 43s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/e48e95e8730f4ca6af2591f84fe1ff2e : SUCCESS in 1h 59m 33s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/5bc7520951d443cca59ef220dfc24a85 : SUCCESS in 2h 14m 40s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/48dd01596e9c4911ab2c1f6f2f3fd7b9 : SUCCESS in 4m 15s\n- tripleo-ci-centos-7-scenario001-standalone https://zuul.opendev.org/t/openstack/build/1527f1df25c243678b86bffde19fc98a : FAILURE in 2h 03m 49s\n- tripleo-ci-centos-7-scenario002-standalone https://zuul.opendev.org/t/openstack/build/a5866d7ecb2447668e9fe71da15a21bb : SUCCESS in 1h 27m 29s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/eed8b8b611ab4ece8bf7e7ebe938c08f : SUCCESS in 1h 24m 38s\n- tripleo-ci-centos-7-scenario004-standalone https://zuul.opendev.org/t/openstack/build/3f341f25fed443208a6f6942d493def1 : SUCCESS in 1h 37m 01s\n- tripleo-ci-centos-7-scenario012-standalone https://zuul.opendev.org/t/openstack/build/e64d293c1f174a1998962db3faf45918 : FAILURE in 22m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f2e2e783ac1afe41520747aa6dbb2786c6e9288a","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2020-01-28 23:49:02.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d781cf1e7f4797b3db918a2be27a5f1d3756d88e","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-01-29 02:10:33.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 https://review.rdoproject.org/zuul/build/cdcaf6ed86c745119a974bc4aff41f09 : FAILURE in 2h 20m 42s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d0923d10dfa6e270ac506ac17e9eeeb4c4809c13","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-29 02:27:26.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a121a97344c44b1f8df19a60081ff4b7 : SUCCESS in 9m 08s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6bd6c1871aa94270b8d6b3af73fcd6e4 : SUCCESS in 7m 47s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3fb514d9b5fb47aca32bdcf0e6c4372e : SUCCESS in 5m 51s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/46f35cc5e2ad4702936d02e915e5d774 : SUCCESS in 39m 46s\n- puppet-openstack-unit-6.0-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/968d9384c12d4eb6bf1adf58d66f5dcc : SUCCESS in 20m 44s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3fd527210e3d463b849fbd572c6a57f6 : SUCCESS in 22m 03s (non-voting)\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/fb5acc9889464498a862869657983a39 : SUCCESS in 1h 41m 22s\n- tripleo-ci-centos-7-containers-undercloud-minion https://zuul.opendev.org/t/openstack/build/8cd8fa22752a4b85b610e2d8a6a54fe3 : SUCCESS in 1h 31m 03s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/cf1b59c0fffd44438d7bb1ba88cd540f : FAILURE in 1h 38m 04s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/86f76b2de23c4ed3ab8d1289339f6313 : FAILURE in 55m 56s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/de029c2c40fa42169f8000c4b57db32e : FAILURE in 53m 58s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/1bc16ccf367a44af99c32a6d1803472f : SUCCESS in 2h 22m 43s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/0e4ae71a4bd749df85e8b1e73c803436 : SUCCESS in 1h 39m 16s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/b66b7788b23e485ab3c8cbb413c0390f : SUCCESS in 2h 18m 38s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3cb18b97c31b4c788f8ba930a0fb4ef7 : SUCCESS in 3m 41s\n- tripleo-ci-centos-7-scenario001-standalone https://zuul.opendev.org/t/openstack/build/81cf1f5320aa40e98a451d3f9e9fe27e : FAILURE in 1h 46m 17s\n- tripleo-ci-centos-7-scenario002-standalone https://zuul.opendev.org/t/openstack/build/1d942b14bdd34fba974ffb0744bd06ef : SUCCESS in 1h 53m 59s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/98f1f1d92ef74c49a228167a30befdc0 : SUCCESS in 1h 31m 50s\n- tripleo-ci-centos-7-scenario004-standalone https://zuul.opendev.org/t/openstack/build/0bc11b3f429b4937aa6bb63c841ed77e : SUCCESS in 2h 02m 34s\n- tripleo-ci-centos-7-scenario012-standalone https://zuul.opendev.org/t/openstack/build/37b13778adca4ffa8f967ffc15f08fd1 : FAILURE in 27m 50s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f54c14dc9c65d022df06925215ce804342189fe5","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2020-01-29 03:04:07.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"902fe437318d824f397954d62347ae88a455c105","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-01-29 05:20:41.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 https://review.rdoproject.org/zuul/build/d7bc3aee0dca4d1bb242c0b757ee4005 : FAILURE in 2h 15m 39s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"1e4a2a2d2166cc0a849f42f8688687360e30a95b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-29 05:28:42.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/be6671b6ad984cd597da81f3a830963b : SUCCESS in 4m 40s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/597bba3653304e33ba1730ddd9f745bc : SUCCESS in 4m 23s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2f14cacfe9aa4675a8ebd858abe946c2 : SUCCESS in 5m 34s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/340b657a0cdf4940a2a92ee2bfd56e8a : SUCCESS in 30m 56s\n- puppet-openstack-unit-6.0-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/840167020d11473cbca0b74579fbae34 : SUCCESS in 19m 19s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/716a13e288204533941d06ab8c37f685 : SUCCESS in 19m 28s (non-voting)\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/7736faf15b3e4dc0a97bd574dd920253 : SUCCESS in 1h 34m 04s\n- tripleo-ci-centos-7-containers-undercloud-minion https://zuul.opendev.org/t/openstack/build/b9d4a58debea438d8a9b91d76ede934b : SUCCESS in 1h 15m 01s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/1e90eaacbe3f48c393a8e78c429b342a : FAILURE in 1h 24m 00s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/686fd6ffcee1471cb97d8dc108a38ace : SUCCESS in 2h 19m 59s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/c73ef93dd3424547a7da63f291e28304 : SUCCESS in 2h 21m 03s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/a2f31a35eb36420aa8cb9062864c5147 : SUCCESS in 1h 57m 35s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/f96487b50f674c86bf47aa60609b3565 : SUCCESS in 1h 24m 53s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/7817d52b584241298e3d33a14c511158 : FAILURE in 1h 50m 20s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c197f2b134b94423864f100de4a795c9 : SUCCESS in 3m 39s\n- tripleo-ci-centos-7-scenario001-standalone https://zuul.opendev.org/t/openstack/build/269049b047c64105aaae6c6028d7fbcb : FAILURE in 1h 32m 20s\n- tripleo-ci-centos-7-scenario002-standalone https://zuul.opendev.org/t/openstack/build/6df02eab781f4bc494a8b98f96adf638 : SUCCESS in 1h 32m 53s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/9e3aead2ebea46a399f2355466c9050e : SUCCESS in 1h 11m 46s\n- tripleo-ci-centos-7-scenario004-standalone https://zuul.opendev.org/t/openstack/build/6d23cd8194564665933f1919ef9a46d7 : SUCCESS in 1h 40m 41s\n- tripleo-ci-centos-7-scenario012-standalone https://zuul.opendev.org/t/openstack/build/b4d09025a1a8480781fc38a279f1d2da : FAILURE in 21m 40s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4817fdcf71abb3ec9c55db0684d5807499ab461a","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-01-30 00:25:39.000000000","message":"Patch Set 2:\n\nSeems that scenario001 is failing because of the following error in glance-api.\n\n~~~\n2020-01-29 04:23:22.224 67 ERROR glance.common.wsgi [req-0586d075-0830-4fba-a513-a3926fdc8c32 10e46d4ccc5b4af5b03e2c3638fd769d fea844d452d24b16a282739c69c8a39e - default default] Caught error: The image metadata key store has an invalid type of \u003ctype \u0027str\u0027\u003e. Only dict, list, and unicode are supported.: BackendException: The image metadata key store has an invalid type of \u003ctype \u0027str\u0027\u003e. Only dict, list, and unicode are supported.\n~~~","accounts_in_message":[],"_revision_number":2},{"id":"706a74347a5c8bce7828943d837c09e92892cc2a","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2020-01-30 00:30:58.000000000","message":"Patch Set 2: Workflow-1\n\nYes, I\u0027m looking at that now. The issue is the value of the \u0027store\u0027 metadata is somehow being converted from unicode to a py2 str. I\u0027m able to reproduce the problem using cli commands when creating an image from a nova server.","accounts_in_message":[],"_revision_number":2},{"id":"15d895e21ee0b392498c52ab893725eec28c9c92","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2020-01-31 15:07:02.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"cba2351d7056ed001b7b76a462125088e21c0fd5","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2020-01-31 15:07:39.000000000","message":"Patch Set 3:\n\nFailure traced to [1] and should be fixed by [2].\n\n[1] https://bugs.launchpad.net/glance/+bug/1861501\n[2] https://review.opendev.org/705229","accounts_in_message":[],"_revision_number":3},{"id":"454744841bf9a0c1b0a8dc7a8f46e88d9ba8159a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-31 18:12:59.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/802c087a75694aac98c4e5b555d6025e : SUCCESS in 5m 13s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5d498bcb158a4513b381e029cf2df11e : SUCCESS in 4m 12s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/529a95ef88734d88b8a66bbe8de8b0bc : SUCCESS in 5m 16s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/a27f79afbc2946c291db1ba751ab6f02 : POST_FAILURE in 37m 06s\n- puppet-openstack-unit-6.0-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3f8d6533586242d590c57496d6f9207e : SUCCESS in 22m 24s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f30076143d0e4749a18e17496ad532b6 : SUCCESS in 23m 52s (non-voting)\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/f02a2b629a064227854ee86edc3c63e3 : SUCCESS in 1h 51m 20s\n- tripleo-ci-centos-7-containers-undercloud-minion https://zuul.opendev.org/t/openstack/build/773a4abafb0b494ebd58160e57a1a3ee : SUCCESS in 1h 39m 39s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/72dbef90cfa842408b0f6a8b24a3e76e : FAILURE in 2h 45m 54s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/084c81c2e81f47608d2252bb9693dd14 : SUCCESS in 2h 53m 20s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/ce000322b673403e99a099f431511741 : FAILURE in 1h 54m 03s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/754fe74c2fee457b82ac1a203b56cda7 : FAILURE in 2h 35m 56s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/f2908b0455f3444591d3f3e88b1ef366 : SUCCESS in 1h 54m 47s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/594b4d0832ea4ac98895f68361648441 : FAILURE in 2h 00m 04s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b1ad326aab494473a6b363fcf15493c0 : SUCCESS in 5m 09s\n- tripleo-ci-centos-7-scenario001-standalone https://zuul.opendev.org/t/openstack/build/ca56501e6b174a8e81ca87df5c81e2e0 : SUCCESS in 1h 49m 35s\n- tripleo-ci-centos-7-scenario002-standalone https://zuul.opendev.org/t/openstack/build/2960181a09be4b5aa86fc778b1bbefed : SUCCESS in 2h 15m 35s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/a2dd29610c204080bc25f52212a240e8 : SUCCESS in 1h 29m 16s\n- tripleo-ci-centos-7-scenario004-standalone https://zuul.opendev.org/t/openstack/build/2e5e199e58204c49b7a37be532387862 : SUCCESS in 3h 00m 51s\n- tripleo-ci-centos-7-scenario012-standalone https://zuul.opendev.org/t/openstack/build/22eb665ff97045d2860c0b9d67474e18 : FAILURE in 35m 25s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"803ab2ee9f582e6c528ef6138278e36a8125d705","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-01-31 18:18:43.000000000","message":"Patch Set 3: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":3},{"id":"e521d75c58236b57d4922a852283f7cf28a0fc6b","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2020-02-01 15:43:42.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"c49444c5362b8d940bd8cfbe3701df0ce9b7df9d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-01 18:09:21.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/935dc00d790a4eaf9bf49cefa73d6907 : SUCCESS in 6m 00s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/84703f7d89594b10a3b7b4ead65ab91a : SUCCESS in 10m 01s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3222b500eb0e43d4bf43492865f64d56 : SUCCESS in 8m 02s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/c3d0866ce8a94e2c9921f42a0a1096cc : SUCCESS in 46m 04s\n- puppet-openstack-unit-6.0-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f213769be2004bac996175b1e23ffbad : SUCCESS in 20m 49s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bf4c30e1b2c745369ba877ead808b9b2 : SUCCESS in 21m 19s (non-voting)\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/2d100cb5203044c495642d1d88724f60 : SUCCESS in 1h 37m 09s\n- tripleo-ci-centos-7-containers-undercloud-minion https://zuul.opendev.org/t/openstack/build/e417400e481944639dbb1c2921f03942 : SUCCESS in 1h 19m 11s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/23dc9db5e1bb4f06901c4bb00ef05dfa : SUCCESS in 2h 03m 52s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/2c75e230732b464c92f6bf7664196c77 : SUCCESS in 2h 16m 23s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/06e2d68d8ce34228b6b29756a91955f8 : SUCCESS in 1h 47m 13s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/2cedf5b7f8fa4373be4aa1bcc746b1d9 : SUCCESS in 1h 57m 45s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/e39247f2e6a54b72959675b3b66c2784 : SUCCESS in 1h 22m 19s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/6539c2f76a2c47b1a7b4b5a9192002ed : SUCCESS in 2h 02m 37s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8d3042cf1e6640c788811216b43a56a2 : SUCCESS in 3m 20s\n- tripleo-ci-centos-7-scenario001-standalone https://zuul.opendev.org/t/openstack/build/0067cad9cecd4b92a67040b8c4545ab4 : FAILURE in 1h 32m 00s\n- tripleo-ci-centos-7-scenario002-standalone https://zuul.opendev.org/t/openstack/build/2445414e85c6430faa90ea519bfffc53 : SUCCESS in 1h 31m 45s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/68895c292bfc4e38be213aaf5fb15c0d : SUCCESS in 1h 13m 39s\n- tripleo-ci-centos-7-scenario004-standalone https://zuul.opendev.org/t/openstack/build/6ef85b8abaa94ff8aafd7652c4590040 : SUCCESS in 2h 24m 29s\n- tripleo-ci-centos-7-scenario012-standalone https://zuul.opendev.org/t/openstack/build/205230d319a049f4a702642d477166b6 : FAILURE in 33m 36s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"7f41e8fe7cc1811f2d164acad45652c30a2d31e0","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-02-01 19:51:23.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 https://review.rdoproject.org/zuul/build/34709597714a4743b59388e3c4d9c756 : SUCCESS in 3h 50m 57s (non-voting)\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 https://review.rdoproject.org/zuul/build/2af2f2dcd3874c53b8860115d0d7a689 : FAILURE in 2h 33m 58s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"2a371e148deb18cf31166187211a25f74b691a4f","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-02-03 13:44:03.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"421159962cda60d1e7f1749f0ac1175ea7283bfc","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2020-02-03 14:34:36.000000000","message":"Patch Set 3: Workflow-1\n\nThe scenario001 test passed [1] when the glance patch was in the pipeline. But now that it\u0027s merged, we need a new glance release that includes that patch. I\u0027ll ask the glance team for a new release.\n\n[1] https://zuul.opendev.org/t/openstack/build/ca56501e6b174a8e81ca87df5c81e2e0","accounts_in_message":[],"_revision_number":3},{"id":"f2f66ec3abfffa86ffdbde26ee1806335437c0c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-03 17:07:21.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/54aab85e59154cd0aadf10fd41816878 : SUCCESS in 5m 27s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d315ff5891a74d96a1f9b1ce08ae1b41 : SUCCESS in 5m 13s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b5e965597b694ea4a6598504105f6e84 : SUCCESS in 5m 51s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/df4336739dff4055bb09cd300f02cfb8 : SUCCESS in 39m 22s\n- puppet-openstack-unit-6.0-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/50c2accc73b14a44aa6340b13a81bd79 : SUCCESS in 22m 38s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2f1f908e3fcb4bcab9ca76e2f0b8a09e : SUCCESS in 23m 00s (non-voting)\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/faf856f51a234f6f8d1c494faf14ed0a : SUCCESS in 1h 49m 52s\n- tripleo-ci-centos-7-containers-undercloud-minion https://zuul.opendev.org/t/openstack/build/ebd1ce04e9db4322be241d84e9f9b16a : SUCCESS in 1h 40m 45s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/7029d18ac6c548d9bcb446b81814c054 : SUCCESS in 2h 57m 46s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/f8bf50661b3742febab45b79366ba4ea : SUCCESS in 3h 06m 48s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/3e14d8d060254fb98419df0db058081d : SUCCESS in 2h 37m 54s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/e56fbf8c855c4a8483205cf18ed1760c : SUCCESS in 2h 44m 05s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/d5848e51ad064e5ba8b8dee7e2e5991e : SUCCESS in 1h 35m 02s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/14f27409505a47098adc50e9f71d0d62 : SUCCESS in 2h 00m 31s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/20082fcd830045b9ada9aa2a91a9d29e : SUCCESS in 6m 09s\n- tripleo-ci-centos-7-scenario001-standalone https://zuul.opendev.org/t/openstack/build/29079fcc01314a61b6b97794213b8c84 : FAILURE in 1h 39m 07s\n- tripleo-ci-centos-7-scenario002-standalone https://zuul.opendev.org/t/openstack/build/298b5cf9cde746ed9f8ec9b3b597c468 : SUCCESS in 1h 33m 52s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/acd4506640684689b508e60fdd8c2bdb : SUCCESS in 1h 21m 36s\n- tripleo-ci-centos-7-scenario004-standalone https://zuul.opendev.org/t/openstack/build/261d78aaed05442aa01b7f49816f27ce : SUCCESS in 2h 44m 44s\n- tripleo-ci-centos-7-scenario012-standalone https://zuul.opendev.org/t/openstack/build/a27e397402784249bb65ec156f48563d : FAILURE in 46m 03s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"e497460c4e41e7aa79a5bcc3d343c24e7a0c4398","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-02-03 17:52:09.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 https://review.rdoproject.org/zuul/build/d073bbd8d18f465590b4a305bb805c21 : SUCCESS in 4h 06m 26s (non-voting)\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 https://review.rdoproject.org/zuul/build/4aa16222efb144e2b7cfa0482f189f96 : FAILURE in 2h 47m 23s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"ab42cadc0a30afb5012e7a865a1907994a1fe116","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-02-13 17:53:11.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"bbb11173ec49e450403ad9d453f47a1e445d4ece","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-02-13 21:21:31.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 https://review.rdoproject.org/zuul/build/f905fdd0f4e34d1eaf6af635b6611cdb : SUCCESS in 3h 27m 00s (non-voting)\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 https://review.rdoproject.org/zuul/build/0bcc1f2c07b845d098d44d39d0ff92a9 : FAILURE in 3h 04m 42s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"c8295d1e499782914e7456c1d44afd1faf92a1a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-13 21:26:01.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2c1b069155884a6c8682d0d5d2fbf862 : SUCCESS in 6m 50s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c3be0c1ede044a5c8a96bba3f2a795fb : SUCCESS in 5m 57s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fecccfde228d40e6ab0a58768ea2b1b5 : SUCCESS in 7m 09s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/47d152967b984baa95199762bf035b0b : SUCCESS in 44m 38s\n- puppet-openstack-unit-6.0-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ebd4b6cd771d44ac87d78973ef476d34 : SUCCESS in 24m 14s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b19b902510be4dafac6e77ec9d77e983 : SUCCESS in 23m 46s (non-voting)\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/bdcc2ad114c64415806805bf4a777036 : SUCCESS in 1h 58m 01s\n- tripleo-ci-centos-7-containers-undercloud-minion https://zuul.opendev.org/t/openstack/build/38bca3d846744d8d9b2ddc54671fc59f : SUCCESS in 1h 39m 55s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/70caea1239744fbb8a03eef4c5c3177d : SUCCESS in 2h 19m 42s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/0a9a582b67594c94b0cf55407fa626b5 : SUCCESS in 2h 55m 02s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/19e10881d739451d9b43f524dd08700e : SUCCESS in 2h 09m 16s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/54f4ecefaea740bc863e420358d07452 : SUCCESS in 2h 15m 16s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/b42eaf35c6a8486ea580d0272c90d59b : SUCCESS in 1h 48m 04s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/75b0139b2b1f46bea66b17923cc7d7f2 : SUCCESS in 2h 04m 51s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8991f36accca40bbb6ff07f39acfd96d : SUCCESS in 5m 01s\n- tripleo-ci-centos-7-scenario001-standalone https://zuul.opendev.org/t/openstack/build/187b44ffe8784d2b92b6e86257af727f : FAILURE in 2h 14m 02s\n- tripleo-ci-centos-7-scenario002-standalone https://zuul.opendev.org/t/openstack/build/d053511ea7b54c04b7215a408becfeb1 : SUCCESS in 1h 42m 53s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/e8e64d6dad014308841e2c3803cb73d6 : SUCCESS in 1h 17m 39s\n- tripleo-ci-centos-7-scenario004-standalone https://zuul.opendev.org/t/openstack/build/3b4ae259e7024457bbd25c63dc1c77a7 : SUCCESS in 1h 49m 10s\n- tripleo-ci-centos-7-scenario012-standalone https://zuul.opendev.org/t/openstack/build/48f0e261811641f4baa561db54155adf : FAILURE in 1h 15m 29s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"21d9a9620c5d93bd6d00570c81ff27fb5a298632","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2020-02-13 23:53:20.000000000","message":"Patch Set 3:\n\nThe tripleo container image for glance-api hasn\u0027t been rebuilt with the ussuri m2 code yet (see [1]). I\u0027ll monitor the situation, and recheck when I know there\u0027s a fresh glance-api image.\n\n[1] https://hub.docker.com/r/tripleomaster/centos-binary-glance-api/tags","accounts_in_message":[],"_revision_number":3},{"id":"991b6c033dbb4ccf626a90d0d5efd1b55e2d19ff","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-02-25 08:49:32.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"6f6a1c1d09bb12238164419d066834a8877311e3","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-02-25 11:48:43.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 https://review.rdoproject.org/zuul/build/91437711a0a94c7ba0871567e523ecd5 : SUCCESS in 2h 58m 14s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"052fd05fceb917da6ecf2aede1176e224a2152ee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-25 12:37:58.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8dc7998b176549758c375438a01eca95 : SUCCESS in 6m 42s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/033478d44ac745beaeb41c4c0663499d : SUCCESS in 5m 56s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/485d6f8b6c8849fda7daf94ea77bb257 : SUCCESS in 7m 21s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/a0827fceaaf04d8e91cb1463f815c0c7 : SUCCESS in 36m 33s\n- puppet-openstack-unit-6.0-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/daeb8db9c26b4529a49d72ab91ab1ab3 : SUCCESS in 24m 21s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6e577feeda274b498485ae9bc95195d3 : SUCCESS in 24m 14s (non-voting)\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/0b13a5594dbe49959e1783172db3b24e : SUCCESS in 1h 51m 22s\n- tripleo-ci-centos-7-containers-undercloud-minion https://zuul.opendev.org/t/openstack/build/e3a5b33365cc49edb8281f6c3d1fcea4 : SUCCESS in 1h 53m 02s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/f0018f29d9244e82917026a88a546ce2 : SUCCESS in 2h 55m 58s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/0db614f18d6c4bf58d7ef3fb79a96d13 : SUCCESS in 2h 36m 50s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/0e6271ca8fa045d29d8bb46474de8b7f : SUCCESS in 2h 13m 34s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/30ee89ff75324faf8ee0934ff6bacc61 : FAILURE in 2h 24m 23s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/60cf10f98eca43b0bf30cb2c0f67edb6 : SUCCESS in 1h 32m 51s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/90160653f9fb49509ea9dbf4637e2f33 : SUCCESS in 2h 38m 08s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dc132219de3a446f9ca4ffe057b0395a : SUCCESS in 6m 35s\n- tripleo-ci-centos-7-scenario001-standalone https://zuul.opendev.org/t/openstack/build/8ba53f61db9a46ffbfb8acf7a7b8a763 : SUCCESS in 2h 15m 23s\n- tripleo-ci-centos-7-scenario002-standalone https://zuul.opendev.org/t/openstack/build/00171a283365480a9d8fb82bf9de0e75 : SUCCESS in 1h 54m 46s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/ecbbcb61470f4c6ea8febf1799ecf454 : SUCCESS in 1h 29m 07s\n- tripleo-ci-centos-7-scenario004-standalone https://zuul.opendev.org/t/openstack/build/bfb74097988649909c2dbc07b9252048 : SUCCESS in 1h 40m 47s\n- tripleo-ci-centos-7-scenario012-standalone https://zuul.opendev.org/t/openstack/build/3eaf449c8784458c80755884ceadb9b3 : FAILURE in 1h 16m 43s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"a4ff6a2b72232bca355d9f210aa421294a1d75a2","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2020-02-25 13:53:00.000000000","message":"Patch Set 3: -Workflow\n\nThanks for your patience, the patch is ready for reviews now.","accounts_in_message":[],"_revision_number":3},{"id":"3c34d280231ac551b4bde47931a057e9214d87ea","author":{"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},"date":"2020-02-26 14:24:33.000000000","message":"Patch Set 3: Code-Review+2\n\nthanks!","accounts_in_message":[],"_revision_number":3},{"id":"787be886bf0daad2bacfd48fbbdbe655cd91c998","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-02-26 17:34:41.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"e64971948595863391e34c390b37184d35c54f9b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-26 17:34:58.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"bed81fe95be263614208f094960f46498afbaf76","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-26 22:30:07.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2f11c54c1524405f968f8c9ccd1aa306 : SUCCESS in 6m 26s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e8417df6ad354bac8140492aaffa23b5 : SUCCESS in 5m 55s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a460973e753247faa8d70088a84c969e : SUCCESS in 6m 24s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/8b1f4e80280d48469f191cde29f3702a : SUCCESS in 36m 32s\n- puppet-openstack-unit-6.0-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f175befe493e4afdb2fad815a3ee9bb9 : SUCCESS in 20m 43s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/c0deabbe5b3642cb85963920ae73087c : SUCCESS in 1h 45m 53s\n- tripleo-ci-centos-7-containers-undercloud-minion https://zuul.opendev.org/t/openstack/build/dd17c7bac4224599a1f49ccb9bb5da46 : SUCCESS in 2h 04m 25s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/c40cf01cc2c545a28bfe40d7288aec6a : SUCCESS in 2h 29m 47s\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/36ba9f4e74e44674a30af76dfa41c847 : SUCCESS in 1h 43m 59s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/1899413b85ab4a6db6f1965d116875e5 : SUCCESS in 2h 25m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/53e813fefe264aeab66f7be929d37ca0 : SUCCESS in 5m 41s\n- tripleo-ci-centos-7-scenario001-standalone https://zuul.opendev.org/t/openstack/build/e0ccafd385374bbe899000f5a75c8857 : SUCCESS in 2h 07m 10s\n- tripleo-ci-centos-7-scenario002-standalone https://zuul.opendev.org/t/openstack/build/ef86e74e1c6b44fcbad39b7bef6393ff : SUCCESS in 1h 47m 43s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/f013ec885bcb44e2b407dbd36da24d0b : SUCCESS in 1h 36m 39s\n- tripleo-ci-centos-7-scenario004-standalone https://zuul.opendev.org/t/openstack/build/033e04e4503141b59f3ce7f4fa01258d : SUCCESS in 1h 52m 43s","accounts_in_message":[],"_revision_number":3},{"id":"96a53e7b31b2123616c6f93f9c705d9ce4e96a9f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-26 22:30:07.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"5dbaa0ad4c26a12978d17ca498037fee284c5bdc","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-26 22:31:45.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c309cf01c1b14d4ab799155409d590bf : SUCCESS in 1m 15s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"c7b9b90dbdeef35cc16b1c0b2007152c1a21b613","revisions":{"3c08d46e6b6d312c5f2970256a5594b6b620e8e3":{"kind":"REWORK","_number":1,"created":"2020-01-27 17:43:07.000000000","uploader":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"ref":"refs/changes/73/704373/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tripleo","ref":"refs/changes/73/704373/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/73/704373/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/73/704373/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/73/704373/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tripleo refs/changes/73/704373/1"}}},"commit":{"parents":[{"commit":"8c0d43a193d60214d8615a88334c94ee683ff13a","subject":"Merge \"Make the bundle user configurable via hiera\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/8c0d43a193d60214d8615a88334c94ee683ff13a"}]}],"author":{"name":"Alan Bishop","email":"abishop@redhat.com","date":"2020-01-20 14:09:59.000000000","tz":-480},"committer":{"name":"Alan Bishop","email":"abishop@redhat.com","date":"2020-01-27 15:31:13.000000000","tz":-480},"subject":"Add support for glance multistore","message":"Add support for glance multistore\n\nAdd new tripleo::profile::base::glance::api::multistore_config parameter\nto support configuring multiple glance-api backends. The parameter is\noptional, and represents a hash of settings for each additional backend.\nThe existing \u0027glance_backend\u0027 parameter specifies the default backend.\n\nIn order to support DCN/Edge deployments, the syntax supports multiple\ninstances of the \u0027rbd\u0027 backend type. Restrictions are imposed to allow\nonly a single instance of the \u0027cinder\u0027, \u0027file\u0027 and \u0027swift\u0027 backend types.\n\nChange-Id: I41ab9b3593bf3d078c5bbd1826df8308e3f5e7af\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/3c08d46e6b6d312c5f2970256a5594b6b620e8e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/3c08d46e6b6d312c5f2970256a5594b6b620e8e3"}]},"branch":"refs/heads/master"},"c56fa601c8ad197f63befa56f0b32ef1b6666218":{"kind":"REWORK","_number":2,"created":"2020-01-28 23:49:02.000000000","uploader":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"ref":"refs/changes/73/704373/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tripleo","ref":"refs/changes/73/704373/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/73/704373/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/73/704373/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/73/704373/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tripleo refs/changes/73/704373/2"}}},"commit":{"parents":[{"commit":"8c0d43a193d60214d8615a88334c94ee683ff13a","subject":"Merge \"Make the bundle user configurable via hiera\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/8c0d43a193d60214d8615a88334c94ee683ff13a"}]}],"author":{"name":"Alan Bishop","email":"abishop@redhat.com","date":"2020-01-20 14:09:59.000000000","tz":-480},"committer":{"name":"Alan Bishop","email":"abishop@redhat.com","date":"2020-01-28 23:48:41.000000000","tz":-480},"subject":"Add support for glance multistore","message":"Add support for glance multistore\n\nAdd new tripleo::profile::base::glance::api::multistore_config parameter\nto support configuring multiple glance-api backends. The parameter is\noptional, and represents a hash of settings for each additional backend.\nThe existing \u0027glance_backend\u0027 parameter specifies the default backend.\n\nIn order to support DCN/Edge deployments, the syntax supports multiple\ninstances of the \u0027rbd\u0027 backend type. Restrictions are imposed to allow\nonly a single instance of the \u0027cinder\u0027, \u0027file\u0027 and \u0027swift\u0027 backend types.\n\nChange-Id: I41ab9b3593bf3d078c5bbd1826df8308e3f5e7af\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/c56fa601c8ad197f63befa56f0b32ef1b6666218"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/c56fa601c8ad197f63befa56f0b32ef1b6666218"}]},"branch":"refs/heads/master"},"c7b9b90dbdeef35cc16b1c0b2007152c1a21b613":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2020-01-31 15:07:02.000000000","uploader":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"ref":"refs/changes/73/704373/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tripleo","ref":"refs/changes/73/704373/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/73/704373/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/73/704373/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/73/704373/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tripleo refs/changes/73/704373/3"}}},"commit":{"parents":[{"commit":"8c0d43a193d60214d8615a88334c94ee683ff13a","subject":"Merge \"Make the bundle user configurable via hiera\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/8c0d43a193d60214d8615a88334c94ee683ff13a"}]}],"author":{"name":"Alan Bishop","email":"abishop@redhat.com","date":"2020-01-20 14:09:59.000000000","tz":-480},"committer":{"name":"Alan Bishop","email":"abishop@redhat.com","date":"2020-01-31 15:06:16.000000000","tz":-480},"subject":"Add support for glance multistore","message":"Add support for glance multistore\n\nAdd new tripleo::profile::base::glance::api::multistore_config parameter\nto support configuring multiple glance-api backends. The parameter is\noptional, and represents a hash of settings for each additional backend.\nThe existing \u0027glance_backend\u0027 parameter specifies the default backend.\n\nIn order to support DCN/Edge deployments, the syntax supports multiple\ninstances of the \u0027rbd\u0027 backend type. Restrictions are imposed to allow\nonly a single instance of the \u0027cinder\u0027, \u0027file\u0027 and \u0027swift\u0027 backend types.\n\nChange-Id: I41ab9b3593bf3d078c5bbd1826df8308e3f5e7af\nDepends-On: I5a1c61430879a910e7b6c79effba538431959d56\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/c7b9b90dbdeef35cc16b1c0b2007152c1a21b613"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/c7b9b90dbdeef35cc16b1c0b2007152c1a21b613"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
