)]}'
{"id":"openstack%2Fpuppet-tripleo~734130","triplet_id":"openstack%2Fpuppet-tripleo~stable%2Ftrain~I83b4d400947da16229c565c5311ca033b4c76d73","project":"openstack/puppet-tripleo","branch":"stable/train","topic":"bug/1875176","hashtags":[],"change_id":"I83b4d400947da16229c565c5311ca033b4c76d73","subject":"Support for Dell EMC VXFlexOS Backend","status":"MERGED","created":"2020-06-08 15:39:23.000000000","updated":"2020-06-11 04:38:47.000000000","submitted":"2020-06-11 04:37:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"734130-1591850260715-2acfc7a0","meta_rev_id":"efd63ae0db5b4bc4a3f30d499c2327184031c81b","_number":734130,"virtual_id_number":734130,"owner":{"_account_id":10379,"name":"Rajini Karthik","email":"rrajini@gmail.com","username":"rajinir"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-11 04:37:40.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"all":[{"value":2,"date":"2020-06-10 22:32:10.000000000","_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"all":[{"value":1,"date":"2020-06-10 22:32:10.000000000","_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-10 22:32:10.000000000","updated_by":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"reviewer":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"state":"REVIEWER"},{"updated":"2020-06-11 04:37:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"62b564323694e2778913ec01913ddad337ed98de","author":{"_account_id":10379,"name":"Rajini Karthik","email":"rrajini@gmail.com","username":"rajinir"},"date":"2020-06-08 15:39:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a077f0c015d85f75d8953314f2927c3d472892d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-08 19:04:02.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9f660261ae8b412397dff90c4d41a876 : SUCCESS in 6m 17s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5c3fdf86b388459d94215e87c811e973 : SUCCESS in 4m 52s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/45b752d63e82416cb63ff9d149345322 : SUCCESS in 6m 15s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/2e9bf6fcad7a4e908ebac936e26f333f : FAILURE in 46m 21s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/dea72649e47f4a22a00b8412a076944a : FAILURE in 25m 06s (non-voting)\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/c91299dc5d624a0b922b5b3bb64bfde8 : SUCCESS in 1h 26m 55s\n- tripleo-ci-centos-7-containers-undercloud-minion https://zuul.opendev.org/t/openstack/build/6da0b3e55e774d439a2aa735792258c9 : SUCCESS in 1h 12m 59s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/f7390afcf3764432ae331d9478cfc48a : FAILURE in 2h 28m 02s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/569174d2798e4e38ba13ddb9966630f0 : SUCCESS in 3h 06m 01s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/fa26497e4ef048e99aec53bc37781468 : SUCCESS in 2h 01m 02s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/8ea5ff211d984d61815d66e4115e8b28 : FAILURE in 35m 53s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/42ae2b682a9b403f911242f17c9db15e : SUCCESS in 1h 46m 06s\n- tripleo-ci-centos-7-standalone-upgrade-train https://zuul.opendev.org/t/openstack/build/a5f4507e49da4271861804e02c3d8a14 : FAILURE in 52m 30s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d0f962ca7ade40cebd38fbee95a59ab1 : SUCCESS in 5m 10s\n- tripleo-ci-centos-7-scenario001-standalone https://zuul.opendev.org/t/openstack/build/279c717f101845a0990ef95618ee46e1 : SUCCESS in 1h 47m 51s\n- tripleo-ci-centos-7-scenario002-standalone https://zuul.opendev.org/t/openstack/build/a57454857c514cc5943a6a7c43886161 : SUCCESS in 1h 35m 51s","accounts_in_message":[],"_revision_number":1},{"id":"62c0a6ebca745a2e8c186ba025c9688bf2587365","author":{"_account_id":10379,"name":"Rajini Karthik","email":"rrajini@gmail.com","username":"rajinir"},"date":"2020-06-09 21:10:54.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"29ae7adbf8fe37e3b66c5927e621785bc80560db","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-09 23:58:09.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/878fcc6821214d689582e8a431214e04 : SUCCESS in 5m 32s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/31946bba07f84db78232f8084f71ab94 : SUCCESS in 4m 38s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9afce30e880f41549424fd5c66188b73 : SUCCESS in 4m 48s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/ab4256ac2de748c08a65cd4ba5f5e015 : FAILURE in 43m 49s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cfe45f7bf8da47d29d106e76b94b9c5e : FAILURE in 20m 00s (non-voting)\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/9f64cd2803c1463a80de21abc58482e6 : SUCCESS in 1h 52m 42s\n- tripleo-ci-centos-7-containers-undercloud-minion https://zuul.opendev.org/t/openstack/build/2080ccd1f6e24d6082f5ad1dbdc4b682 : SUCCESS in 1h 11m 59s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/4942ee22389b43ccbbddb2fa09a7dc35 : FAILURE in 1h 35m 20s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/6636396a8c624cd1bae73bc0299b7f8e : SUCCESS in 2h 42m 04s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/62fa27f2493345938ea70053478582b6 : SUCCESS in 2h 14m 10s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/220e56220fc44ee598bd607795f68c84 : SUCCESS in 2h 10m 34s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/d0345e484aea476daea4a11410b12187 : SUCCESS in 1h 45m 47s\n- tripleo-ci-centos-7-standalone-upgrade-train https://zuul.opendev.org/t/openstack/build/a0531772fb1e46d29b169edbd7712294 : SUCCESS in 2h 26m 11s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/eeaad0937f784e96b6fe970024710daf : SUCCESS in 3m 17s\n- tripleo-ci-centos-7-scenario001-standalone https://zuul.opendev.org/t/openstack/build/1ecefb9eef5747a5832685eef6d1e856 : SUCCESS in 1h 59m 45s\n- tripleo-ci-centos-7-scenario002-standalone https://zuul.opendev.org/t/openstack/build/a3190e9ce1264d1ba9d56f1a6530ebda : SUCCESS in 1h 43m 55s","accounts_in_message":[],"_revision_number":2},{"id":"bbc5dc8d93bdab4704a56ea0a00dc06c0ee46062","author":{"_account_id":10379,"name":"Rajini Karthik","email":"rrajini@gmail.com","username":"rajinir"},"date":"2020-06-10 19:20:06.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"375b64974edd6f21ba35d33c125212f148fd54a2","author":{"_account_id":10379,"name":"Rajini Karthik","email":"rrajini@gmail.com","username":"rajinir"},"date":"2020-06-10 19:21:33.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"cef6613a5a9d0d1608eedd976b77dba6ebdd78a9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-10 22:19:24.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5ef35cc1c021481aa0a5bcb11bd3d88a : SUCCESS in 5m 11s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e0c63e08625a4cb19ef679e86ae930b2 : SUCCESS in 6m 18s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3c5cb780e4b144d884bac89a49fec0f3 : SUCCESS in 5m 00s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/40f9530a6c86460abfc35714b3e64d08 : SUCCESS in 41m 26s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2d065b5c890148dc9a7c86f0c9099ca4 : SUCCESS in 25m 19s (non-voting)\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/3058d89be74d40f293120f790d7fe5ec : SUCCESS in 1h 54m 38s\n- tripleo-ci-centos-7-containers-undercloud-minion https://zuul.opendev.org/t/openstack/build/df87ca56845c41a885e4fac8b137e632 : SUCCESS in 1h 49m 58s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/172e1d7700944104a9395a3aa0b0a8fe : FAILURE in 2h 14m 55s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/76a54c410fab44d89b0735df8fa0e2f0 : SUCCESS in 2h 47m 41s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/e0a15afaee3f4fdc82fb8506700edcfb : SUCCESS in 2h 48m 52s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/88b6803269924ac2b624105ae724e113 : SUCCESS in 2h 15m 10s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/4e67c743a5334aca9d82dee10ecc35af : SUCCESS in 1h 30m 39s\n- tripleo-ci-centos-7-standalone-upgrade-train https://zuul.opendev.org/t/openstack/build/7d6d01a94a484d51bcbf223f70958ded : SUCCESS in 2h 06m 02s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7250eb86c3d4463e9c46414de741cda5 : SUCCESS in 4m 25s\n- tripleo-ci-centos-7-scenario001-standalone https://zuul.opendev.org/t/openstack/build/410681179e7a45f3837fa9e082abcff9 : SUCCESS in 2h 10m 25s\n- tripleo-ci-centos-7-scenario002-standalone https://zuul.opendev.org/t/openstack/build/093ac01f30ca47f88372ad5406465de4 : SUCCESS in 1h 36m 36s","accounts_in_message":[],"_revision_number":4},{"id":"09a10eeb964c83aa324c5be02796224a412487bd","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2020-06-10 22:32:10.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\nCorresponding puppet-cinder patch for train (https://review.opendev.org/727279) has already merged.","accounts_in_message":[],"_revision_number":4},{"id":"48ac266db287582cc9b565b5f838b03975b6e3dd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-10 22:32:29.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"b8a4f0ddff767dbdd98f2c7f64b89b2139e5b1d1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-11 04:37:40.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/95a204381aac47399f49e4fdcdefb8b0 : SUCCESS in 6m 43s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/209aaf02f1404b729bf2e5f0c1a12a4d : SUCCESS in 4m 57s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1abcec9452664d12823369f2ace9c14f : SUCCESS in 6m 09s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/ed48d5910cb1411cb52c95eab317e356 : SUCCESS in 35m 07s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/3573a0b258c34002a7b0a12f53451178 : SUCCESS in 2h 22m 01s\n- tripleo-ci-centos-7-containers-undercloud-minion https://zuul.opendev.org/t/openstack/build/c4b2cf5979a7468abbb568b8e1abe24b : SUCCESS in 1h 34m 05s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/beb1b50bea7a45aea27269fb2ff3695a : SUCCESS in 2h 33m 33s\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/1677272a81a24684afa80610a1c0f3e0 : SUCCESS in 1h 34m 21s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/5f7f0faf6b584922bf6edf191e9cdfcc : SUCCESS in 2h 30m 52s\n- tripleo-ci-centos-7-standalone-upgrade-train https://zuul.opendev.org/t/openstack/build/f5b426e73de34b8da52bdd5ef869f938 : SUCCESS in 1h 31m 23s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5cb923d1a14d4039b9064c06a61be99b : SUCCESS in 3m 44s\n- tripleo-ci-centos-7-scenario001-standalone https://zuul.opendev.org/t/openstack/build/63b44a9fd8e4431c95549a622f0e45f2 : SUCCESS in 1h 50m 34s\n- tripleo-ci-centos-7-scenario002-standalone https://zuul.opendev.org/t/openstack/build/43fea99452e046d8b50a4eff06d0a2d3 : SUCCESS in 2h 03m 24s","accounts_in_message":[],"_revision_number":4},{"id":"f4f8ad8e3ebdb212208e1daa593f8d5d92215c3a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-11 04:37:40.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"9ce98e4240a04cb7e712039cf8eba4aceb6575de","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-11 04:38:47.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bd0e4d6a665c4a3fa942d12021c799fa : SUCCESS in 52s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"b81fc16488e3f662a2795082738108a21616639a","revisions":{"c4aaa88e3ce6f36ab44d28f2b19c4744a98a185e":{"kind":"REWORK","_number":1,"created":"2020-06-08 15:39:23.000000000","uploader":{"_account_id":10379,"name":"Rajini Karthik","email":"rrajini@gmail.com","username":"rajinir"},"ref":"refs/changes/30/734130/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tripleo","ref":"refs/changes/30/734130/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/30/734130/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/30/734130/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/30/734130/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tripleo refs/changes/30/734130/1"}}},"commit":{"parents":[{"commit":"2a2deaa7269fa2f07562872126e5f7dea1f3f09f","subject":"Merge \"Support for Dell EMC SC Volume Config\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/2a2deaa7269fa2f07562872126e5f7dea1f3f09f"}]}],"author":{"name":"rajinir","email":"rajini.karthik@dell.com","date":"2020-05-17 23:17:33.000000000","tz":-300},"committer":{"name":"rajinir","email":"rajini.karthik@dell.com","date":"2020-06-08 15:39:11.000000000","tz":-300},"subject":"Support for Dell EMC VXFlexOS Backend","message":"Support for Dell EMC VXFlexOS Backend\n\nAdding support for VXFlexOS Volume Backend\n\nChange-Id: I83b4d400947da16229c565c5311ca033b4c76d73\nCloses-Bug: 1875176\n(cherry picked from commit c71e527f46ffa9898a7def6afacb837d32faae61)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/c4aaa88e3ce6f36ab44d28f2b19c4744a98a185e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/c4aaa88e3ce6f36ab44d28f2b19c4744a98a185e"}]},"branch":"refs/heads/stable/train"},"ae35c1699a981990e408986be2659705917c3e08":{"kind":"REWORK","_number":2,"created":"2020-06-09 21:10:54.000000000","uploader":{"_account_id":10379,"name":"Rajini Karthik","email":"rrajini@gmail.com","username":"rajinir"},"ref":"refs/changes/30/734130/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tripleo","ref":"refs/changes/30/734130/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/30/734130/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/30/734130/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/30/734130/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tripleo refs/changes/30/734130/2"}}},"commit":{"parents":[{"commit":"2a2deaa7269fa2f07562872126e5f7dea1f3f09f","subject":"Merge \"Support for Dell EMC SC Volume Config\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/2a2deaa7269fa2f07562872126e5f7dea1f3f09f"}]}],"author":{"name":"rajinir","email":"rajini.karthik@dell.com","date":"2020-05-17 23:17:33.000000000","tz":-300},"committer":{"name":"rajinir","email":"rajini.karthik@dell.com","date":"2020-06-09 21:10:43.000000000","tz":-300},"subject":"Support for Dell EMC VXFlexOS Backend","message":"Support for Dell EMC VXFlexOS Backend\n\nAdding support for VXFlexOS Volume Backend\n\nChange-Id: I83b4d400947da16229c565c5311ca033b4c76d73\nCloses-Bug: 1875176\n(cherry picked from commit c71e527f46ffa9898a7def6afacb837d32faae61)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/ae35c1699a981990e408986be2659705917c3e08"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/ae35c1699a981990e408986be2659705917c3e08"}]},"branch":"refs/heads/stable/train"},"501726ae64e0015d82fd464824ebc43d698b7b03":{"kind":"REWORK","_number":3,"created":"2020-06-10 19:20:06.000000000","uploader":{"_account_id":10379,"name":"Rajini Karthik","email":"rrajini@gmail.com","username":"rajinir"},"ref":"refs/changes/30/734130/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tripleo","ref":"refs/changes/30/734130/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/30/734130/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/30/734130/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/30/734130/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tripleo refs/changes/30/734130/3"}}},"commit":{"parents":[{"commit":"2a2deaa7269fa2f07562872126e5f7dea1f3f09f","subject":"Merge \"Support for Dell EMC SC Volume Config\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/2a2deaa7269fa2f07562872126e5f7dea1f3f09f"}]}],"author":{"name":"rajinir","email":"rajini.karthik@dell.com","date":"2020-05-17 23:17:33.000000000","tz":-300},"committer":{"name":"rajinir","email":"rajini.karthik@dell.com","date":"2020-06-10 19:19:59.000000000","tz":-300},"subject":"Support for Dell EMC VXFlexOS Backend","message":"Support for Dell EMC VXFlexOS Backend\n\nAdding support for VXFlexOS Volume Backend\n\nChange-Id: I83b4d400947da16229c565c5311ca033b4c76d73\nCloses-Bug: 1875176\n(cherry picked from commit c71e527f46ffa9898a7def6afacb837d32faae61)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/501726ae64e0015d82fd464824ebc43d698b7b03"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/501726ae64e0015d82fd464824ebc43d698b7b03"}]},"branch":"refs/heads/stable/train"},"b81fc16488e3f662a2795082738108a21616639a":{"kind":"REWORK","_number":4,"created":"2020-06-10 19:21:33.000000000","uploader":{"_account_id":10379,"name":"Rajini Karthik","email":"rrajini@gmail.com","username":"rajinir"},"ref":"refs/changes/30/734130/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tripleo","ref":"refs/changes/30/734130/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/30/734130/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/30/734130/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/30/734130/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tripleo refs/changes/30/734130/4"}}},"commit":{"parents":[{"commit":"2a2deaa7269fa2f07562872126e5f7dea1f3f09f","subject":"Merge \"Support for Dell EMC SC Volume Config\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/2a2deaa7269fa2f07562872126e5f7dea1f3f09f"}]}],"author":{"name":"rajinir","email":"rajini.karthik@dell.com","date":"2020-05-17 23:17:33.000000000","tz":-300},"committer":{"name":"rajinir","email":"rajini.karthik@dell.com","date":"2020-06-10 19:21:28.000000000","tz":-300},"subject":"Support for Dell EMC VXFlexOS Backend","message":"Support for Dell EMC VXFlexOS Backend\n\nAdding support for VXFlexOS Volume Backend\n\nChange-Id: I83b4d400947da16229c565c5311ca033b4c76d73\nCloses-Bug: 1875176\n(cherry picked from commit c71e527f46ffa9898a7def6afacb837d32faae61)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/b81fc16488e3f662a2795082738108a21616639a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/b81fc16488e3f662a2795082738108a21616639a"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
