)]}'
{"id":"openstack%2Fpuppet-tripleo~744192","triplet_id":"openstack%2Fpuppet-tripleo~stable%2Ftrain~Id54537205538740a09a570a392f291a14d759f61","project":"openstack/puppet-tripleo","branch":"stable/train","topic":"pcs-setup-extra-stable/train","hashtags":[],"change_id":"Id54537205538740a09a570a392f291a14d759f61","subject":"Fix up cluster_setup_extras usage","status":"MERGED","created":"2020-07-31 14:58:00.000000000","updated":"2020-08-06 09:02:51.000000000","submitted":"2020-08-06 09:02:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"744192-1596704571488-76a6aaaf","meta_rev_id":"d7806a3965c53c904aabc88a5016ce6ac84e966c","_number":744192,"virtual_id_number":744192,"owner":{"_account_id":20778,"name":"Damien Ciabrini","email":"dciabrin@redhat.com","username":"dciabrin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":20778,"name":"Damien Ciabrini","email":"dciabrin@redhat.com","username":"dciabrin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-06 09:02:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2020-08-05 14:59:05.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2020-08-05 13:21:19.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":2,"date":"2020-08-05 13:21:19.000000000","_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":20778,"name":"Damien Ciabrini","email":"dciabrin@redhat.com","username":"dciabrin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":1,"date":"2020-08-06 05:49:58.000000000","_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":20778,"name":"Damien Ciabrini","email":"dciabrin@redhat.com","username":"dciabrin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"_account_id":20778,"name":"Damien Ciabrini","email":"dciabrin@redhat.com","username":"dciabrin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-31 18:47:02.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-08-03 17:03:01.000000000","updated_by":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"reviewer":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"state":"REVIEWER"},{"updated":"2020-08-04 06:04:23.000000000","updated_by":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"reviewer":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"state":"REVIEWER"},{"updated":"2020-08-05 14:59:05.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-08-06 05:49:58.000000000","updated_by":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"reviewer":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"state":"REVIEWER"},{"updated":"2020-08-06 09:02:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"402fd98aaa79ceaf755c09d528ec14284f10b545","author":{"_account_id":20778,"name":"Damien Ciabrini","email":"dciabrin@redhat.com","username":"dciabrin"},"date":"2020-07-31 14:58:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"80e7ba427aa7614250d5136397d704e1a9014aba","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-07-31 17:41:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/b484e93d75cb429a8f9beffc337e5c25 : SUCCESS in 2h 42m 50s","accounts_in_message":[],"_revision_number":1},{"id":"b07c42841fde7df54f66af8cdc9dc404f87578c6","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-07-31 18:47:02.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e3bc74d6e325a51144a627b7ba1751734ad4cc5d","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-07-31 18:47:05.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5b8c182890af69e140cfbce5bc9f8bd60e022694","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-31 22:25:45.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c84ef9baef7c4bbcbca4f34aa071c701 : FAILURE in 5m 59s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cc3d3b0bdf4544d0b379fc388fd7a9ad : SUCCESS in 3m 58s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3d0235e6ec2c49a584414371ff8fccb8 : SUCCESS in 4m 23s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/d1296e05e42f4a35af5cdbcccb003381 : SUCCESS in 56m 55s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c5cc676869fe4a7182d3315e216703d9 : FAILURE in 8m 36s (non-voting)\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/4d3a0966663c4552a222147f3dc3421c : SUCCESS in 1h 58m 36s\n- tripleo-ci-centos-7-containers-undercloud-minion https://zuul.opendev.org/t/openstack/build/1d188c4a9f5443b9b9b8ac2a73799f6c : SUCCESS in 1h 09m 48s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/b51205a293494b93ae4972f0bc456cdf : FAILURE in 2h 05m 26s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/4ef27890322c43e8b8c32dc833b3a4d4 : SUCCESS in 2h 56m 57s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/dfa68ce9198044b1b2b7c7c4eccef1a9 : SUCCESS in 2h 40m 37s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/caae0da0f8404bb185fb4c2ea71371f9 : SUCCESS in 2h 43m 19s (non-voting)\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-train https://zuul.opendev.org/t/openstack/build/7d2b884804a7494687213e6bf60d7c79 : FAILURE in 2h 22m 44s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/79e7104be7bf4b52af716cbf33bb60aa : SUCCESS in 1h 34m 02s\n- tripleo-ci-centos-7-standalone-upgrade-train https://zuul.opendev.org/t/openstack/build/839231ec6f144bec8e2f68e3ca466900 : SUCCESS in 2h 12m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4c80438725c7e1e43862b6a93c0ceb667ef3e640","author":{"_account_id":20778,"name":"Damien Ciabrini","email":"dciabrin@redhat.com","username":"dciabrin"},"date":"2020-08-01 08:17:10.000000000","message":"Patch Set 1:\n\nstuck with puppet-openstack-lint-ubuntu-bionic, that is being tracked in https://review.opendev.org/#/c/741572/","accounts_in_message":[],"_revision_number":1},{"id":"0a1858c7d33b66ca4a4b317564399a3e59fd1e03","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2020-08-03 17:03:01.000000000","message":"Patch Set 1: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c987707ab432b56f942a5d3dcae1d1e193a21946","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-08-03 19:48:17.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/f34f53b9f28045c0bc1d8392b83feb70 : SUCCESS in 2h 44m 25s","accounts_in_message":[],"_revision_number":1},{"id":"a067a2a2d4cbab0dce72213c6c8abce06245950d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-03 22:10:33.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/740525114dec4bd1880d3e019907df30 : FAILURE in 6m 05s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/95ca8e61ca9f4b38a71b3cca735af808 : SUCCESS in 5m 55s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ab391782262c4d31821942ee03c310d1 : SUCCESS in 5m 54s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/3337dedb736c42c89483240bdd89814d : SUCCESS in 33m 13s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/57e2f523cb8444c6871a6c5b0d1556c5 : FAILURE in 8m 51s (non-voting)\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/0528c9747b144dafb7d3ec79cf6b6efe : SUCCESS in 1h 46m 44s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/4a3d7de54fed445aa0179f5f7b18d657 : FAILURE in 2h 18m 22s (non-voting)\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/5b93b823f8ae449cbd7fa2d8775b2bbf : SUCCESS in 2h 19m 07s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/ebb8dc469bea49c8b109f7a82bd360dc : SUCCESS in 2h 25m 36s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/16070af8fe354a71b97db80b2f22fff3 : FAILURE in 55m 04s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/7214bbe0027e4406a62c29d628443ae5 : SUCCESS in 2h 17m 37s (non-voting)\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-train https://zuul.opendev.org/t/openstack/build/3339383265924c24b733612831c2abac : FAILURE in 2h 03m 06s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/d6c1761ba3a946cb82842a8ee9c2fafe : SUCCESS in 1h 50m 52s\n- tripleo-ci-centos-7-standalone-upgrade-train https://zuul.opendev.org/t/openstack/build/dfa1489fa25d41c6a4e3acbc773523c6 : SUCCESS in 2h 49m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1750146d3a90e652d6ec1ea3ba0f2bf1a9d6238a","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2020-08-04 02:04:46.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"e40c1d3c78caaae5722dcac072036bc8b6497387","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-08-04 04:48:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/489d28d61e9243dcb9db97bf0765a671 : SUCCESS in 2h 42m 16s","accounts_in_message":[],"_revision_number":1},{"id":"1dd56785d84db7f7c7bbbc4a1a05cdeba6574fc9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-04 05:11:43.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1f1b0596614849ca8c2eb03bcb532a81 : FAILURE in 5m 06s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f45ad87d016e4fcda9299b94c89dbb47 : SUCCESS in 4m 30s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/abd4f4400fb74542bdf095ab8893e1e8 : SUCCESS in 4m 38s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/6e3e9fe120ac4a1da54b481e77a66085 : SUCCESS in 40m 40s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b6d056aa5e5f468aaf895603afa3572d : FAILURE in 9m 17s (non-voting)\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/d8e7293f086e483b8726eb7bddfab1f6 : SUCCESS in 1h 56m 54s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/6253ae051aae4409832c827eeab60c65 : FAILURE in 1h 59m 37s (non-voting)\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/4b75c15d22854e689e76bfc5b7b06480 : SUCCESS in 2h 56m 56s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/1d1dbdd7217e49ecb31b854f0bf85840 : SUCCESS in 2h 25m 30s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/3703d1537c5f43bdadf7af4c859c3d62 : SUCCESS in 2h 27m 23s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/e5d165b61d7349318d8b72bf5dc29214 : SUCCESS in 2h 38m 30s (non-voting)\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-train https://zuul.opendev.org/t/openstack/build/796c4b54b6f443bdaf5d4a6dc6445095 : FAILURE in 2h 04m 15s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/f35eaeceb6ca4107b7cd55090e9776b6 : SUCCESS in 1h 38m 16s\n- tripleo-ci-centos-7-standalone-upgrade-train https://zuul.opendev.org/t/openstack/build/5282f20a39ad4d2a8ef8121a6035ee7f : SUCCESS in 2h 24m 22s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4df1eccaff3684af92ea67b8524ff148eb903eeb","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2020-08-04 08:15:07.000000000","message":"Patch Set 1:\n\ncurrently blocked by https://review.opendev.org/#/c/744499/2 to fix the puppet-openstack-lint-ubuntu-bionic FAILURE in 5m 06s (thanks chkumar|rover)","accounts_in_message":[],"_revision_number":1},{"id":"e995d6c56678922c3cb3eb145263953509c02361","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2020-08-04 20:54:56.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"00e6fedd4dec553ccba1430f56bbbedf680c0fb8","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-08-05 00:03:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/3a0ca5757b7a45cf8f3bc2dde1fe9ee2 : SUCCESS in 3h 07m 34s","accounts_in_message":[],"_revision_number":1},{"id":"b429ac22e498b1a79e896368ed0615e18dfbfe86","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-05 01:00:36.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/884fe7260b0a4a879c4005a6039ee8cc : SUCCESS in 4m 14s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8ba29e8cbba24a359b14af3ecb60fe3b : SUCCESS in 5m 17s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f2c22b642fac405b98bf52522205f916 : SUCCESS in 5m 39s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/e876bc8f815e46d18a80c8736ff2ccd0 : SUCCESS in 47m 22s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/96209fc5068f45f499c65e284f24df40 : SUCCESS in 34m 18s (non-voting)\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/082a872c35ad4ee39fb0307c8b92e0b0 : TIMED_OUT in 3h 56m 13s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/bc7d33bc907a46a2ba016ef3c0bda407 : TIMED_OUT in 3h 42m 47s (non-voting)\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/c972ffa366e14b4ba73a32fe6900c46b : TIMED_OUT in 4h 00m 55s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/247b3f611250416ba83c16025d32c9ca : FAILURE in 38m 22s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/8f377e1126ed432ca74c31e7134da7c9 : TIMED_OUT in 3h 46m 17s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/b4058015491145e3802f1f4fb4c0adfe : TIMED_OUT in 3h 46m 30s (non-voting)\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-train https://zuul.opendev.org/t/openstack/build/4d2d41cc3e1440438b9e215590576a24 : TIMED_OUT in 3h 42m 06s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/efae3844e0e44754887d0df553fec39b : SUCCESS in 2h 58m 25s\n- tripleo-ci-centos-7-standalone-upgrade-train https://zuul.opendev.org/t/openstack/build/760a28d0699b4000b3177eb34a19ea2a : TIMED_OUT in 3h 26m 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"90337bfc400c063e27884a56def5d8f64df6eea0","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2020-08-05 03:16:03.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"b16bf92f15b955e6092ea28bfde77485c1b7306b","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-08-05 03:59:46.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/ccaf5e922aa148548870248ea234fac7 : FAILURE in 42m 44s","accounts_in_message":[],"_revision_number":1},{"id":"50d89f15d7a1c68b36b40356b7597c7ed963a770","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-05 06:29:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a13d039ef1b84bd0afc7c940ea9059b3 : SUCCESS in 3m 52s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0711d5dbf69640c0afe1d74c63793bf1 : SUCCESS in 3m 40s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fc1b7a6dcaea4a22a011d9d4dead12be : SUCCESS in 4m 02s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/cea2bd38ce274384b0176d646bd1bf61 : SUCCESS in 44m 35s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a18236894bdc45fea3a007818b9be970 : SUCCESS in 14m 56s (non-voting)\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/0bb409fbbc0d46f585f3c4ea29c3e053 : SUCCESS in 1h 54m 23s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/fe1b0dcf47ef4106923d5bffd2ae5f3e : FAILURE in 2h 14m 11s (non-voting)\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/2d4ba0dfd36540498ef68bc6f2a5e501 : SUCCESS in 2h 20m 50s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/26177a1509084d8992b1d79b68049ed2 : SUCCESS in 3h 07m 56s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/a1914f2e86ec46c589cfd078583c617b : SUCCESS in 2h 01m 15s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/9d00a475809c445b845e69833ae9762d : SUCCESS in 2h 18m 46s (non-voting)\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-train https://zuul.opendev.org/t/openstack/build/dd71de0284bb4ec18ba1c3cb377f8bd5 : FAILURE in 2h 02m 33s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/0262f9d503b244fe8be442103d621604 : SUCCESS in 1h 26m 22s\n- tripleo-ci-centos-7-standalone-upgrade-train https://zuul.opendev.org/t/openstack/build/1fb4274149b6441a8cd0457c6f6c122c : SUCCESS in 2h 22m 05s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"122da3786db782560443867462b5866cd405d7b3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-05 06:30:03.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0cfb930df9b51ae3c5a801f80117c37f98c83d34","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-05 12:32:26.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2d61f7f4713d432089d4ba9362a3bd2a : SUCCESS in 6m 00s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/21ac6dcc953e480cb68232e09f28c8db : SUCCESS in 4m 33s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4df19a20c42d40429cde04fa9c717304 : SUCCESS in 6m 24s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/a23df4698e6d4bd89cab3fe9736aff1b : SUCCESS in 40m 13s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/36440116957c42709f4d473bbb01d97e : SUCCESS in 2h 08m 49s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/37f079501ec547ac8caf9de6ec5bf12b : FAILURE in 51m 07s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/bd37865b0df245ccb7c43f82b67d6899 : FAILURE in 1h 57m 09s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/17352e9d03464921b23a316b8a263979 : FAILURE in 1h 13m 16s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/c880e4ce976a4bdbbe2701532c3cf951 : SUCCESS in 2h 20m 37s\n- tripleo-ci-centos-7-standalone-upgrade-train https://zuul.opendev.org/t/openstack/build/728ad9e59cd84e63be17dd8e6395dcbb : FAILURE in 1h 53m 41s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"26365879e7eee16c0f3fbd9577205ab8b9d24c58","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2020-08-05 13:21:19.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"d976d28e73f1979447e08987e06e357382002bab","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2020-08-05 13:21:29.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"b266b5e9c4c0df32885613b21581d7acb2af0e1f","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-08-05 14:58:41.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/79ac8abca21e4dbe9a4a97ef6fa7d4b2 : FAILURE in 1h 35m 44s","accounts_in_message":[],"_revision_number":1},{"id":"11842b4b7bdb30ef38337856ac386a2933d15fca","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-08-05 14:59:05.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/1ab0b843e81743ce9f9a014d5213de8f : FAILURE in 1h 37m 01s","accounts_in_message":[],"_revision_number":2},{"id":"592cbcdfe8c088c07df828b24752f1cb3539a63d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-05 16:58:47.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/adb7001354044283bcd58cd488bedfa3 : SUCCESS in 5m 00s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4748029e9f2e4e0b815035c9435633f6 : SUCCESS in 4m 15s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/904f025b76dd4174a85e7481838b35f1 : SUCCESS in 6m 26s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/b043691a615746ff946e55d4bce1e627 : SUCCESS in 44m 52s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4ebe9425b6ae44268a2c595ba7ea3b38 : SUCCESS in 22m 20s (non-voting)\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/86021273e3494ae99b76476221742178 : SUCCESS in 2h 01m 47s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/e8f2d3cc1cb04f1f913f686e0c48e96c : FAILURE in 2h 03m 30s (non-voting)\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/88e9b49a665c4c11bd1ee7b4a2920ba6 : SUCCESS in 2h 18m 07s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/0502908549304cf48431371493e6ebe5 : SUCCESS in 2h 26m 37s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/a852fb49bfed4c70a13e40ed4a3515e2 : SUCCESS in 2h 39m 03s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/bb16142b840c4c7691115f39a71af5fe : SUCCESS in 2h 08m 29s (non-voting)\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-train https://zuul.opendev.org/t/openstack/build/45be557612b14f768398d97c338dd8b6 : FAILURE in 2h 19m 34s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/62e942ad949349bca3c39522299b0b52 : SUCCESS in 1h 17m 02s\n- tripleo-ci-centos-7-standalone-upgrade-train https://zuul.opendev.org/t/openstack/build/2f9ea2e312b44cb8a46e807e0c69d0e1 : SUCCESS in 2h 01m 07s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"74dd2bfdcd8453bf3a4e4829aadaf68f2af07e4b","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2020-08-06 05:49:58.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"00026c94c68e347e41ffc2167b0dd37e4ed250aa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-06 05:50:10.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d2da62e65d9b8cfbeafe1003484c829b64262a65","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-06 09:02:51.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"ca0dbb1fbde889ea46ae710903517b59273a982a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-06 09:02:51.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0f84aacd761b4487a2e32cc65b327359 : SUCCESS in 4m 08s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/543a72e4e689443fbbfcae51b1f1313a : SUCCESS in 3m 46s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/81b263c3c04d4fa8b36ca18261ab14d8 : SUCCESS in 4m 47s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/99a59bbfcb044775b35232f5fdb94e73 : SUCCESS in 47m 27s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/5f4c70f1c0e0480c8869f39575aba717 : SUCCESS in 1h 57m 27s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/8d0a192c39d241589ec02494d3ee80f0 : SUCCESS in 3h 07m 39s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/00b0e7d8fc8744cead8fc08660289e49 : SUCCESS in 2h 48m 33s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/d753ea34a25c47c398e46e5a03f8e565 : SUCCESS in 1h 54m 55s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/8b4da468d0be44a7b80e59757a155eb8 : SUCCESS in 2h 33m 53s\n- tripleo-ci-centos-7-standalone-upgrade-train https://zuul.opendev.org/t/openstack/build/7a3c6769e3c643bb8b03f83edbb61672 : SUCCESS in 2h 26m 38s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"abbc4e4c409f002ac0bc2ebb594aa1684f7b3424","revisions":{"664cbe4d511dfebaa0d94f8d5efd1e85af58ac36":{"kind":"REWORK","_number":1,"created":"2020-07-31 14:58:00.000000000","uploader":{"_account_id":20778,"name":"Damien Ciabrini","email":"dciabrin@redhat.com","username":"dciabrin"},"ref":"refs/changes/92/744192/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tripleo","ref":"refs/changes/92/744192/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/92/744192/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/92/744192/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/92/744192/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tripleo refs/changes/92/744192/1"}}},"commit":{"parents":[{"commit":"a0725f8645a1de4f294b8175629040133235c4e2","subject":"Merge \"HA: fix rabbitmq readiness check for rabbitmq-server 3.8\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/a0725f8645a1de4f294b8175629040133235c4e2"}]}],"author":{"name":"Damien Ciabrini","email":"dciabrin@redhat.com","date":"2020-07-31 14:48:13.000000000","tz":120},"committer":{"name":"Damien Ciabrini","email":"dciabrin@redhat.com","date":"2020-07-31 14:57:46.000000000","tz":120},"subject":"Fix up cluster_setup_extras usage","message":"Fix up cluster_setup_extras usage\n\nStarting with pcs 0.10 (CentOS/RHEL 8) the arguments to pcs cluster\nsetup changed completely in an incompatible way and in fact are ignored\nby puppet-pacemaker. As we would like to reenable them, we first need to\nmake sure that whatever we pass is not understood by pcs.\n\nWe can also drop ipv6 and encryption parameters as corosync/pcs do not\nneed them any longer.\n\n(manually adapted from Iaecced845b80e958c7ad0ef96896f1202259cca2 to\nsupport CentOS/RHEL 7 and 8)\n\nChange-Id: Id54537205538740a09a570a392f291a14d759f61\nRelated-Bug: #1886789\nRelated-Bug: #1889764\n(cherry picked from commit ef7818ecc2de85e85e08125a1304eaa17891a77c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/664cbe4d511dfebaa0d94f8d5efd1e85af58ac36"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/664cbe4d511dfebaa0d94f8d5efd1e85af58ac36"}]},"branch":"refs/heads/stable/train"},"abbc4e4c409f002ac0bc2ebb594aa1684f7b3424":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-08-05 13:21:19.000000000","uploader":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"ref":"refs/changes/92/744192/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tripleo","ref":"refs/changes/92/744192/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/92/744192/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/92/744192/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/92/744192/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tripleo refs/changes/92/744192/2"}}},"commit":{"parents":[{"commit":"5ec551b0b79abeb923132343f25eb14c8c4f7d53","subject":"Switch includes to absolute","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/5ec551b0b79abeb923132343f25eb14c8c4f7d53"}]}],"author":{"name":"Damien Ciabrini","email":"dciabrin@redhat.com","date":"2020-07-31 14:48:13.000000000","tz":120},"committer":{"name":"wes hayutin","email":"whayutin@redhat.com","date":"2020-08-05 13:21:19.000000000","tz":0},"subject":"Fix up cluster_setup_extras usage","message":"Fix up cluster_setup_extras usage\n\nStarting with pcs 0.10 (CentOS/RHEL 8) the arguments to pcs cluster\nsetup changed completely in an incompatible way and in fact are ignored\nby puppet-pacemaker. As we would like to reenable them, we first need to\nmake sure that whatever we pass is not understood by pcs.\n\nWe can also drop ipv6 and encryption parameters as corosync/pcs do not\nneed them any longer.\n\n(manually adapted from Iaecced845b80e958c7ad0ef96896f1202259cca2 to\nsupport CentOS/RHEL 7 and 8)\n\nChange-Id: Id54537205538740a09a570a392f291a14d759f61\nRelated-Bug: #1886789\nRelated-Bug: #1889764\n(cherry picked from commit ef7818ecc2de85e85e08125a1304eaa17891a77c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/abbc4e4c409f002ac0bc2ebb594aa1684f7b3424"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/abbc4e4c409f002ac0bc2ebb594aa1684f7b3424"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
