)]}'
{"id":"openstack%2Fpuppet-tripleo~745255","triplet_id":"openstack%2Fpuppet-tripleo~stable%2Ftrain~I0ea26253355a57b3721bfa6ceef3972eaabc5b1d","project":"openstack/puppet-tripleo","branch":"stable/train","topic":"etcd-cert-revamp","hashtags":[],"change_id":"I0ea26253355a57b3721bfa6ceef3972eaabc5b1d","subject":"Add certmonger-etcd-refresh.sh script","status":"MERGED","created":"2020-08-07 00:15:57.000000000","updated":"2020-08-10 19:45:51.000000000","submitted":"2020-08-10 19:45:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"745255-1597088751830-12991b16","meta_rev_id":"da350ec566c5bba7758c783e0419648f3292dfb1","_number":745255,"virtual_id_number":745255,"owner":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-10 19:45:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-08-07 02:39:10.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2020-08-10 12:52:41.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2020-08-10 12:52:41.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-07 02:39:10.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-08-10 12:52:41.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-08-10 19:45:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"858e67ecb18119ce5099d2719c2a3e663af7585d","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2020-08-07 00:15:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f6b92e19ee4014aeea327a328aac6d3ebb9dda93","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-08-07 02:39:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/91d6a5cdb5f5478199adcf8715693255 : SUCCESS in 2h 22m 26s","accounts_in_message":[],"_revision_number":1},{"id":"e1b927eab12164ebf6617e270fa86358d6f44062","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-07 03:08:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a18b61edc9f74bc9b340c093af710e88 : SUCCESS in 5m 26s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fe064459f0b44a038e0726a49aaf49a4 : SUCCESS in 3m 45s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cc38da6f65454d06a9c8233beee8cf52 : SUCCESS in 5m 24s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/5888c39db3eb4d69bd9d2b980346e388 : SUCCESS in 42m 52s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c692093cb1a641e3b901cfff16a5c9a2 : SUCCESS in 20m 20s (non-voting)\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/00a4f5bd74004f17814c8fd71bbbd0d4 : SUCCESS in 1h 48m 24s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/6deca8eec8e2422b8d6b862fae3a710b : FAILURE in 2h 22m 34s (non-voting)\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/7b6068b2abc74088bfea2c9139154766 : SUCCESS in 2h 15m 49s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/2abbeee231874fc7a63d72bc2eefdc97 : SUCCESS in 2h 45m 04s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/e9d7d7e427934e28b3f21accc7ab74b6 : SUCCESS in 2h 16m 27s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/9cc3d22dbd7748978519b283db45d9db : SUCCESS in 2h 11m 15s (non-voting)\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-train https://zuul.opendev.org/t/openstack/build/65218cb8839b43c9941bd6d90c586fdf : FAILURE in 2h 36m 33s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/ac091cb869604bf59b89c1aec630c456 : SUCCESS in 1h 30m 59s\n- tripleo-ci-centos-7-standalone-upgrade-train https://zuul.opendev.org/t/openstack/build/bdf8dc496ca84df78737d84f539ff9ec : FAILURE in 1h 57m 40s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f724e9f0bbe42f699f570e057d1f20a817b85679","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-08-10 12:52:41.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0dc9b4e2aa262c91657a286ba4a21f1ca9906b18","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-10 12:53:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cb555ae18adfdc04c61a7268a7e03b6621201e57","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-10 19:45:51.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"3e2ace320e32fd176320c2d3f90ea7248cfe888b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-10 19:45:51.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/87b7abca740c421e8f943b7b890c3029 : SUCCESS in 4m 41s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/aa472da411294b7aaee4c3670ee1411a : SUCCESS in 4m 10s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2a220e5484de4c148c71ae06c11bc5ae : SUCCESS in 5m 27s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/78f32f938c8e47f4a3c56899bd8bb78f : SUCCESS in 40m 15s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/4846eeaf3f91497991eafc0d40891a69 : SUCCESS in 1h 53m 50s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/d49ec7968173446db5d152360ee6c12e : SUCCESS in 2h 39m 25s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/89475e0dec0c4b42b5a91e681e978ff6 : SUCCESS in 2h 29m 49s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/6da0d11cce8443e19483f37397b49263 : SUCCESS in 1h 40m 54s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/e6fc3a77af1949bf8be27e8f60e9becf : SUCCESS in 2h 17m 59s\n- tripleo-ci-centos-7-standalone-upgrade-train https://zuul.opendev.org/t/openstack/build/6b4b27706468416498559a72faa74d79 : SUCCESS in 2h 30m 05s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d833f860835362b2a28108264ebe44fb8286f9a0","revisions":{"d833f860835362b2a28108264ebe44fb8286f9a0":{"kind":"REWORK","_number":1,"created":"2020-08-07 00:15:57.000000000","uploader":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"ref":"refs/changes/55/745255/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tripleo","ref":"refs/changes/55/745255/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/55/745255/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/55/745255/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/55/745255/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tripleo refs/changes/55/745255/1"}}},"commit":{"parents":[{"commit":"a612d62611c881ebcfff66cf5a02ed4a6dde37bb","subject":"Merge \"Fix up cluster_setup_extras usage\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/a612d62611c881ebcfff66cf5a02ed4a6dde37bb"}]}],"author":{"name":"Alan Bishop","email":"abishop@redhat.com","date":"2020-07-29 17:00:31.000000000","tz":-420},"committer":{"name":"Alan Bishop","email":"abishop@redhat.com","date":"2020-08-07 00:08:29.000000000","tz":-420},"subject":"Add certmonger-etcd-refresh.sh script","message":"Add certmonger-etcd-refresh.sh script\n\nAdd a script that refreshes the etcd cert and key files in all\ncontainers that reference them. This includes etcd itself, plus any\ncinder services that access etcd.\n\nChange-Id: I0ea26253355a57b3721bfa6ceef3972eaabc5b1d\n(cherry picked from commit 95db8b4947f0a80a0109131eda8f66ed4acb90d6)\n(cherry picked from commit 37b8666f64d8d8e65c483ec37efe99ac763f62c9)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/d833f860835362b2a28108264ebe44fb8286f9a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/d833f860835362b2a28108264ebe44fb8286f9a0"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
