)]}'
{"id":"openstack%2Fpuppet-tripleo~746149","triplet_id":"openstack%2Fpuppet-tripleo~stable%2Fussuri~Iad23663d6c98e4fd3a507980638870e0ad0cee45","project":"openstack/puppet-tripleo","branch":"stable/ussuri","topic":"bug/1888398-stable/ussuri","hashtags":[],"change_id":"Iad23663d6c98e4fd3a507980638870e0ad0cee45","subject":"Use pcs 0.9 style authkey/remotes when doing an upgrade","status":"MERGED","created":"2020-08-13 14:46:49.000000000","updated":"2020-08-17 15:16:59.000000000","submitted":"2020-08-17 15:16:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"746149-1597677419456-c9d102ff","meta_rev_id":"5cbe67664f3baee973826278dd669dc9ebd72bc2","_number":746149,"virtual_id_number":746149,"owner":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":20778,"name":"Damien Ciabrini","email":"dciabrin@redhat.com","username":"dciabrin"},{"value":0,"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-17 15:16:59.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-08-14 10:22:37.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":20778,"name":"Damien Ciabrini","email":"dciabrin@redhat.com","username":"dciabrin"},"all":[{"value":2,"date":"2020-08-17 07:57:50.000000000","_account_id":20778,"name":"Damien Ciabrini","email":"dciabrin@redhat.com","username":"dciabrin"},{"value":0,"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"all":[{"value":0,"_account_id":20778,"name":"Damien Ciabrini","email":"dciabrin@redhat.com","username":"dciabrin"},{"value":1,"date":"2020-08-17 10:20:15.000000000","_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"_account_id":20778,"name":"Damien Ciabrini","email":"dciabrin@redhat.com","username":"dciabrin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-14 10:22:37.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-08-17 07:57:50.000000000","updated_by":{"_account_id":20778,"name":"Damien Ciabrini","email":"dciabrin@redhat.com","username":"dciabrin"},"reviewer":{"_account_id":20778,"name":"Damien Ciabrini","email":"dciabrin@redhat.com","username":"dciabrin"},"state":"REVIEWER"},{"updated":"2020-08-17 15:16:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2b684ba80837b8277dfb7889f87bb9eaa7b6a1a8","author":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"date":"2020-08-13 14:46:49.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"059e7c92fe55c2f2524abd4645594857a4c10835","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-08-13 16:13:57.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/c1b0e33bc8d04063acc2da03e49588c4 : FAILURE in 1h 26m 25s","accounts_in_message":[],"_revision_number":1},{"id":"659a87287fec8823a4358ffcda4cc97aa4e207ca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-13 18:30:49.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f4fca98ab141487c836783f48e6ef065 : SUCCESS in 4m 24s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/23458f0adde04fb286ab2a5db5469d60 : SUCCESS in 3m 43s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d5900071c71d489999e637d95b30316a : SUCCESS in 5m 20s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/1539b84db00f45f288b2bdbe0c735187 : SUCCESS in 24m 00s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/722a4bc338c1442faee95186ebc43bc1 : FAILURE in 6m 48s (non-voting)\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/b479917f5b1949dfaa6fa982cabe82cb : SUCCESS in 1h 45m 54s\n- tripleo-ci-centos-8-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/801d4fa45f85473a8c4c1d5ef46e91c3 : FAILURE in 2h 11m 51s (non-voting)\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/bef68fc27ebc4a05a7e0e54d804a3282 : FAILURE in 1h 12m 39s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-ussuri https://zuul.opendev.org/t/openstack/build/af87ab4e0b854defb33653a1a7efd9fd : FAILURE in 1h 24m 57s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-upgrades-ussuri https://zuul.opendev.org/t/openstack/build/579ec0a10021457789d64fa9e336281a : SUCCESS in 3h 11m 55s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/cbf793130b33429b82be9074bf8a9a89 : FAILURE in 1h 24m 35s\n- tripleo-ci-centos-8-standalone-upgrade-ussuri https://zuul.opendev.org/t/openstack/build/bd69504d3cef44ff8eb1cac078665fc9 : FAILURE in 1h 24m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"51b3b99e26863c34995fdfc58cfe9670f381947d","author":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"date":"2020-08-14 08:11:37.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"a84086c61ac470bdb26be643eddfd1b0d325353b","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-08-14 10:22:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/059712555f5c43acbe8ca2b7078c3f5f : SUCCESS in 2h 08m 00s","accounts_in_message":[],"_revision_number":1},{"id":"b70555cddcdaac204e2987842f75975220b113be","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-14 10:28:04.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/24d754ae3b1945c8b45ac68cd596ab22 : SUCCESS in 5m 55s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8a5e00cecb7f4b038e5de319192c8f23 : SUCCESS in 5m 43s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5531d20ef56840f8ab0649b45d238824 : SUCCESS in 5m 52s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/6dfc6c475efb448d98050974fb56a769 : SUCCESS in 22m 57s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9ee6173b671f4ba8a226668f469cf591 : FAILURE in 9m 15s (non-voting)\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/8213321cea9f4f438d7955eac0fe77bc : SUCCESS in 1h 30m 27s\n- tripleo-ci-centos-8-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/8fd6d3597af54f26b7a3d17642298150 : FAILURE in 2h 05m 00s (non-voting)\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/a9ae10f73db34da8936dd9d4a7ef41ae : FAILURE in 2h 04m 54s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-ussuri https://zuul.opendev.org/t/openstack/build/162f48b16e3248bf81ceb31310024c24 : FAILURE in 2h 06m 31s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-upgrades-ussuri https://zuul.opendev.org/t/openstack/build/f9505d0762324f2a924f89fa828eec54 : FAILURE in 2h 07m 09s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/32da439bc060433a8f4396da0287af46 : SUCCESS in 1h 36m 40s\n- tripleo-ci-centos-8-standalone-upgrade-ussuri https://zuul.opendev.org/t/openstack/build/18fea61debc24e9fa9baf8a0f65fac48 : FAILURE in 1h 26m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ec2d5324056b58b228e9219366defdd1d6cea39f","author":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"date":"2020-08-15 09:49:00.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"a26055b467834ebec80ad7a44b11a58e9f0839d1","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-08-15 12:09:32.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/f6767507a6ee4114988ac80d8b2415ff : SUCCESS in 2h 18m 30s","accounts_in_message":[],"_revision_number":1},{"id":"2153d4df0decc44232abc89c4cd4f678890ab57b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-15 13:03:46.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3f605baa659c4c38af786d9440f33b24 : SUCCESS in 4m 57s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9a291a29e2634abeb0c723934ab7b8fd : SUCCESS in 4m 36s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4acb6d6d481649f8a56bd083e76680a9 : SUCCESS in 4m 48s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/f97d83e8c6cf460a9327e245487e9639 : SUCCESS in 20m 06s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9198b8fa829c48eeabd43057e66527c6 : FAILURE in 9m 12s (non-voting)\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/0f8360a7e4384220a1149050f9d375c3 : SUCCESS in 2h 00m 43s\n- tripleo-ci-centos-8-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/f637ed3b28f447fc9cc8cce506e4b74b : SUCCESS in 3h 03m 40s (non-voting)\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/4731ded3e68b4f3dac21e20f7288b8d7 : SUCCESS in 2h 45m 10s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-ussuri https://zuul.opendev.org/t/openstack/build/625af735faf2463f8314d8b0bee8d4ac : TIMED_OUT in 3h 11m 31s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-upgrades-ussuri https://zuul.opendev.org/t/openstack/build/1893df68dd4346c9a6cbedb7f749d2f7 : SUCCESS in 2h 11m 53s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/c3820638d2fd49c396a4b19c0d747363 : SUCCESS in 2h 08m 49s\n- tripleo-ci-centos-8-standalone-upgrade-ussuri https://zuul.opendev.org/t/openstack/build/9ed0c60d6c5c4ccbb216732e6f1c9db3 : FAILURE in 1h 20m 26s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"00d7b93916532aedaa66f3f969ff9866aee25479","author":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"date":"2020-08-16 14:08:13.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"5da770eef09a844648e64f91af948eafc02dcfa4","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-08-16 16:20:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/0bf3bae50a384c11abaa87b7d804263c : SUCCESS in 2h 11m 10s","accounts_in_message":[],"_revision_number":1},{"id":"dcdce3b9516b633031fc1ea763b2b85873daf869","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-16 17:29:11.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/122963f0858040bc83c171d0a80534a0 : SUCCESS in 4m 26s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f0400980b691407f884b6987ee2e62c2 : SUCCESS in 5m 06s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5da9f876753343c8a408dcca145fe5f6 : SUCCESS in 4m 37s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/3b4365e5a50047158dfa2819fcb62573 : SUCCESS in 26m 06s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1bfab7f9390a47ceb138a40ab0cf135e : FAILURE in 9m 46s (non-voting)\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/86f131bbcfcf4b68b8f675dd266e0cc5 : SUCCESS in 1h 45m 09s\n- tripleo-ci-centos-8-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/79877bb79adc4564a159e39e2a8fb192 : SUCCESS in 2h 41m 44s (non-voting)\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/7799da4032c34fbc8652ba391fb71dbc : SUCCESS in 2h 36m 49s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-ussuri https://zuul.opendev.org/t/openstack/build/576cf880a62f42a3a24e1e821c9eda9e : SUCCESS in 2h 21m 49s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-upgrades-ussuri https://zuul.opendev.org/t/openstack/build/4274da57eb6b448dadb49bab65134997 : TIMED_OUT in 3h 18m 26s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/2f90bfbf673d417d837b692d3cb751e2 : SUCCESS in 1h 59m 33s\n- tripleo-ci-centos-8-standalone-upgrade-ussuri https://zuul.opendev.org/t/openstack/build/2cf51383b65c4b8e8aaef4f19df31c57 : FAILURE in 1h 50m 18s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"106f5230b7ebf06dd0bf6278c0295422f038c463","author":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"date":"2020-08-17 07:03:27.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c9fc049e0a6761f127250c46a24034734160ce08","author":{"_account_id":20778,"name":"Damien Ciabrini","email":"dciabrin@redhat.com","username":"dciabrin"},"date":"2020-08-17 07:57:50.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"421179f947f769d187986c098713be4da0caf6cf","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-08-17 09:20:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/de7344ec58ec434bafda6acbc7ad44a8 : SUCCESS in 2h 15m 56s","accounts_in_message":[],"_revision_number":1},{"id":"beb2f1334fc4651d89818d12b3061db052692932","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-17 10:05:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bd78fb2d25394a3781cb1b2b264fd3ab : SUCCESS in 4m 24s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fa293743dee640d88c1cc8ef77bb7bb9 : SUCCESS in 5m 32s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d37808c361ce4e81911e711e17bbe7d9 : SUCCESS in 5m 33s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/6df3bfbcdb984e6a93cd389e0e45b1c8 : SUCCESS in 25m 52s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7245fc26a59246c0aca71da8554bb70e : FAILURE in 8m 25s (non-voting)\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/d0993ddf788c415aa06fc8d53b81d383 : SUCCESS in 1h 35m 01s\n- tripleo-ci-centos-8-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/ad547fed78ae4308a11ba0b8c51c331b : SUCCESS in 2h 48m 25s (non-voting)\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/73a52ac985ab4e9fa2a29bf7c6265a6f : SUCCESS in 2h 22m 08s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-ussuri https://zuul.opendev.org/t/openstack/build/ca4bbd72c3f4452b86e3b240c8dd71bc : SUCCESS in 2h 03m 18s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-upgrades-ussuri https://zuul.opendev.org/t/openstack/build/a9721fb98e504658a65855ac94ed002e : SUCCESS in 2h 35m 54s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/30a704ffe8764b16b1d6fd4eb9844fb7 : SUCCESS in 1h 52m 34s\n- tripleo-ci-centos-8-standalone-upgrade-ussuri https://zuul.opendev.org/t/openstack/build/4abdd614c3d44b1aa48b289bbda14369 : FAILURE in 1h 37m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7b8936959f6dfa50a2b0341c3ff3a5aeca62c853","author":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"date":"2020-08-17 10:20:15.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"88d78bcee9c5b184d3ec7cf6d6d71c76db38e584","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-17 10:20:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"550564236d8f899b86c12e278171284d15d19f76","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-17 15:16:59.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0840e70395294ad8af848053698b0e1d : SUCCESS in 5m 26s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/30576952000946a28bf77143ff822c4c : SUCCESS in 5m 26s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/513494e3ead04c19af9207bd35ea6638 : SUCCESS in 4m 46s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/bd4f9acfac1b43b18174696d8ad93bf9 : SUCCESS in 22m 01s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/8fed411dd2b7443292212a9c33127638 : SUCCESS in 1h 29m 46s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/9ec454d64a654b6287bb10b0bc353452 : SUCCESS in 2h 34m 49s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/bed9d8c52c3141598bc8b2456189593e : SUCCESS in 1h 38m 46s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-ussuri https://zuul.opendev.org/t/openstack/build/fad493ed54d744c2a64f31ab4de92421 : SUCCESS in 2h 39m 48s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-upgrades-ussuri https://zuul.opendev.org/t/openstack/build/81ecca9ecf364b8c9c3cf65294c71b50 : SUCCESS in 2h 10m 35s","accounts_in_message":[],"_revision_number":1},{"id":"9f6cdad4e92ced587b24bb905aa4fa2a5fd978d2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-17 15:16:59.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"16bb6f5d7f24e4686c78e54bb5072efc2603086a","revisions":{"16bb6f5d7f24e4686c78e54bb5072efc2603086a":{"kind":"REWORK","_number":1,"created":"2020-08-13 14:46:49.000000000","uploader":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"ref":"refs/changes/49/746149/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tripleo","ref":"refs/changes/49/746149/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/49/746149/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/49/746149/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/49/746149/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tripleo refs/changes/49/746149/1"}}},"commit":{"parents":[{"commit":"e9cc37eafc94fce35b2c03aa034a8d52fa494277","subject":"Merge \"Fix up cluster_setup_extras usage\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/e9cc37eafc94fce35b2c03aa034a8d52fa494277"}]}],"author":{"name":"Michele Baldessari","email":"michele@acksyn.org","date":"2020-07-15 12:59:48.000000000","tz":120},"committer":{"name":"Michele Baldessari","email":"michele@acksyn.org","date":"2020-08-13 14:46:49.000000000","tz":0},"subject":"Use pcs 0.9 style authkey/remotes when doing an upgrade","message":"Use pcs 0.9 style authkey/remotes when doing an upgrade\n\nWe leverage the _override keys for pacemaker and pacemaker_remote\nservices in order to decide if we should use the old way of managing\nremotes (managed authkey and pcs 0.9 way of doing remotes).\n\nThe pacemaker_remote override keys are introduced via\nhttps://review.opendev.org/741610\n\nRelated-Bug: #1888398\n\nChange-Id: Iad23663d6c98e4fd3a507980638870e0ad0cee45\n(cherry picked from commit 5cb4565c016d4de48958f7cb9cffce5b685290eb)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/16bb6f5d7f24e4686c78e54bb5072efc2603086a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/16bb6f5d7f24e4686c78e54bb5072efc2603086a"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
