)]}'
{"id":"openstack%2Fpuppet-tripleo~749882","triplet_id":"openstack%2Fpuppet-tripleo~stable%2Ftrain~I8f92469aa8dd0b8c21821b3ec62b2d43361ee428","project":"openstack/puppet-tripleo","branch":"stable/train","topic":"fix-haproxy-2.2-stable/train","hashtags":[],"change_id":"I8f92469aa8dd0b8c21821b3ec62b2d43361ee428","subject":"Replace union to create tcp-check list for redis","status":"MERGED","created":"2020-09-04 07:22:40.000000000","updated":"2020-09-09 02:08:40.000000000","submitted":"2020-09-09 02:08:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"749882-1599617320182-a980bdbc","meta_rev_id":"2ec5bb65d3f58a040ec41b0ca4c86159e094645d","_number":749882,"virtual_id_number":749882,"owner":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-09 02:08:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-09-04 14:32:42.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2020-09-08 13:31:26.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2020-09-04 16:10:43.000000000","_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2020-09-08 13:31:26.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-04 14:32:42.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-09-04 16:10:43.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2020-09-08 13:31:26.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-09-09 02:08:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3fb2d38946b09c3a5e186e20424c95b00401648c","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2020-09-04 07:22:40.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"69f246f9adfe111a2c6fda78d689085b71d698d5","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-09-04 10:04:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/9016a2c3b18d4811a3f231e6013e82d8 : SUCCESS in 2h 40m 46s","accounts_in_message":[],"_revision_number":1},{"id":"4911277170dfbd57b5e6b9694d03df309bac7be3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-04 10:47:14.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/49b916297b8144a4832f8f4c34637733 : SUCCESS in 4m 54s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3a0921839afb4dbea450abb5759bfc4a : SUCCESS in 4m 34s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/be6ea259d2f44c00b17ce50e3eeb3aa1 : SUCCESS in 4m 42s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/b4773639354745a39a702eff6bd07bd3 : FAILURE in 14m 52s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1ce9f143804744aa9cff47f0e9eac3e3 : FAILURE in 13m 58s (non-voting)\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/31693507ddaf47cbb665bd4f8cb35aca : SUCCESS in 1h 40m 07s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/701f4428742941ef85191686c7015a24 : FAILURE in 2h 07m 26s (non-voting)\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/bbd6e5a6dcd74564853e9eb4bb4d9f1d : SUCCESS in 2h 16m 40s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/532825ef99de4f06920a7ae17a914a86 : SUCCESS in 2h 24m 10s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/869ad19a2c5048e4a64e450cb1323a8a : SUCCESS in 1h 55m 46s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/d66f8e966f1245e6a93f45123af66ffb : FAILURE in 1h 59m 00s (non-voting)\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-train https://zuul.opendev.org/t/openstack/build/d5da6b85cc8b48d18b127c308285ec36 : SUCCESS in 1h 59m 48s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/cafe5011deb446caa26bc98f81a2a089 : SUCCESS in 1h 23m 41s\n- tripleo-ci-centos-7-standalone-upgrade-train https://zuul.opendev.org/t/openstack/build/58ce9de6f9114c1082298fb1cf7053c7 : SUCCESS in 2h 32m 21s (non-voting)\n- tripleo-ci-centos-8-scenario001-standalone https://zuul.opendev.org/t/openstack/build/e1f9c4a243ec4ddf8683480764831f2d : SUCCESS in 1h 43m 39s\n- tripleo-ci-centos-8-scenario003-standalone https://zuul.opendev.org/t/openstack/build/06d2e30ed5f041e88fb84c3f75b0049a : SUCCESS in 1h 04m 51s\n- tripleo-ci-centos-8-scenario004-standalone https://zuul.opendev.org/t/openstack/build/b1b6b4963f05428cac79a953f9b40f70 : SUCCESS in 1h 36m 26s\n- tripleo-ci-centos-8-scenario010-standalone https://zuul.opendev.org/t/openstack/build/f1ec3c892f5c4ab98f5854ee75c2584c : SUCCESS in 2h 04m 58s (non-voting)\n- tripleo-ci-centos-8-scenario010-ovn-provider-standalone https://zuul.opendev.org/t/openstack/build/80ceb8b9c0fb44488984767f3bff6085 : SUCCESS in 1h 20m 36s (non-voting)\n- tripleo-ci-centos-8-scenario012-standalone https://zuul.opendev.org/t/openstack/build/f03c13cbcd1543bcad8f82dfee0d723b : SUCCESS in 1h 12m 50s\n- tripleo-ci-centos-8-scenario007-standalone https://zuul.opendev.org/t/openstack/build/c233e70284284da584a9dd8522cb4658 : SUCCESS in 1h 01m 02s","accounts_in_message":[],"_revision_number":1},{"id":"5b6da26c2373d1e6512cf7c317de90fda8c93340","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2020-09-04 11:53:48.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"a54a1c18c16c10b2214a1cfb781798852f43affd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-04 14:27:00.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6446e4a0aafe483d88f9250bb185eba0 : SUCCESS in 4m 57s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9c068430c2954206a736f17c52be5d95 : SUCCESS in 5m 39s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/51417d3cff2641c8b3a5daa7f6757996 : SUCCESS in 7m 20s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/6f46e7c4b95641a49a30610940980ab7 : SUCCESS in 42m 54s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/54126eb755b844a6b41c52af89c3e58f : SUCCESS in 39m 07s (non-voting)\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/7ecdf760489747e3929eb873de2cee1b : SUCCESS in 1h 40m 34s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/46d564320c42481a8754c51bef4064e1 : FAILURE in 2h 01m 14s (non-voting)\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/c65289ab4fa14257a8f02357ad12d20b : SUCCESS in 2h 14m 58s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/daccd1a889f84a33bf3ee3392203f884 : SUCCESS in 2h 30m 44s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/5bcec7d1b50e4c1d82f5b3fe7ea38a5f : SUCCESS in 1h 57m 38s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/68bf0e6df1eb428e8f10a8a51597d0a0 : SUCCESS in 2h 06m 04s (non-voting)\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-train https://zuul.opendev.org/t/openstack/build/c4933a477f9145f09124023f4c71dc42 : SUCCESS in 1h 51m 32s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/d56ec3758bc84bcf93bad504c6db9422 : SUCCESS in 1h 27m 30s\n- tripleo-ci-centos-7-standalone-upgrade-train https://zuul.opendev.org/t/openstack/build/d5bafe31277145e29c5c88cc97350916 : SUCCESS in 2h 12m 28s (non-voting)\n- tripleo-ci-centos-8-scenario001-standalone https://zuul.opendev.org/t/openstack/build/027528ca686c470b962ebc04f88ac00f : SUCCESS in 1h 43m 07s\n- tripleo-ci-centos-8-scenario003-standalone https://zuul.opendev.org/t/openstack/build/6f59f50d52604784b9d17cb2c5cc56bd : SUCCESS in 1h 11m 28s\n- tripleo-ci-centos-8-scenario004-standalone https://zuul.opendev.org/t/openstack/build/24f42b3dd131406289dd756ab0988b87 : SUCCESS in 2h 15m 35s\n- tripleo-ci-centos-8-scenario010-standalone https://zuul.opendev.org/t/openstack/build/5dd9c076598445d882cf62f507137dd5 : SUCCESS in 1h 53m 01s (non-voting)\n- tripleo-ci-centos-8-scenario010-ovn-provider-standalone https://zuul.opendev.org/t/openstack/build/baaa0101db82439aa9df9ccd961da970 : SUCCESS in 1h 44m 40s (non-voting)\n- tripleo-ci-centos-8-scenario012-standalone https://zuul.opendev.org/t/openstack/build/48b269ad9af34f869842c2cb6d5f601f : SUCCESS in 1h 13m 24s\n- tripleo-ci-centos-8-scenario007-standalone https://zuul.opendev.org/t/openstack/build/8ea8fc4dafbb4418bd189c631d59c038 : SUCCESS in 59m 32s","accounts_in_message":[],"_revision_number":2},{"id":"686608daea7a5cf9e69a1f23c5b78b5bd21070ad","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-09-04 14:32:42.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/932fbf9affdb4bab8f9867a6db768cfb : SUCCESS in 2h 37m 54s","accounts_in_message":[],"_revision_number":2},{"id":"71e171db346d52057263b8d19997d06bf9028334","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-09-04 16:10:43.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"c921a7056506810145a325f1e63b10004ef129cc","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-09-08 13:31:26.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"225a83666f66e5da46f324896a175d7d9fea0efc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-08 13:31:55.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"11c497902da3cf5f2f7430a902134dd4ad20f106","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-09 02:08:39.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4182c337ae84442e888f82f8c73b05fd : SUCCESS in 4m 56s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fdc5e2044e5442bfb6fbddc9e20a665c : SUCCESS in 5m 10s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1dc10e35624646279e6491dad6b1d1fc : SUCCESS in 4m 25s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/bea95494b1d940a9ba613020ddb22a13 : SUCCESS in 42m 40s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/a7a44c8b2c964890b95edb65dc62056c : SUCCESS in 1h 40m 25s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/3bbd35a301954f0786de90b143253096 : SUCCESS in 2h 21m 18s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/7cb65146eaa640ad9d5d21b301a8af6d : SUCCESS in 2h 05m 21s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/e818828d04f94807b80f572872450ded : SUCCESS in 1h 08m 05s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/be75194067d941d2a72ae40dee0c8784 : SUCCESS in 1h 39m 36s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-train https://zuul.opendev.org/t/openstack/build/360152a6f7f0427e923ae7c832f98d85 : SUCCESS in 1h 44m 50s\n- tripleo-ci-centos-7-standalone-upgrade-train https://zuul.opendev.org/t/openstack/build/187e4fd42202427fb11b5d205d11699d : FAILURE in 2h 00m 53s (non-voting)\n- tripleo-ci-centos-8-scenario001-standalone https://zuul.opendev.org/t/openstack/build/d0923d23f9e245d89ebe3449f499dd7b : SUCCESS in 1h 28m 24s\n- tripleo-ci-centos-8-scenario003-standalone https://zuul.opendev.org/t/openstack/build/09d1b1cce6244c009281e856496e5c11 : SUCCESS in 1h 07m 37s\n- tripleo-ci-centos-8-scenario004-standalone https://zuul.opendev.org/t/openstack/build/81e59e31027143578d34043cf1f78851 : SUCCESS in 1h 29m 52s\n- tripleo-ci-centos-8-scenario007-standalone https://zuul.opendev.org/t/openstack/build/6e6902b35354472cae44a2bf4b5e9f12 : SUCCESS in 43m 03s\n- tripleo-ci-centos-8-scenario012-standalone https://zuul.opendev.org/t/openstack/build/8da4173a31aa41079f97ed7dccb8a963 : SUCCESS in 52m 20s","accounts_in_message":[],"_revision_number":2},{"id":"5e396837656db6b7a61638967d0981e25cf39746","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-09 02:08:40.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"f716ef5a85bdcf20bcda9ae976cecf51faa0ac58","revisions":{"de3437312b2ed15bab70bd2f406aac0c1f490ddf":{"kind":"REWORK","_number":1,"created":"2020-09-04 07:22:40.000000000","uploader":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"ref":"refs/changes/82/749882/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tripleo","ref":"refs/changes/82/749882/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/82/749882/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/82/749882/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/82/749882/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tripleo refs/changes/82/749882/1"}}},"commit":{"parents":[{"commit":"72ad205fa85377663b72d9dc7d297452098a41cb","subject":"Fix tls - zaqar wss default port","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/72ad205fa85377663b72d9dc7d297452098a41cb"}]}],"author":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2020-09-02 14:32:21.000000000","tz":120},"committer":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2020-09-04 07:22:40.000000000","tz":0},"subject":"Replace union to create tcp-check list for redis","message":"Replace union to create tcp-check list for redis\n\nWhen using union, repeated arrays members are removed, so in this case\nwe are missing a \u0027expect string +ok\u0027 after the QUIT. To avoid that i\u0027m\nusinhg + operator to join arrays.\n\nChange-Id: I8f92469aa8dd0b8c21821b3ec62b2d43361ee428\n(cherry picked from commit 3130cc880a1f714ee8b38855f471b85a18834c1d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/de3437312b2ed15bab70bd2f406aac0c1f490ddf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/de3437312b2ed15bab70bd2f406aac0c1f490ddf"}]},"branch":"refs/heads/stable/train"},"f716ef5a85bdcf20bcda9ae976cecf51faa0ac58":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-09-04 11:53:48.000000000","uploader":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"ref":"refs/changes/82/749882/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tripleo","ref":"refs/changes/82/749882/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/82/749882/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/82/749882/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/82/749882/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tripleo refs/changes/82/749882/2"}}},"commit":{"parents":[{"commit":"819c2592bf9baa4659510c21f2fd40aad403973d","subject":"Change branch name for selinux_core puppet module","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/819c2592bf9baa4659510c21f2fd40aad403973d"}]}],"author":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2020-09-02 14:32:21.000000000","tz":120},"committer":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2020-09-04 11:53:48.000000000","tz":0},"subject":"Replace union to create tcp-check list for redis","message":"Replace union to create tcp-check list for redis\n\nWhen using union, repeated arrays members are removed, so in this case\nwe are missing a \u0027expect string +ok\u0027 after the QUIT. To avoid that i\u0027m\nusinhg + operator to join arrays.\n\nChange-Id: I8f92469aa8dd0b8c21821b3ec62b2d43361ee428\n(cherry picked from commit 3130cc880a1f714ee8b38855f471b85a18834c1d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/f716ef5a85bdcf20bcda9ae976cecf51faa0ac58"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/f716ef5a85bdcf20bcda9ae976cecf51faa0ac58"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
