)]}'
{"id":"openstack%2Fpuppet-tripleo~757142","triplet_id":"openstack%2Fpuppet-tripleo~master~I4da96f2164cf1d136f9471f1d6251bdd8cfd2d0b","project":"openstack/puppet-tripleo","branch":"master","topic":"key_size_cert","hashtags":[],"change_id":"I4da96f2164cf1d136f9471f1d6251bdd8cfd2d0b","subject":"Adding key_size option on the certmonger_certificate function","status":"MERGED","created":"2020-10-09 16:00:05.000000000","updated":"2021-01-05 02:52:29.000000000","submitted":"2021-01-05 02:52:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"757142-key_size_cert","meta_rev_id":"3eb878747ef5a90684a6828d4ef2dd2ad8efef32","_number":757142,"virtual_id_number":757142,"owner":{"_account_id":8866,"name":"Raildo Mascena de Sousa Filho","email":"rmascena@redhat.com","username":"raildo"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8866,"name":"Raildo Mascena de Sousa Filho","email":"rmascena@redhat.com","username":"raildo"},{"value":0,"_account_id":28182,"name":"Jiaping LI","email":"lijiaping@yovole.com","username":"lijiaping"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-05 02:52:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-01-04 18:01:33.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"recommended":{"_account_id":28182,"name":"Jiaping LI","email":"lijiaping@yovole.com","username":"lijiaping"},"all":[{"value":0,"_account_id":8866,"name":"Raildo Mascena de Sousa Filho","email":"rmascena@redhat.com","username":"raildo"},{"value":1,"date":"2020-12-27 10:47:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28182,"name":"Jiaping LI","email":"lijiaping@yovole.com","username":"lijiaping"},{"value":2,"date":"2021-01-04 19:47:16.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2021-01-04 19:48:13.000000000","_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":1,"date":"2020-12-26 04:32:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":8866,"name":"Raildo Mascena de Sousa Filho","email":"rmascena@redhat.com","username":"raildo"},{"value":0,"_account_id":28182,"name":"Jiaping LI","email":"lijiaping@yovole.com","username":"lijiaping"},{"value":1,"date":"2021-01-04 20:25:41.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"_account_id":8866,"name":"Raildo Mascena de Sousa Filho","email":"rmascena@redhat.com","username":"raildo"},{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":28182,"name":"Jiaping LI","email":"lijiaping@yovole.com","username":"lijiaping"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-19 22:31:27.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-20 02:41:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-04 21:10:28.000000000","updated_by":{"_account_id":8866,"name":"Raildo Mascena de Sousa Filho","email":"rmascena@redhat.com","username":"raildo"},"reviewer":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"state":"REVIEWER"},{"updated":"2020-12-18 15:25:13.000000000","updated_by":{"_account_id":8866,"name":"Raildo Mascena de Sousa Filho","email":"rmascena@redhat.com","username":"raildo"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2020-12-27 10:47:07.000000000","updated_by":{"_account_id":28182,"name":"Jiaping LI","email":"lijiaping@yovole.com","username":"lijiaping"},"reviewer":{"_account_id":28182,"name":"Jiaping LI","email":"lijiaping@yovole.com","username":"lijiaping"},"state":"REVIEWER"},{"updated":"2021-01-04 19:48:13.000000000","updated_by":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"reviewer":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"state":"REVIEWER"}],"messages":[{"id":"8351692cf54bff7e758414f39fc214935bd9e7d3","author":{"_account_id":8866,"name":"Raildo Mascena de Sousa Filho","email":"rmascena@redhat.com","username":"raildo"},"date":"2020-10-09 16:00:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"eac5a0d4398c826cd2d3bed3353ed203a4d127c6","author":{"_account_id":8866,"name":"Raildo Mascena de Sousa Filho","email":"rmascena@redhat.com","username":"raildo"},"date":"2020-10-09 16:01:54.000000000","message":"Patch Set 1:\n\nWaiting for https://github.com/saltedsignal/puppet-certmonger/pull/27 to be merged","accounts_in_message":[],"_revision_number":1},{"id":"568bc9a70ea00e75bdd3032045c5bdcb63afffff","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-10-09 16:52:27.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/8f7450c6cbdc4ca586e240e2e3c0ed63 : FAILURE in 49m 48s","accounts_in_message":[],"_revision_number":1},{"id":"1d0d7b9a50997cc873e1c5f48c5f48e3a490c01c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-10 05:23:55.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/33a7fdb4ad62451ea5220fbff40fcc43 : SUCCESS in 3m 49s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ab453eb3184e462a89e79b06682fd642 : SUCCESS in 3m 28s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/ec31ecf9069f4e8fba6ea939875d3996 : FAILURE in 27m 03s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3418a76974ea469ea27e5c0ca0317313 : FAILURE in 22m 42s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3407a0ff96244ae9974ad289b60f5c8f : FAILURE in 30m 25s (non-voting)\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/943fd061281548ee9f7e19ebe9fdfdb9 : FAILURE in 43m 05s\n- tripleo-ci-centos-8-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/419c4148e209449fa2afabbd8abbf937 : FAILURE in 1h 25m 10s (non-voting)\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/b9931aae281d49f798a488ac3332955a : SUCCESS in 2h 32m 57s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/513a52c6cd5143cfa1637bb1fd6b5d0f : SUCCESS in 1h 57m 05s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/46af703de1d64d9cb39c732001b0868a : SUCCESS in 2h 10m 22s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/44a3402ab16545d9894c4dec644f0255 : SUCCESS in 1h 47m 19s\n- tripleo-ci-centos-8-scenario001-standalone https://zuul.opendev.org/t/openstack/build/fa58642f33e54c6c90eb36ab4910885e : SUCCESS in 2h 07m 36s\n- tripleo-ci-centos-8-standalone-upgrade https://zuul.opendev.org/t/openstack/build/10a6ef4e5cca46cc88b27c628a22d411 : FAILURE in 2h 12m 35s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"35bb744e882af875532926ab693fc3edc9c5c0e8","author":{"_account_id":8866,"name":"Raildo Mascena de Sousa Filho","email":"rmascena@redhat.com","username":"raildo"},"date":"2020-11-03 13:29:58.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"9f82d533de00a3d3079be7060cdbacb0133998c9","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-11-03 14:20:15.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/a0fd210514eb4563b812fab0a05a9c08 : FAILURE in 49m 21s","accounts_in_message":[],"_revision_number":2},{"id":"8838d42b102bea684653f4aef0706a5b0eb8e5d6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-03 22:46:04.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/049f1ac8d63f4b659223098ac8f964a9 : SUCCESS in 4m 04s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ccb67a9e454040e7ad04b480aea4a8f4 : SUCCESS in 4m 50s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/75e554ee5e574469be2939b1430d4a21 : FAILURE in 19m 33s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7ae5a160fb354843a44eff9c1df2d9b7 : FAILURE in 25m 09s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fd7f6e75ad154429aeb9b16159a5c20b : FAILURE in 27m 46s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/c777128eeb7240f492986cbdff901916 : SUCCESS in 9h 12m 12s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/c0960c6bb0de40e38f0e9fddd52ffcdd : SUCCESS in 1h 22m 13s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/fed993f6fd874f50aeec588820a02805 : SUCCESS in 42m 13s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/f84ab8938a424e12a358531b27836c22 : FAILURE in 34m 38s\n- tripleo-ci-centos-8-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/f0acab6ad2944c1b8ab7f1498ace35fb : FAILURE in 21m 46s (non-voting)\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/a404c584350c40669061081a03b92ec7 : SUCCESS in 1h 56m 29s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/6d3f281048ef438bbe9ff02bb049110f : FAILURE in 1h 40m 57s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/5283415b159e472ab0473f36bc868032 : SUCCESS in 2h 38m 34s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/ae889b55d9b846a2a1006106f206970d : SUCCESS in 1h 56m 21s\n- tripleo-ci-centos-8-scenario001-standalone https://zuul.opendev.org/t/openstack/build/e8ec656a8f1e4597b59729feb9bdd10c : SUCCESS in 1h 50m 47s\n- tripleo-ci-centos-8-standalone-upgrade https://zuul.opendev.org/t/openstack/build/683d95cba25e4b5c894bcfe8703f50e2 : FAILURE in 21m 00s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d3790e1c0986c4817f0dcd674a33a863d02468f4","author":{"_account_id":8866,"name":"Raildo Mascena de Sousa Filho","email":"rmascena@redhat.com","username":"raildo"},"date":"2020-11-12 13:09:59.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"0661df9e9e3baedea45766c614d37cf3c94ffc93","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-11-12 13:59:57.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/13acf49f6bf74b3f9f104b202eaa84c5 : FAILURE in 48m 56s","accounts_in_message":[],"_revision_number":3},{"id":"ce0793e37a1df14823e3d483aac1f0f573982dfb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-12 17:51:07.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d7655be9816c4d8a9b677d30b7543c7e : SUCCESS in 4m 08s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3049089e643b43a98a4a27ecf55f3f04 : SUCCESS in 3m 36s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/7abf42de7af547de883c3a0a98d45ccb : FAILURE in 24m 09s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/611ecec6f5914ddea5b21ba918170a34 : FAILURE in 23m 14s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fdb0670cefc04c88995c2bda185ac669 : FAILURE in 29m 23s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/2082725325804044ab4921ee62b66a29 : SUCCESS in 3h 40m 26s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/c40434cd5f8e41ab94f787a2531505df : SUCCESS in 1h 54m 27s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/b9ee1d81e6374ce6b1b03661364969a6 : TIMED_OUT in 3h 56m 46s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/58864933d46f4147919f78f283ce23bf : FAILURE in 42m 31s\n- tripleo-ci-centos-8-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/e984698f2e7645198bf4c0c0b8f33d61 : FAILURE in 28m 16s (non-voting)\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/7e8c01018ed742a6bbd0ef56eb999417 : SUCCESS in 2h 04m 25s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/d7dbe8a051fa4732ab07c769f904a977 : SUCCESS in 2h 04m 55s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/2923331ef82741fe8a2409659ac0462d : SUCCESS in 2h 17m 06s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/20592715560e4658b2f1524ffefca51e : SUCCESS in 1h 35m 59s\n- tripleo-ci-centos-8-scenario001-standalone https://zuul.opendev.org/t/openstack/build/b1ff0b4a30f54f0181a40a2d2161264b : SUCCESS in 1h 59m 21s\n- tripleo-ci-centos-8-standalone-upgrade https://zuul.opendev.org/t/openstack/build/162e0527f2204368997ed12c0fafc077 : FAILURE in 29m 34s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"ad2a2ac8dcfb1ee8bba7366209f182380f9a574e","author":{"_account_id":8866,"name":"Raildo Mascena de Sousa Filho","email":"rmascena@redhat.com","username":"raildo"},"date":"2020-11-19 21:42:24.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"157e787eb714a1a526283f94fa0eb2f780cf2b63","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-11-19 22:31:27.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/1b594d2672f14e0b817448e55f710024 : FAILURE in 48m 00s","accounts_in_message":[],"_revision_number":4},{"id":"7b378c4048488478a5878748ccc5498cf57fcfe1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-20 02:41:57.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2a32a4259d064ef39674eb4ffbbe554a : SUCCESS in 3m 11s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1704e57bd12f46ffa65efdcf35ba9958 : SUCCESS in 3m 07s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/0dfe9512f1814e4196a6bf9fa92b370f : FAILURE in 24m 04s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e6f1560d593548fcbe2ff143a47ce67c : FAILURE in 22m 31s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ecc4cb3e3e78403ca9658be6d1278a98 : FAILURE in 29m 06s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/71688adb44994c6a8c0b31568941f9ac : FAILURE in 1h 02m 50s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/d3523a813dd44bbdb2605e769e81dd47 : SUCCESS in 1h 10m 28s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/2a13c9ca626d4e8b952ffe1724a91bc7 : SUCCESS in 1h 03m 44s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/80348a1d220d413ea72585d40df6064c : SUCCESS in 55m 46s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-scenario001-standalone https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-standalone-upgrade https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"9a2c92ff7b1023a67fcebd46156942e1ec04a58c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8866,"name":"Raildo Mascena de Sousa Filho","email":"rmascena@redhat.com","username":"raildo"},"date":"2020-11-30 12:16:01.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"1d48b5ed23817d2b3c6ba3cbb21e9e9b4ed26fad","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-11-30 13:23:28.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/e696f656422b4018b899dd7cbb387726 : FAILURE in 17m 05s","accounts_in_message":[],"_revision_number":5},{"id":"88962462906bc5ffdbf9a9447a7fa1c4e1a284ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-30 14:45:59.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e46f7bd093eb41059210d23078a3a6b4 : SUCCESS in 3m 42s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cbc487f037164e358c2ce0fc5a736022 : SUCCESS in 3m 31s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/4be529bb4ef849eab84cce1dd5bb468f : FAILURE in 7m 18s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0511d0192edf47e7a29dd4ecf4a5dd06 : FAILURE in 4m 53s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/54c6cc4c67af49899b2e686d087269b6 : FAILURE in 4m 45s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/e5e2d0a9e808483cbc83513a97592f7d : FAILURE in 13m 17s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/4a6846e080a045fe986b8744548ad9f8 : FAILURE in 8m 31s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/6835a3d311d04ed9894643700e2abb4e : FAILURE in 12m 35s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/4442d37094b54c788b5a884857dbdd80 : FAILURE in 7m 07s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-scenario001-standalone https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-standalone-upgrade https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"a31a2c91577076732dbd3f223dbfca2df5f4f9c8","author":{"_account_id":8866,"name":"Raildo Mascena de Sousa Filho","email":"rmascena@redhat.com","username":"raildo"},"date":"2020-12-06 16:01:43.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"ff0d66c09cd0c6349599d7e388d2a6ade525c156","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-12-06 16:02:26.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/None : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":5},{"id":"1d435df49715eb8fd6c40751d0fecdfaf6e180eb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-06 17:06:30.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3d6f246468cc47bc9aace5397df5d462 : SUCCESS in 3m 49s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/10e8688d203845bea7c7d8be13c946c5 : SUCCESS in 3m 31s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/7f9a5f3fa10b41dca46b0950df88dcb8 : FAILURE in 10m 22s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b0d77583a55d44bf8f7308e401112722 : FAILURE in 7m 51s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/15185a8843ca4dbfbd98771f09dd98df : FAILURE in 8m 02s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/a9a1a33337ca4f8a836d122c556537f8 : FAILURE in 57m 07s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/4c1ec181765e419f892df12ccd8bf9cf : SUCCESS in 1h 02m 20s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/a80cf86a5bc94cec802049a22f973844 : SUCCESS in 54m 09s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/5c3cd0d9699543d5a5bb572415921fa2 : SUCCESS in 46m 19s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-scenario001-standalone https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-standalone-upgrade https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"355430992db8d045d29c42c255eec16f9bd8cfbd","author":{"_account_id":8866,"name":"Raildo Mascena de Sousa Filho","email":"rmascena@redhat.com","username":"raildo"},"date":"2020-12-08 12:13:30.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"41773d596259cbbfde9230bc83398ab1004fe004","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-12-08 13:30:38.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/11b03204fc2c462bae01e72dd07dad0f : FAILURE in 1h 15m 05s","accounts_in_message":[],"_revision_number":5},{"id":"e9e4b562768b211e368823c77ee36ffc5ddea6e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-08 17:19:04.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/44e0ab2a52124914b1090c3af4ca6456 : SUCCESS in 4m 28s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/65a2d747c9714295a71e405dc03660c9 : SUCCESS in 4m 10s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/d0232ec356d440af8dba1a9694705e1d : FAILURE in 8m 22s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ca05b3e3552e455491130a2fbe749ea1 : FAILURE in 10m 27s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6ad270f4672e43a18c5fc4515c7e4b06 : FAILURE in 5m 42s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/fabed8fab67e442ea5832ca44cf7048e : SUCCESS in 4h 00m 31s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/1f0e7a88fe0a4ad69cd977a8262303bc : TIMED_OUT in 3h 57m 06s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/161ac7d25d3243cb9d8dc9cca439df6c : FAILURE in 1h 32m 09s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/736799d4a9814c0887b28544657be52f : SUCCESS in 1h 03m 22s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/10adf4b6a930480d8d424607feaec9fb : FAILURE in 47m 45s\n- tripleo-ci-centos-8-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/b5a2f26c896949c4a0b70c13f6478d57 : SUCCESS in 2h 10m 56s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/c8c1353f37324dcf82407269141b0207 : SUCCESS in 2h 06m 23s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/73cac297c1004323bb136e5fc6c82103 : SUCCESS in 2h 12m 36s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/e7270175e4074ee485f3cbec34131c55 : SUCCESS in 1h 39m 24s\n- tripleo-ci-centos-8-scenario001-standalone https://zuul.opendev.org/t/openstack/build/42ea9c1a109a4b9f8ee59ca47d733476 : SUCCESS in 1h 46m 49s\n- tripleo-ci-centos-8-standalone-upgrade https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"163b3f2cab7a219ff56aaaaffc4877e3da03063a","author":{"_account_id":8866,"name":"Raildo Mascena de Sousa Filho","email":"rmascena@redhat.com","username":"raildo"},"date":"2020-12-14 12:38:20.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"26e5ebb4ee722a966431c8c4ca1e400adeaf3e06","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-12-14 13:36:34.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/6198b31c06bb40749899cce56f5d980e : FAILURE in 56m 15s","accounts_in_message":[],"_revision_number":5},{"id":"dbb4e21d14335aeaa63525cfeacd1c93a2c893ba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-14 16:44:50.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ac8031bfea8d4052956e46ca6468817b : SUCCESS in 4m 09s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bee78e84c2a9403c8dd696dd4798e320 : SUCCESS in 3m 55s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/8e173458024a48569af958ab63c28bf5 : FAILURE in 8m 14s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/80a97dfb7f0247eab43eacd8fe0f7cf1 : FAILURE in 6m 09s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5d7b4cbde4d24bd780c02c6c1c71b332 : FAILURE in 8m 23s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/d335bcbf784a46ada199b93df5fd6851 : TIMED_OUT in 3h 19m 45s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/8272e2b12fa743cc87c7657ad8141997 : SUCCESS in 3h 28m 09s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/3b77f962c5304c94b1c0d2ec80668eeb : SUCCESS in 1h 20m 39s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/a9da58efc8134818945c940ca7b0ec62 : SUCCESS in 1h 03m 57s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-scenario001-standalone https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-standalone-upgrade https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"780bc0bca1a901635b6114341194a135aae63c22","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8866,"name":"Raildo Mascena de Sousa Filho","email":"rmascena@redhat.com","username":"raildo"},"date":"2020-12-15 18:42:12.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"5d2d3487f3b4ca09264efb2cfbcf8f7d2cd98cde","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-12-15 20:00:42.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/cae788d938a7453cbe4a9a6cbc2facf0 : FAILURE in 1h 16m 57s","accounts_in_message":[],"_revision_number":6},{"id":"3db845d6ba24177a7826883b4313f35b13348734","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-16 04:49:02.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3f28f438c0bf4a5cbadd36404c638d85 : SUCCESS in 3m 23s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8b17ac34b5b84b51908a429bd30e42d8 : SUCCESS in 3m 29s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/75502bb3f14a4a4d977ec1846366f124 : FAILURE in 9m 39s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/59a90575dede4f7ebb9d9a767fc160c4 : FAILURE in 6m 15s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2d9baed215be4d8c8b4056a1ef8e7c89 : FAILURE in 6m 35s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/4282c4688b364d738d3b51f3edb91919 : SUCCESS in 4h 26m 58s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/224e6587efd2432998b06a5b4be06505 : SUCCESS in 4h 27m 16s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/223796f0a22c4e27801994f1a84bb335 : SUCCESS in 1h 30m 49s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/b151b5fc26734a33838c3bd219401395 : SUCCESS in 1h 22m 08s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/93833752bf634fb2b44228dbdafbf66b : FAILURE in 53m 15s\n- tripleo-ci-centos-8-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/7670df0f7fc546d3a8dc25aca4465bca : FAILURE in 1h 16m 49s (non-voting)\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/7d1968a6027f4914a7dccf4261117dea : SUCCESS in 2h 07m 09s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/c020ac59df584703922941550aea0d12 : SUCCESS in 1h 59m 30s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/48590c196601468f804d8d6670540bbc : SUCCESS in 2h 07m 26s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/f3c0c0c4dcee435681b375e4f605ad3a : SUCCESS in 1h 47m 30s\n- tripleo-ci-centos-8-scenario001-standalone https://zuul.opendev.org/t/openstack/build/45e49d01a7934b1db9f256cc1826603a : SUCCESS in 1h 52m 52s\n- tripleo-ci-centos-8-standalone-upgrade https://zuul.opendev.org/t/openstack/build/c95feba622194407a0ebe19c36eafd73 : SUCCESS in 2h 32m 27s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"6ae42119b4f27a1ea800a1402d0175adcbaaae6a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8866,"name":"Raildo Mascena de Sousa Filho","email":"rmascena@redhat.com","username":"raildo"},"date":"2020-12-16 11:57:07.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"36be20ad45e0101ecf6f4235458cbf2784ba4724","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-12-16 13:28:04.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/de11262bd1554786aa7e434cf0b441a5 : FAILURE in 53m 28s","accounts_in_message":[],"_revision_number":7},{"id":"cf10df67cceaaaa9675442ef115576cb3c7ffbbf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-17 01:19:37.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d157cf0efe78420f902d0156a60a36e7 : SUCCESS in 3m 19s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/18eb6569531b403eb3f26453132d0560 : SUCCESS in 4m 10s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/c1ac8c498d674a9e905649ee57da457f : SUCCESS in 26m 23s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/81b99229da2641aeb41dfce57d37472f : SUCCESS in 23m 33s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/afdb7ca3a1f44432a3dbed80695bc2ae : SUCCESS in 28m 26s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/ec458590a9b8482aa464f5f1f77b7627 : SUCCESS in 4h 15m 35s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/a63b0a2a5aa84a72a1e982876ebad154 : SUCCESS in 4h 17m 41s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/3904397eaf904920a99fc7845b490c52 : SUCCESS in 1h 17m 32s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/e95f40cce6b6479a8b844b1ea44b75e4 : SUCCESS in 1h 11m 06s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/b7a7826359254c47a8496f7c4993f9a1 : FAILURE in 37m 53s\n- tripleo-ci-centos-8-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/7ead0c47a7f448ccab070ec7f11ea994 : FAILURE in 1h 15m 19s (non-voting)\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/2aec825296134a2a81a3239609dafc36 : SUCCESS in 2h 28m 49s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/b109018a927841d3aadf22ab2765fae7 : SUCCESS in 1h 47m 52s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/5055d5c0ad144189a7836b31492780ad : SUCCESS in 2h 00m 09s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/528ba215f08649f6bd0ebbbfdfcf22bf : SUCCESS in 1h 54m 30s\n- tripleo-ci-centos-8-scenario001-standalone https://zuul.opendev.org/t/openstack/build/91c5eabaf8dd46618ac7e90ee3b41bf5 : SUCCESS in 2h 17m 42s\n- tripleo-ci-centos-8-standalone-upgrade https://zuul.opendev.org/t/openstack/build/7011ac654d4e406f81ff14935ca4929d : SUCCESS in 2h 50m 32s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"70ba6775d8a9699e8a1cf307352b33f6798e33c4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8866,"name":"Raildo Mascena de Sousa Filho","email":"rmascena@redhat.com","username":"raildo"},"date":"2020-12-17 23:22:36.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"933aece84ac8af2f73dd39d7e7f8c492a2282c58","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-12-18 00:07:14.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/f3b1f9216a034286ba6c3b18e1f16ce2 : RETRY_LIMIT in 14m 23s","accounts_in_message":[],"_revision_number":8},{"id":"71f85949d174ca1a7034e6a061d3d5a25e6e9c54","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-18 02:39:18.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7d0e612d80c84a07be6d055f6bc946c1 : SUCCESS in 3m 30s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/78a9c828280141f1a83f94ab52ce6df6 : SUCCESS in 3m 08s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/9a5b683f99af43468c48c80f4fbdb242 : FAILURE in 7m 37s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fc07ddc8e4ed4a0c8cd442480b94fdc6 : FAILURE in 5m 17s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b0a58768ee6440bc87c1e1622b8832b2 : FAILURE in 5m 29s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/61ac09a9cbe740e7ad699a552ac5ebd0 : SUCCESS in 3h 14m 10s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/a111905779924b1f951c29babafd6599 : SUCCESS in 3h 12m 52s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/6d286761b0ac4c86b7a368bb433afcaf : SUCCESS in 57m 26s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/15a09dfcf51c4ab0b255459a8c1793ab : SUCCESS in 47m 22s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/9393949613bc4039976104a4c1f8492a : FAILURE in 44m 44s\n- tripleo-ci-centos-8-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/3c22327094384c85bc5ebeedcb26739c : FAILURE in 1h 03m 24s (non-voting)\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/cc91809d780546a2bd7394ed5c527e9f : SUCCESS in 1h 52m 11s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/67c778aab6e5458295d93060542b9735 : SUCCESS in 1h 29m 15s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/3997382d0317470c89f88001b3bfcf43 : SUCCESS in 1h 42m 41s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/ad5a3f9f314143ad8fb8be31ce396088 : SUCCESS in 1h 16m 45s\n- tripleo-ci-centos-8-scenario001-standalone https://zuul.opendev.org/t/openstack/build/56e57c0977304488aa186ce3c7e85696 : SUCCESS in 1h 31m 54s\n- tripleo-ci-centos-8-standalone-upgrade https://zuul.opendev.org/t/openstack/build/4d99c78c9076485db7a57743e4f591f8 : SUCCESS in 2h 09m 19s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"7156f11bfe72aaa1f383fc6bc645bc6c2b22df52","author":{"_account_id":8866,"name":"Raildo Mascena de Sousa Filho","email":"rmascena@redhat.com","username":"raildo"},"date":"2020-12-18 18:46:50.000000000","message":"Patch Set 8:\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"c4ae8d6d055b13ee432dbf2ece1eb7991839e752","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-12-18 19:42:48.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/6da0f1b7c2214a278b484ab992f26d99 : FAILURE in 53m 04s","accounts_in_message":[],"_revision_number":8},{"id":"e82d7856ce59e9c0b525499a1ec900b4074e627b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-19 00:25:23.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9f566181decf41f084613fbac2a20fd2 : SUCCESS in 4m 04s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/63df57eeace041a4a63295dee11a5f41 : SUCCESS in 3m 42s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/fc63e17847474dd99243b123a91afb5e : SUCCESS in 20m 04s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bc589b056b2744bbb43f8c7b2a9bc433 : SUCCESS in 21m 03s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/36c44ef8502344a38a1053fa6ed31c1a : SUCCESS in 27m 22s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/3afa9636a1b74d02ba801d77bc5f0416 : SUCCESS in 2h 57m 07s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/0a3e0f263a664744807d96f6b527102d : SUCCESS in 2h 57m 07s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/95b3a3abe89c49aaac963084d77d993d : SUCCESS in 1h 17m 22s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/efaf9f79d2354923b3bebbacf8c9bad4 : SUCCESS in 48m 15s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/978f835032484d888b8968f80872b304 : FAILURE in 32m 45s\n- tripleo-ci-centos-8-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/a59c703fd264487a8e78228ea8897b9e : FAILURE in 1h 12m 44s (non-voting)\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/2c462110dbfa4d5bbdf7d30a6ba0b099 : SUCCESS in 1h 44m 44s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/426c0c893a3c4736a7cecabbe22b0146 : SUCCESS in 1h 42m 36s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/84d2e9e96f474f38a09186f746660e0f : SUCCESS in 1h 48m 44s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/7b6fd46ea936469689a16095383b3094 : SUCCESS in 1h 13m 38s\n- tripleo-ci-centos-8-scenario001-standalone https://zuul.opendev.org/t/openstack/build/37ed331158b340da9c03f58a21b489e1 : SUCCESS in 1h 31m 39s\n- tripleo-ci-centos-8-standalone-upgrade https://zuul.opendev.org/t/openstack/build/3e0f3f8196ed41b0a817fdcd6a2b1116 : FAILURE in 1h 49m 02s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"ffc78b68d79cc85481a7c21b42d33ea2ce555eca","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2020-12-26 04:32:44.000000000","message":"Patch Set 8: Code-Review+1\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"96c800a7706d11a0a1c8155e9420cc5ee39c27c7","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-12-26 05:37:16.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/0eb5b21bbf3a47c698a90f9e7c63fe66 : FAILURE in 1h 01m 43s","accounts_in_message":[],"_revision_number":8},{"id":"f403dfad1c5d0546dee86c511e4523c08dbcef8a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-26 09:56:10.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0bd6f3ba4ecb4bc2812c67a2c2642621 : SUCCESS in 4m 14s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1c9a34c2b1ef4e7da853c88ac9f44536 : SUCCESS in 3m 39s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/1c9674386b96479198914f22fdcec781 : SUCCESS in 24m 00s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9a69b2090aee429aa1af81f7d4f605e1 : SUCCESS in 18m 55s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9ff0466a94cf4a63a3a5c199a7c0bbf8 : SUCCESS in 21m 38s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/e332ad1aa24f4a4f8ca30412fd1abbf6 : SUCCESS in 5h 21m 26s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/308e4e0d36b643fda7564c9f346f8068 : SUCCESS in 5h 21m 40s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/12e2e1f08cec4cff8a27772d9b1853a0 : SUCCESS in 58m 02s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/6dbf6f2906d34f60a51d5fd8a83770d7 : SUCCESS in 51m 26s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/4f9e34153ebc45318dc3cc2676009b7a : FAILURE in 30m 43s\n- tripleo-ci-centos-8-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/a7adfebc8acb4f9eaa6ff9d325476716 : FAILURE in 1h 13m 53s (non-voting)\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/46c7fafd356c4b5bb12e8e613963bef8 : SUCCESS in 1h 49m 48s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/e83ef6970e30405692ea55f5a4a0430d : SUCCESS in 1h 29m 58s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/d74bc8ce86d14caca9e56b6ec2ec509f : SUCCESS in 1h 36m 45s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/bc49c169a5f1464787da6974ae15498e : SUCCESS in 1h 15m 52s\n- tripleo-ci-centos-8-scenario001-standalone https://zuul.opendev.org/t/openstack/build/1aac8cf323064aea8c06e1effed1f273 : SUCCESS in 1h 30m 40s\n- tripleo-ci-centos-8-standalone-upgrade https://zuul.opendev.org/t/openstack/build/c88332c4b88a44b2a79dbb2fe9670a6d : SUCCESS in 2h 24m 08s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"4b7d3f9498202b6a15a2fbf7b6e37fbaca45670d","author":{"_account_id":28182,"name":"Jiaping LI","email":"lijiaping@yovole.com","username":"lijiaping"},"date":"2020-12-27 10:47:07.000000000","message":"Patch Set 8: Code-Review+1","accounts_in_message":[],"_revision_number":8},{"id":"b7aaaec9c1a8797c897acd0d94b73ef2d7f93031","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-01-04 15:16:08.000000000","message":"Patch Set 8:\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"35ae8989346ce7be554c0e9575b5d1769dda9b93","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-01-04 18:01:33.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/170de6a24b664faf9ac9ea1573b6f3ca : SUCCESS in 2h 43m 48s","accounts_in_message":[],"_revision_number":8},{"id":"6cb72f76557326977ba84b684d5f157f4fe14f57","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-04 19:10:06.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8a5f8abb704645faaa412742ed4553ba : SUCCESS in 3m 39s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1217a393618f4e9fafb8f0363d17a8c5 : SUCCESS in 3m 39s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/3e34a6c8779948f2a36eac2348c57d8a : SUCCESS in 25m 59s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/994674bfa0324f288d78d75f8f3963cc : SUCCESS in 24m 34s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/833dbb1ee5cd4105aa93786038988fd8 : SUCCESS in 28m 13s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/0acdcaf5a00a4c598079c0b84ad85a64 : SUCCESS in 3h 31m 50s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/bc4afeeeb25842faaadec0e7e8f7d2f3 : SUCCESS in 3h 28m 01s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/b6e8841ed87d4817a602cfdce11578d5 : SUCCESS in 1h 00m 10s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/86c2eb274e094c0bbe8e9eb131a05908 : SUCCESS in 53m 31s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/b3c298ea928a4ad6913fb2a89e43f6af : SUCCESS in 1h 14m 43s\n- tripleo-ci-centos-8-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/6a4c23fb69ff418ab68879d7cfe91cbc : FAILURE in 2h 04m 03s (non-voting)\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/7031199e00a14f67b968f5e7059fbd7f : SUCCESS in 2h 00m 39s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/f4cfb387f5b34bf3ab34b0c987da2968 : SUCCESS in 1h 49m 15s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/5f587156ce7d406d824eff8f4e684669 : SUCCESS in 2h 08m 45s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/5fe089fd5bd646c9962a14dfbbfea02c : SUCCESS in 1h 26m 58s\n- tripleo-ci-centos-8-scenario001-standalone https://zuul.opendev.org/t/openstack/build/7fcad5054fab4dd38f2a77716b91baa7 : SUCCESS in 1h 50m 53s\n- tripleo-ci-centos-8-standalone-upgrade https://zuul.opendev.org/t/openstack/build/d71abb2080a94ac4b9127e729ed0bb75 : FAILURE in 37m 19s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"28456115a65fa1a275f26f4216444202756f6576","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-01-04 19:47:16.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"807e722ac6e84df2b71194ca290ec4152bfb6f27","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2021-01-04 19:48:13.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"f5ff918f68929bf6ac3dc452564f6a3c4199045a","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-01-04 20:25:41.000000000","message":"Patch Set 8: Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"4de29172113e9a8af80ddb10ce48a86d9aea62fa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-04 20:25:59.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"6a678a9248652585e8fbc3bdfdd60d3d32c366d9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-05 02:52:28.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d8cad39289c848a2a70cb96f67bb1a12 : SUCCESS in 4m 34s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6c4cac5a149743e199df28985932c81f : SUCCESS in 2m 46s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/029ce1403556441bb9cf306045c11e74 : SUCCESS in 26m 04s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/956f6c5580d744bab8d327ad373bee1e : SUCCESS in 23m 05s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/088f95193acd403e985cacfc68d64034 : SUCCESS in 3h 16m 01s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/ee66924a40db469bbd302fba7a2ade51 : SUCCESS in 1h 05m 26s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/cbee14e510884caba23fcbd9e6204c53 : SUCCESS in 1h 05m 54s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/d48ddecd843f44dd99ee32aef51757fc : SUCCESS in 58m 03s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/873376cdb76348aaa8c745da7fb861fa : SUCCESS in 1h 17m 58s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/2feb1bab6f344ad8a75090e27d88f51e : SUCCESS in 2h 00m 18s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/36e38225d8ff4df8980a8b5a2f8eb8e1 : SUCCESS in 1h 34m 37s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/b2850034352d489bacbe7f86dad4a05f : SUCCESS in 1h 53m 35s\n- tripleo-ci-centos-8-scenario001-standalone https://zuul.opendev.org/t/openstack/build/60c40cb74600477e956e1a66ba06de8f : SUCCESS in 1h 57m 15s","accounts_in_message":[],"_revision_number":8},{"id":"3eb878747ef5a90684a6828d4ef2dd2ad8efef32","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-05 02:52:29.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"190aebca609e8ec68586cfa4ced9f2efa65758d1","revisions":{"dd5a6375c7aadde7107174f0029a8176feba9a57":{"kind":"REWORK","_number":1,"created":"2020-10-09 16:00:05.000000000","uploader":{"_account_id":8866,"name":"Raildo Mascena de Sousa Filho","email":"rmascena@redhat.com","username":"raildo"},"ref":"refs/changes/42/757142/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tripleo","ref":"refs/changes/42/757142/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/42/757142/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/42/757142/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/42/757142/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tripleo refs/changes/42/757142/1"}}},"commit":{"parents":[{"commit":"2fbc53ac1ef1ba4a90d0428f7399c07fbd66d252","subject":"Merge \"Make ceph_dashboard haproxy config parametric\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/2fbc53ac1ef1ba4a90d0428f7399c07fbd66d252"}]}],"author":{"name":"Raildo","email":"rmascena@redhat.com","date":"2020-10-09 13:52:26.000000000","tz":-180},"committer":{"name":"Raildo","email":"rmascena@redhat.com","date":"2020-10-09 15:59:11.000000000","tz":-180},"subject":"[WIP]Adding key_size option on the certmonger_certificate function","message":"[WIP]Adding key_size option on the certmonger_certificate function\n\ncertmonger_certificate function currently does not support\ncreating certificates with private keys stronger than 2048bits.\nAdding a key_size option.\n\nkey_size option were added on puppet_certmonger on:\nhttps://github.com/saltedsignal/puppet-certmonger/pull/27\n\nChange-Id: I4da96f2164cf1d136f9471f1d6251bdd8cfd2d0b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/dd5a6375c7aadde7107174f0029a8176feba9a57"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/dd5a6375c7aadde7107174f0029a8176feba9a57"}]},"branch":"refs/heads/master"},"2d7f946223a864a1fb5b271eb9b1171ee7c451b0":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":2,"created":"2020-11-03 13:29:58.000000000","uploader":{"_account_id":8866,"name":"Raildo Mascena de Sousa Filho","email":"rmascena@redhat.com","username":"raildo"},"ref":"refs/changes/42/757142/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tripleo","ref":"refs/changes/42/757142/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/42/757142/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/42/757142/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/42/757142/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tripleo refs/changes/42/757142/2"}}},"commit":{"parents":[{"commit":"e6885a4ee0937255b29bf5bc18afb34a449e615a","subject":"Bump metadata to 13.4.0 preparing to create stable/victoria branch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/e6885a4ee0937255b29bf5bc18afb34a449e615a"}]}],"author":{"name":"Raildo","email":"rmascena@redhat.com","date":"2020-10-09 13:52:26.000000000","tz":-180},"committer":{"name":"Raildo","email":"rmascena@redhat.com","date":"2020-11-03 13:29:39.000000000","tz":-180},"subject":"Adding key_size option on the certmonger_certificate function","message":"Adding key_size option on the certmonger_certificate function\n\ncertmonger_certificate function currently does not support\ncreating certificates with private keys stronger than 2048bits.\nAdding a key_size option.\n\nkey_size option were added on puppet_certmonger on:\nhttps://github.com/saltedsignal/puppet-certmonger/pull/27\n\nChange-Id: I4da96f2164cf1d136f9471f1d6251bdd8cfd2d0b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/2d7f946223a864a1fb5b271eb9b1171ee7c451b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/2d7f946223a864a1fb5b271eb9b1171ee7c451b0"}]},"branch":"refs/heads/master"},"e42db8a030cd74b0740db36f1208dd188cfd22c4":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-11-12 13:09:59.000000000","uploader":{"_account_id":8866,"name":"Raildo Mascena de Sousa Filho","email":"rmascena@redhat.com","username":"raildo"},"ref":"refs/changes/42/757142/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tripleo","ref":"refs/changes/42/757142/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/42/757142/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/42/757142/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/42/757142/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tripleo refs/changes/42/757142/3"}}},"commit":{"parents":[{"commit":"6fd840642f7e42dc891ce103c2fdb178d7ffbd9a","subject":"Merge \"Generic cluster properties support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/6fd840642f7e42dc891ce103c2fdb178d7ffbd9a"}]}],"author":{"name":"Raildo","email":"rmascena@redhat.com","date":"2020-10-09 13:52:26.000000000","tz":-180},"committer":{"name":"Raildo Mascena","email":"rmascena@redhat.com","date":"2020-11-12 13:09:59.000000000","tz":0},"subject":"Adding key_size option on the certmonger_certificate function","message":"Adding key_size option on the certmonger_certificate function\n\ncertmonger_certificate function currently does not support\ncreating certificates with private keys stronger than 2048bits.\nAdding a key_size option.\n\nkey_size option were added on puppet_certmonger on:\nhttps://github.com/saltedsignal/puppet-certmonger/pull/27\n\nChange-Id: I4da96f2164cf1d136f9471f1d6251bdd8cfd2d0b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/e42db8a030cd74b0740db36f1208dd188cfd22c4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/e42db8a030cd74b0740db36f1208dd188cfd22c4"}]},"branch":"refs/heads/master"},"4bd4972126c09a32d6fc8cc8635759500e560133":{"kind":"REWORK","_number":4,"created":"2020-11-19 21:42:24.000000000","uploader":{"_account_id":8866,"name":"Raildo Mascena de Sousa Filho","email":"rmascena@redhat.com","username":"raildo"},"ref":"refs/changes/42/757142/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tripleo","ref":"refs/changes/42/757142/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/42/757142/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/42/757142/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/42/757142/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tripleo refs/changes/42/757142/4"}}},"commit":{"parents":[{"commit":"e6885a4ee0937255b29bf5bc18afb34a449e615a","subject":"Bump metadata to 13.4.0 preparing to create stable/victoria branch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/e6885a4ee0937255b29bf5bc18afb34a449e615a"}]}],"author":{"name":"Raildo","email":"rmascena@redhat.com","date":"2020-10-09 13:52:26.000000000","tz":-180},"committer":{"name":"Raildo","email":"rmascena@redhat.com","date":"2020-11-19 21:42:10.000000000","tz":-180},"subject":"Adding key_size option on the certmonger_certificate function","message":"Adding key_size option on the certmonger_certificate function\n\ncertmonger_certificate function currently does not support\ncreating certificates with private keys stronger than 2048bits.\nAdding a key_size option.\n\nkey_size option were added on puppet_certmonger on:\nhttps://github.com/saltedsignal/puppet-certmonger/pull/27\n\nChange-Id: I4da96f2164cf1d136f9471f1d6251bdd8cfd2d0b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/4bd4972126c09a32d6fc8cc8635759500e560133"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/4bd4972126c09a32d6fc8cc8635759500e560133"}]},"branch":"refs/heads/master"},"7cc3ef8ba410ba4e3ad17fc0a47269d0aa4575e5":{"kind":"REWORK","_number":5,"created":"2020-11-30 12:16:01.000000000","uploader":{"_account_id":8866,"name":"Raildo Mascena de Sousa Filho","email":"rmascena@redhat.com","username":"raildo"},"ref":"refs/changes/42/757142/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tripleo","ref":"refs/changes/42/757142/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/42/757142/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/42/757142/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/42/757142/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tripleo refs/changes/42/757142/5"}}},"commit":{"parents":[{"commit":"eaca38aa674f263ec2a95add28d645b11525e839","subject":"Filter haproxy_certificate_specs if hostname is empty","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/eaca38aa674f263ec2a95add28d645b11525e839"}]}],"author":{"name":"Raildo","email":"rmascena@redhat.com","date":"2020-10-09 13:52:26.000000000","tz":-180},"committer":{"name":"Raildo","email":"rmascena@redhat.com","date":"2020-11-30 12:15:04.000000000","tz":-180},"subject":"Adding key_size option on the certmonger_certificate function","message":"Adding key_size option on the certmonger_certificate function\n\ncertmonger_certificate function currently does not support\ncreating certificates with private keys stronger than 2048bits.\nAdding a key_size option.\n\nkey_size option were added on puppet_certmonger on the v2.4.0\nupstream: https://github.com/saltedsignal/puppet-certmonger/releases/tag/v2.4.0\n\nChange-Id: I4da96f2164cf1d136f9471f1d6251bdd8cfd2d0b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/7cc3ef8ba410ba4e3ad17fc0a47269d0aa4575e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/7cc3ef8ba410ba4e3ad17fc0a47269d0aa4575e5"}]},"branch":"refs/heads/master"},"4d26c81f5d917b70fca3df5e810b07fdc3d72c86":{"kind":"REWORK","_number":6,"created":"2020-12-15 18:42:12.000000000","uploader":{"_account_id":8866,"name":"Raildo Mascena de Sousa Filho","email":"rmascena@redhat.com","username":"raildo"},"ref":"refs/changes/42/757142/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tripleo","ref":"refs/changes/42/757142/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/42/757142/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/42/757142/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/42/757142/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tripleo refs/changes/42/757142/6"}}},"commit":{"parents":[{"commit":"eaca38aa674f263ec2a95add28d645b11525e839","subject":"Filter haproxy_certificate_specs if hostname is empty","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/eaca38aa674f263ec2a95add28d645b11525e839"}]}],"author":{"name":"Raildo","email":"rmascena@redhat.com","date":"2020-10-09 13:52:26.000000000","tz":-180},"committer":{"name":"Raildo Mascena de Sousa Filho","email":"rmascena@redhat.com","date":"2020-12-15 18:41:45.000000000","tz":-180},"subject":"Adding key_size option on the certmonger_certificate function","message":"Adding key_size option on the certmonger_certificate function\n\ncertmonger_certificate function currently does not support\ncreating certificates with private keys stronger than 2048bits.\nAdding a key_size option.\n\nkey_size option were added on puppet_certmonger on the v2.6.0\nupstream: https://github.com/saltedsignal/puppet-certmonger/releases/tag/v2.6.0\n\nChange-Id: I4da96f2164cf1d136f9471f1d6251bdd8cfd2d0b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/4d26c81f5d917b70fca3df5e810b07fdc3d72c86"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/4d26c81f5d917b70fca3df5e810b07fdc3d72c86"}]},"branch":"refs/heads/master"},"ef92ea221d2d083a8b4deb16efea198a087571ea":{"kind":"REWORK","_number":7,"created":"2020-12-16 11:57:07.000000000","uploader":{"_account_id":8866,"name":"Raildo Mascena de Sousa Filho","email":"rmascena@redhat.com","username":"raildo"},"ref":"refs/changes/42/757142/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tripleo","ref":"refs/changes/42/757142/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/42/757142/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/42/757142/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/42/757142/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tripleo refs/changes/42/757142/7"}}},"commit":{"parents":[{"commit":"eaca38aa674f263ec2a95add28d645b11525e839","subject":"Filter haproxy_certificate_specs if hostname is empty","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/eaca38aa674f263ec2a95add28d645b11525e839"}]}],"author":{"name":"Raildo","email":"rmascena@redhat.com","date":"2020-10-09 13:52:26.000000000","tz":-180},"committer":{"name":"Raildo Mascena de Sousa Filho","email":"rmascena@redhat.com","date":"2020-12-16 11:56:57.000000000","tz":-180},"subject":"Adding key_size option on the certmonger_certificate function","message":"Adding key_size option on the certmonger_certificate function\n\ncertmonger_certificate function currently does not support\ncreating certificates with private keys stronger than 2048bits.\nAdding a key_size option.\n\nkey_size option were added on puppet_certmonger on the v2.6.0\nupstream: https://github.com/saltedsignal/puppet-certmonger/releases/tag/v2.6.0\n\nChange-Id: I4da96f2164cf1d136f9471f1d6251bdd8cfd2d0b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/ef92ea221d2d083a8b4deb16efea198a087571ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/ef92ea221d2d083a8b4deb16efea198a087571ea"}]},"branch":"refs/heads/master"},"190aebca609e8ec68586cfa4ced9f2efa65758d1":{"kind":"REWORK","_number":8,"created":"2020-12-17 23:22:36.000000000","uploader":{"_account_id":8866,"name":"Raildo Mascena de Sousa Filho","email":"rmascena@redhat.com","username":"raildo"},"ref":"refs/changes/42/757142/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tripleo","ref":"refs/changes/42/757142/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/42/757142/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/42/757142/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/42/757142/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tripleo refs/changes/42/757142/8"}}},"commit":{"parents":[{"commit":"eaca38aa674f263ec2a95add28d645b11525e839","subject":"Filter haproxy_certificate_specs if hostname is empty","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/eaca38aa674f263ec2a95add28d645b11525e839"}]}],"author":{"name":"Raildo","email":"rmascena@redhat.com","date":"2020-10-09 13:52:26.000000000","tz":-180},"committer":{"name":"Raildo Mascena de Sousa Filho","email":"rmascena@redhat.com","date":"2020-12-17 22:42:02.000000000","tz":-180},"subject":"Adding key_size option on the certmonger_certificate function","message":"Adding key_size option on the certmonger_certificate function\n\ncertmonger_certificate function currently does not support\ncreating certificates with private keys stronger than 2048bits.\nAdding a key_size option.\n\nkey_size option were added on puppet_certmonger on the v2.6.0\nupstream: https://github.com/saltedsignal/puppet-certmonger/releases/tag/v2.6.0\n\nChange-Id: I4da96f2164cf1d136f9471f1d6251bdd8cfd2d0b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/190aebca609e8ec68586cfa4ced9f2efa65758d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/190aebca609e8ec68586cfa4ced9f2efa65758d1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
