)]}'
{"id":"openstack%2Fpuppet-tripleo~779784","triplet_id":"openstack%2Fpuppet-tripleo~stable%2Ftrain~I3db9cd42aa63379ed6332ee77f67d1dfe4be8504","project":"openstack/puppet-tripleo","branch":"stable/train","topic":"1926602","hashtags":[],"change_id":"I3db9cd42aa63379ed6332ee77f67d1dfe4be8504","subject":"Fix live-migration with libvirt \u003e\u003d 6.8.0","status":"MERGED","created":"2021-03-10 15:40:43.000000000","updated":"2021-03-18 22:51:50.000000000","submitted":"2021-03-18 22:48:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"779784-1926602","cherry_pick_of_change":779783,"cherry_pick_of_patch_set":1,"meta_rev_id":"a9fb8187b8958a9abf8cfcae4929c3f9964495f7","_number":779784,"virtual_id_number":779784,"owner":{"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-18 22:48:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-03-10 19:21:34.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"recommended":{"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},"all":[{"value":1,"date":"2021-03-17 10:55:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},{"value":2,"date":"2021-03-18 14:51:17.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},{"value":1,"date":"2021-03-18 14:51:17.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"}],"reviewers":{"REVIEWER":[{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-10 15:40:43.000000000","updated_by":{"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},"reviewer":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"state":"REVIEWER"},{"updated":"2021-03-10 15:40:43.000000000","updated_by":{"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},"reviewer":{"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},"state":"REVIEWER"},{"updated":"2021-03-10 15:40:43.000000000","updated_by":{"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2021-03-10 15:40:43.000000000","updated_by":{"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},"reviewer":{"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},"state":"REVIEWER"},{"updated":"2021-03-10 15:40:43.000000000","updated_by":{"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-10 15:40:43.000000000","updated_by":{"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-10 15:40:43.000000000","updated_by":{"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},"reviewer":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"state":"REVIEWER"}],"messages":[{"id":"7aa0451aaf9ad4334e6a15c923f081a3031f4bf4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},"date":"2021-03-10 15:40:43.000000000","message":"Patch Set 1: Cherry Picked from branch stable/ussuri.","accounts_in_message":[],"_revision_number":1},{"id":"27fd94972fe98bb63ec89ac8ee3dfb6076ae6b59","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-03-10 19:21:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/0c60eac9dd304b1e945bf5b514b1552e : SUCCESS in 3h 05m 08s","accounts_in_message":[],"_revision_number":1},{"id":"bc4ab1550f1223c2b052827481da85eebc3607f8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-10 23:19:50.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/259f6fc3027b4da388b01942733c34ba : SUCCESS in 5m 31s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/71d50261bea043f09815abc15caa9c4e : SUCCESS in 5m 10s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ca6a9d6492614e0698c90a092ae5aff8 : SUCCESS in 4m 50s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/9a3e63d4767d474f82ff24df0cac7417 : FAILURE in 6m 46s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/65866f824bee433ab5ad6a4170a8af64 : SUCCESS in 34m 49s (non-voting)\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/d8adc7c958f94c10baacc7c298a5ef1e : SUCCESS in 1h 42m 19s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/56876afd55e54610926052885bd3bc24 : SUCCESS in 3h 38m 59s\n- tripleo-ci-centos-7-content-provider https://zuul.opendev.org/t/openstack/build/d33938973cc44044b9a5ba9f0eafd2a0 : SUCCESS in 3h 38m 13s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/f29cb8dc68254ce3baaf1d5ec081a2ab : SUCCESS in 2h 12m 31s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/709fd409b36c4bfb85ae3a127fe29968 : SUCCESS in 2h 12m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/17925b22a7494682b754feba06cdcb7f : SUCCESS in 5m 43s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/4195f35a4a5843e68b983d731428448b : SUCCESS in 1h 26m 52s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-train https://zuul.opendev.org/t/openstack/build/fd11b74a7bad4c5b9ffacae502461bb1 : SUCCESS in 2h 00m 39s","accounts_in_message":[],"_revision_number":1},{"id":"58e43e80481556177b3365ba2ec698c27b1b9293","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-03-11 14:55:40.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"72120ee0c84f94f3d3dab90907b74809f1e6f21c","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-03-11 17:51:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/bdd5c3fa0d0e499a8496abb1fa7130c6 : SUCCESS in 2h 52m 32s","accounts_in_message":[],"_revision_number":1},{"id":"80e2554d89930a662b4c95e682fb14273d13a13a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-11 19:01:23.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6537a5df8eda4e02b620babe6b646e5c : SUCCESS in 7m 23s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e132ec12ae6c4004bbe6bc32b2eb285c : SUCCESS in 5m 13s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7653ad7f8dd24db1aabc3f0d9cc13afa : SUCCESS in 4m 14s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/b054dc124ed04cbcb44f69ac61c55717 : FAILURE in 8m 53s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8f3279ad0cd7482b969238713553f4ff : SUCCESS in 38m 45s (non-voting)\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/c6bfe8a72e804fc7925f2ceac0989f04 : SUCCESS in 1h 15m 45s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/c6bfb58f44d448fe945b4462cfd33f1c : FAILURE in 1h 54m 12s\n- tripleo-ci-centos-7-content-provider https://zuul.opendev.org/t/openstack/build/ab2fc1d27d374a3ca0d6463f0fedb656 : SUCCESS in 3h 37m 47s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/0a4d7ec299db46fe986257f93b16cd3d : SUCCESS in 2h 13m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/160fbc7157dd4027baecbbd0df65b912 : SUCCESS in 4m 02s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-train https://zuul.opendev.org/t/openstack/build/None : SKIPPED","accounts_in_message":[],"_revision_number":1},{"id":"a222b872ef907bdccb36077cca9616ee7b968177","author":{"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},"date":"2021-03-17 06:37:17.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"e5a66650690800be7ae7326756b824a030f937b6","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-03-17 09:35:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/0d0b7acdbabe44b186e190cbfa0c8116 : SUCCESS in 2h 57m 04s","accounts_in_message":[],"_revision_number":1},{"id":"97373a9ba9c9e78cc6000a535f835cfdc01b266a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-17 10:51:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0e03d0585e2a47eb9a85142aeda0a8db : SUCCESS in 6m 13s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cc56ac090df144b28adc231b5c2ce42a : SUCCESS in 4m 47s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/98d81e6afbeb40ec820a7bfb988b2b9e : SUCCESS in 4m 58s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/6eef5f08c59b47a48aa50f91e0525a56 : SUCCESS in 47m 44s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/aa58c1e41db44c0dbcdbe406a8187f62 : SUCCESS in 30m 27s (non-voting)\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/156e33da68204567bb3f2a46dd40c659 : SUCCESS in 1h 34m 39s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/1d8410d354cd47da92408a71dd85214a : SUCCESS in 3h 23m 59s\n- tripleo-ci-centos-7-content-provider https://zuul.opendev.org/t/openstack/build/93eeb4c9f23b4f0699c652234421c2b5 : SUCCESS in 3h 57m 01s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/6e9127b26b0f40eea4de4323f169412d : SUCCESS in 2h 10m 27s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/955b7d3073314c238f37fa0fe687c4c3 : SUCCESS in 2h 29m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5a5b243fa1bf41c0a0a22fa8a4186fb2 : SUCCESS in 4m 09s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/186abb0a38414b4182470580b5148a33 : SUCCESS in 1h 35m 22s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-train https://zuul.opendev.org/t/openstack/build/ce5cc3a6dbf94c3e859cc840121b0b21 : SUCCESS in 2h 06m 35s","accounts_in_message":[],"_revision_number":1},{"id":"53383a9a4250866bfcd6432ba5a2edf0f096bfda","author":{"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},"date":"2021-03-17 10:55:11.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"92adafaf7ae45b2569e1ede31432e2bde1cff74a","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-03-18 14:51:17.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3a124ada123e59093b219879d2600b952e8a10d4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-18 14:55:17.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3410d1fb8c5bf0765c063ccdc3d3457e7c888cc5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-18 22:48:10.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b4bb693f38ce4fe1b0a4a21e1ebb2014 : SUCCESS in 5m 44s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/52024c65265f4f9fa43d1a1f2815d666 : SUCCESS in 4m 43s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1705e5b9b7ad4180887b31da50c25371 : SUCCESS in 5m 23s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/f462f0a55bc04338be17eacedcc5ad52 : SUCCESS in 43m 43s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/1417d9f16d264c6385b3ca75e3502a1a : SUCCESS in 1h 45m 13s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/5f86c6919d15494f9033bcb6c3393897 : SUCCESS in 3h 36m 07s\n- tripleo-ci-centos-7-content-provider https://zuul.opendev.org/t/openstack/build/2b5473ba436847e785f096bb840bbea4 : SUCCESS in 4h 21m 57s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/52853e352d314fcf90fead2b1f756bd4 : SUCCESS in 2h 12m 14s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/ba51d6f77459425ca417cee1c49d7ce0 : SUCCESS in 2h 34m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/be723b07ab644409a4af08cd6fa42312 : SUCCESS in 5m 15s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/f1065a6decf64d27a37ccb614e4c73af : SUCCESS in 1h 16m 19s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-train https://zuul.opendev.org/t/openstack/build/6b07f2fd37aa42adbc3b9d031b8b0120 : SUCCESS in 2h 04m 36s","accounts_in_message":[],"_revision_number":1},{"id":"bde66b8393c741fe85341c4e6362b619ac889f0e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-18 22:48:11.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"a9fb8187b8958a9abf8cfcae4929c3f9964495f7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-18 22:51:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8ff8e9e04c7e466a81de1733b9a5e3d3 : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c765a81ea4ed420bbde9bcfe335683a3f0cdfd46","revisions":{"c765a81ea4ed420bbde9bcfe335683a3f0cdfd46":{"kind":"REWORK","_number":1,"created":"2021-03-10 15:40:43.000000000","uploader":{"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},"ref":"refs/changes/84/779784/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tripleo","ref":"refs/changes/84/779784/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/84/779784/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/84/779784/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/84/779784/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tripleo refs/changes/84/779784/1"}}},"commit":{"parents":[{"commit":"8d5bb4b8948ecdef0a0659e4b6bb10756b7cc1bc","subject":"Add ability to specify memcached port","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/8d5bb4b8948ecdef0a0659e4b6bb10756b7cc1bc"}]}],"author":{"name":"Martin Schuppert","email":"mschuppert@redhat.com","date":"2021-03-08 17:45:24.000000000","tz":60},"committer":{"name":"David Vallee Delisle","email":"dvd@redhat.com","date":"2021-03-10 15:40:43.000000000","tz":0},"subject":"Fix live-migration with libvirt \u003e\u003d 6.8.0","message":"Fix live-migration with libvirt \u003e\u003d 6.8.0\n\nlibvirt 6.8.0 introduces virt-ssh-helper which prepends the libvirt\nssh command with a \"which virt-ssh-helper\". libvirt used to first\ncheck for `nc` (netcat). But these two libvirt commits[1][2] have now\nchanged it to first look for `virt-ssh-helper`, if it not available,\nthen fall back to `nc`.\n\nThe nova-migration-wrapper doesn\u0027t accept this command and denies\nthe connection.\n\nUntil nova-migration-wrapper is changed to allow the usage of\nvirt-ssh-helper, this change force to use \"netcat\" (`nc`) by\nappending to the migration URI: \"\u0026proxy\u003dnetcat\"\n\n[1] https://libvirt.org/git/?p\u003dlibvirt.git;a\u003dcommit;h\u003df8ec7c842d (rpc:\nuse new virt-ssh-helper binary for remote tunnelling, 2020-07-08)\n[2] https://libvirt.org/git/?p\u003dlibvirt.git;a\u003dcommit;h\u003d7d959c302d (rpc:\nFix virt-ssh-helper detection, 2020-10-27)\n\nCloses-Bug: 1918250\n\nRelated: https://bugzilla.redhat.com/show_bug.cgi?id\u003d1926602\n\nChange-Id: I3db9cd42aa63379ed6332ee77f67d1dfe4be8504\nCo-Authored-By: David Vallee Delisle \u003cdvd@redhat.com\u003e\n(cherry picked from commit 04a97f92e4d944ce51492011584e2ec1126042a1)\n(cherry picked from commit bf5a4a9849d7cb067d36ca57fd8538faed016715)\n(cherry picked from commit 960dbfe986b06a9b390b4adac1fca57c8badb3e7)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/c765a81ea4ed420bbde9bcfe335683a3f0cdfd46"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/c765a81ea4ed420bbde9bcfe335683a3f0cdfd46"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
