)]}'
{"id":"openstack%2Fpuppet-tripleo~787771","triplet_id":"openstack%2Fpuppet-tripleo~master~I5192f8c306422d0966eef7544f719cc21172bd92","project":"openstack/puppet-tripleo","branch":"master","topic":"virtlogd_wrapper","hashtags":[],"change_id":"I5192f8c306422d0966eef7544f719cc21172bd92","subject":"Introduce nova virtlogd wrapper","status":"MERGED","created":"2021-04-23 15:11:10.000000000","updated":"2021-05-05 12:29:18.000000000","submitted":"2021-05-05 12:27:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":12,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"787771-virtlogd_wrapper","meta_rev_id":"2144cbf6362bfea3c4f8056709fc3293c4f36f16","_number":787771,"virtual_id_number":787771,"owner":{"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},{"value":0,"_account_id":23811,"name":"Oliver Walsh","email":"owalsh@redhat.com","username":"owalsh"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-05 12:27:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"value":1,"date":"2021-04-29 12:16:37.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},{"value":0,"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"recommended":{"_account_id":23811,"name":"Oliver Walsh","email":"owalsh@redhat.com","username":"owalsh"},"all":[{"value":0,"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},{"value":1,"date":"2021-05-04 11:46:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23811,"name":"Oliver Walsh","email":"owalsh@redhat.com","username":"owalsh"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-05-05 08:33:24.000000000","_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":1,"date":"2021-04-29 11:44:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},{"value":2,"date":"2021-04-29 13:43:50.000000000","_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"all":[{"value":0,"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},{"value":0,"_account_id":23811,"name":"Oliver Walsh","email":"owalsh@redhat.com","username":"owalsh"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-05-05 08:33:24.000000000","_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},{"value":0,"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"}],"reviewers":{"REVIEWER":[{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":23811,"name":"Oliver Walsh","email":"owalsh@redhat.com","username":"owalsh"},{"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-23 15:46:08.000000000","updated_by":{"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},"reviewer":{"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},"state":"REVIEWER"},{"updated":"2021-04-23 17:58:37.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-23 22:56:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-26 19:54:19.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2021-04-28 09:09:05.000000000","updated_by":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"reviewer":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"state":"REVIEWER"},{"updated":"2021-04-30 07:18:13.000000000","updated_by":{"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},"reviewer":{"_account_id":23811,"name":"Oliver Walsh","email":"owalsh@redhat.com","username":"owalsh"},"state":"REVIEWER"},{"updated":"2021-04-30 07:18:13.000000000","updated_by":{"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},"reviewer":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"state":"REVIEWER"},{"updated":"2021-04-30 07:18:13.000000000","updated_by":{"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},"reviewer":{"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},"state":"REVIEWER"}],"messages":[{"id":"33619c74eab2181b93833b56762aebf6ec003807","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},"date":"2021-04-23 15:11:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4b340141530139afb38ed6c5ff5698740eac82f6","author":{"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},"date":"2021-04-23 15:11:46.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"4f6ec893d5e974d9b0772e2dc4dd459c30f67979","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-04-23 17:58:37.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/84252e61b9504ccb8fa0da4de94d2ec4 : FAILURE in 2h 26m 14s","accounts_in_message":[],"_revision_number":1},{"id":"10565896994628a5519cfde04d7feebbbb84dd3e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-23 22:56:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c6268a9b0ed3483eb93b962d6325531d : SUCCESS in 3m 54s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6462045b2227418486ba9d457378a408 : SUCCESS in 4m 52s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/848e037a5b824697b4ead9e88349633b : SUCCESS in 21m 17s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/540552f559544cb698ea34c4f696dbb1 : SUCCESS in 18m 32s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/477ce232747746adb648bbe24c1548d6 : SUCCESS in 18m 45s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/8d66121e93b04b65aa43a7908f107f95 : SUCCESS in 2h 52m 49s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/bf438a45d7854a1781d652effde33ecc : SUCCESS in 52m 53s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/5d3f0cc1c84d425ca201e70cc8600de9 : SUCCESS in 1h 45m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c7e9659117ca44b4a72ebf36850c6e54 : SUCCESS in 5m 22s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/6b1dd596940545dfb15977419adfdafa : SUCCESS in 1h 05m 24s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/40192ca02d5844b1a6fe3fa7ce713e07 : SUCCESS in 1h 32m 08s","accounts_in_message":[],"_revision_number":1},{"id":"7b35bb92fdd64ac541754dde1b75f8ee628960ce","author":{"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},"date":"2021-04-26 07:02:10.000000000","message":"Patch Set 1:\n\ncheck-rdo","accounts_in_message":[],"_revision_number":1},{"id":"3d7d8cb1456c90e784db49321892ef1dc5f6ca58","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-04-26 09:35:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/f523d55d9ecf46a7bd24e2dbefa4ca49 : SUCCESS in 2h 32m 09s","accounts_in_message":[],"_revision_number":1},{"id":"c1f03e139dc712fe49fff3f72937f6e89f4756da","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},"date":"2021-04-26 13:30:14.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"41c7694fbd91091fef3d12381e282331a48875d5","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-04-26 15:09:05.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/21651d653f4a4a9d86be7b9e3efb37fe : RETRY_LIMIT in 21m 28s","accounts_in_message":[],"_revision_number":2},{"id":"358d7cf9159e74aa1a855d45bf962bc61419525e","author":{"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},"date":"2021-04-26 15:54:49.000000000","message":"Patch Set 2:\n\ncheck-rdo","accounts_in_message":[],"_revision_number":2},{"id":"60b28992fd5a6ec55c96d1169454d9051f617c78","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-26 17:19:54.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2afafb20a8a740a1a1cc636eeba43ac5 : SUCCESS in 3m 58s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cb55ed9b46934a45a01d5eb03d5f2052 : SUCCESS in 3m 38s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/993477e228e74b61a5433cbee81b6760 : SUCCESS in 23m 26s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/21cbd253a3a84170879c5468f20a4d8c : SUCCESS in 22m 36s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/36a1f6b81ff94ae9a31dae56ab002ba8 : SUCCESS in 21m 02s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/70911cc3fb0749aa9442246dfbd8fdd9 : SUCCESS in 3h 44m 07s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/9d229a6a5791421fa8099d41917a4f53 : SUCCESS in 1h 19m 50s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/327a0444b1124d03802cfa8dedd885a6 : SUCCESS in 1h 58m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6bffc938a65f417383202d567b03cadb : SUCCESS in 4m 05s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/99b5069f399b4f0eb959ae66b8a2f7e9 : SUCCESS in 1h 46m 12s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/1475ece46ca54480be2e12928fcd115e : SUCCESS in 2h 00m 28s","accounts_in_message":[],"_revision_number":2},{"id":"57313c0912951f1d2d0f0321237ff265249cc912","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-04-26 19:18:50.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/642fc89d67a3496c9b7a257500853625 : SUCCESS in 2h 50m 49s","accounts_in_message":[],"_revision_number":2},{"id":"edcc9bebdf91f829757099927a89b9db3ac812fb","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-04-26 19:54:19.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"003ac7cc9f6b766712acfe50e3a37fe6ef2abaf3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},"date":"2021-04-27 05:49:45.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"f68f7a38d196367df2a4bb85b89dcff3c7ed370f","author":{"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},"date":"2021-04-27 05:50:49.000000000","message":"Patch Set 3: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"539eca654be34525d90f8f06cdb6f47816845327","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-27 08:53:28.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8eaa46c35f534a4a93434afd8169004c : SUCCESS in 3m 45s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f75af127e6c04ab3833ac739dd61e07c : SUCCESS in 3m 37s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/7a44ea95629d47a08f49ee5fec046fee : SUCCESS in 31m 58s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f4afea229c464344af7234b5164dc9e7 : SUCCESS in 20m 42s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bf38254e3cca4858b56b0889f1d9eca3 : SUCCESS in 20m 47s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/02a99176dd0a4539865f398b5927af6a : SUCCESS in 3h 01m 51s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/1f0ac1ea4f77448f97647e1b48a98d0f : SUCCESS in 1h 09m 38s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/6ac7506b0de846ab98a724999bce0e0e : SUCCESS in 1h 45m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e91a68e2998b47698a093be8117bbab0 : SUCCESS in 3m 47s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/be16dde160514f6591ee12f59635e43c : SUCCESS in 1h 21m 55s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/1ad660b04f7f41be8768157db6f43577 : SUCCESS in 1h 49m 04s","accounts_in_message":[],"_revision_number":3},{"id":"6d3cc7c7dc5f78c99cc588e037d51fc4c7db2f44","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},"date":"2021-04-27 09:01:33.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"a8027da2061bad5df2df136a4dc2a6520e84e02e","author":{"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},"date":"2021-04-27 09:02:57.000000000","message":"Patch Set 4: Workflow-1","accounts_in_message":[],"_revision_number":4},{"id":"76b091d2c07402d2ddbf087b88a9ef8851b6abf9","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-04-27 12:51:35.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/43872761b7db469396eed8afa28001d9 : SUCCESS in 3h 03m 01s","accounts_in_message":[],"_revision_number":4},{"id":"eb3d6687bcdd734d09d2d3f1c2c57892793a0369","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-27 13:22:28.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d44ec1112d934ab3af43c9788c9181f6 : SUCCESS in 4m 42s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/80ec6914a9204c759ac5608f2bfbd2f3 : SUCCESS in 4m 09s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/d90d231008a5445ab26c852a8c96052f : SUCCESS in 21m 40s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7062d9fff2d04541b61014198f9ee516 : SUCCESS in 21m 15s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8b110680138a4abca67bc098383d7952 : SUCCESS in 21m 47s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/b324a65615a14aa38f4177d204089e11 : SUCCESS in 3h 16m 37s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/245e3e7ace47491fb89c6eba25d8ee93 : SUCCESS in 1h 04m 15s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/0ace374ca2be48c781fd8479b2429b72 : SUCCESS in 2h 03m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4ba40a875a9b43d589746d9ed0eb6c57 : SUCCESS in 5m 39s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/6ad65bf6cf7a4c8fa0b4de45d39fdf10 : SUCCESS in 1h 16m 00s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/cbab778baf2f4ac7bd2f3e34afedb937 : SUCCESS in 1h 34m 03s","accounts_in_message":[],"_revision_number":4},{"id":"bb6a98c044775dcc48a4d37a327a0f599d921cf9","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2021-04-28 09:09:05.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"777ff7976232fb1b3acb6650fcfc2f89c920d737","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},"date":"2021-04-28 10:51:01.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"4a3f864221c538e572beeb6d52296cdc4b659574","author":{"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},"date":"2021-04-28 10:51:22.000000000","message":"Patch Set 5: Workflow-1","accounts_in_message":[],"_revision_number":5},{"id":"186e28e25bfbbc70cdb1c5d95318ba0764da69ff","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-04-28 13:56:43.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/48278146d0a742ffae15314d32b99d8c : SUCCESS in 3h 04m 27s","accounts_in_message":[],"_revision_number":5},{"id":"1abc22b54232041432e1ee91fdb6db43e9b1c914","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-28 15:57:23.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5d2f16dedb3a4f3da30d470d1f32629f : FAILURE in 3m 36s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/12a4405188144f98a114106ca0dfc620 : SUCCESS in 4m 23s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/397f165f605143c0b95f350ded7c6293 : SUCCESS in 21m 45s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/027109e571d74244a313f42030b285e6 : SUCCESS in 17m 00s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/15644ed47cac4a8ebde4d24d3e61dae8 : SUCCESS in 16m 52s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/c4444ea1219642098c7bc48ecf922327 : SUCCESS in 5h 02m 34s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/6c8cdb75a3e14ecc8cd99a8acc302a1f : SUCCESS in 1h 17m 15s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/15d8630f4db340449e42cf6934986289 : SUCCESS in 1h 57m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/aaff6ed7de264e7798cfb876ba43f93e : SUCCESS in 4m 53s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/a9ee91c26f06411dba2580382da80115 : SUCCESS in 1h 13m 05s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/97e9fae8eee847ccbc47715aa02331da : SUCCESS in 2h 38m 26s","accounts_in_message":[],"_revision_number":5},{"id":"707012f08824f848462b7926245db02ff589a667","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},"date":"2021-04-28 16:20:08.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"e592dbfdbed0c8f5475e4dcdb54bfce9ad9dd0a9","author":{"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},"date":"2021-04-28 16:20:21.000000000","message":"Patch Set 6: Workflow-1","accounts_in_message":[],"_revision_number":6},{"id":"79399154c0c45cede4ec552daf05e008c0ef31c0","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-04-28 19:31:27.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/f5308444a96840afbec5eb5c47dce0d9 : SUCCESS in 3h 02m 24s","accounts_in_message":[],"_revision_number":6},{"id":"f110c463bcfbb7b29274d24461eb3c5575e41efe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-28 20:34:48.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f8b7c1f1c468450ab346fb90d4b42690 : SUCCESS in 5m 01s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c38bc6fabd0941f2aaec7a54808710cd : SUCCESS in 4m 26s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/b0f0abb336b04cac833e3189cb3e5a96 : SUCCESS in 25m 51s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/90bbfd1e0e924a85be4de94216f68b7d : SUCCESS in 22m 12s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a516b3d954cc4f91b72675778e6a72e1 : SUCCESS in 21m 53s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/c657f445684d4c72b351ec2e3668624f : SUCCESS in 3h 27m 22s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/41816568c39e420f8e157c353603bc8e : SUCCESS in 1h 21m 21s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/3f6a96fa59ec463d9b5469d9d580ce36 : SUCCESS in 2h 01m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/23919efb3b784add9ac0095e69be420c : SUCCESS in 3m 51s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/6a10deca69f24b3c8c022f44d3882646 : SUCCESS in 1h 35m 34s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/9020caf314aa4363bc28d01fc738bb93 : SUCCESS in 1h 57m 42s","accounts_in_message":[],"_revision_number":6},{"id":"9d3e3b1fb90586ca5661b12949ed0885128dc4b2","author":{"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},"date":"2021-04-29 06:42:11.000000000","message":"Patch Set 6: -Workflow\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"83bac0d9e88f7f642848fc658047180df50833d2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},"date":"2021-04-29 08:14:12.000000000","message":"Uploaded patch set 7: Commit message was updated.","accounts_in_message":[],"_revision_number":7},{"id":"d4d93d8c81cc630a1a6279c4c2d08e3b8662dab6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-29 11:03:33.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/531f5a71c79d4cf986455e28caedf0de : SUCCESS in 3m 24s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8284eea6de164937ba019ee7e231fa85 : SUCCESS in 3m 19s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/b1583c9e1801429190b69f19c43d1331 : SUCCESS in 24m 01s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c332b1ddaba54ef8a4b35c4a25f9818e : SUCCESS in 17m 09s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/df1beab6116b43ab8a5a9ff006f38749 : SUCCESS in 21m 30s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/49dcb9621c8c47f8a917c7f37f8ac3b6 : SUCCESS in 2h 47m 01s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/75c40f1d9fd348e787cc7b188f55a818 : SUCCESS in 1h 05m 01s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/888d0d171c374864afcb41b6c8c330ef : SUCCESS in 1h 41m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c9b8b4bcf14148c98d91c8955db88c26 : SUCCESS in 4m 05s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/5a850acc3a8741a1a4ce720cdbcb0078 : SUCCESS in 1h 16m 20s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/ab36bbc7c6fe40bfa725e254d6cf7732 : SUCCESS in 1h 37m 47s","accounts_in_message":[],"_revision_number":7},{"id":"92b912e95d72a4dda7a232815ad591247e8e3029","author":{"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},"date":"2021-04-29 11:44:09.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"de60b06278849c732ab922ad49959bf68f9332a6","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-04-29 12:16:37.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/17143706bdfc469d82495fd331a66fd2 : SUCCESS in 3h 41m 45s","accounts_in_message":[],"_revision_number":7},{"id":"d5998e28b7630a093a5c5f7bdeb6b0198516c775","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2021-04-29 13:43:50.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"b4dae7f2e096508ffbb5eb7eb385c964431798e3","author":{"_account_id":23811,"name":"Oliver Walsh","email":"owalsh@redhat.com","username":"owalsh"},"date":"2021-05-04 09:30:52.000000000","message":"Patch Set 7:\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"fe68f1d742e4c1b7b5f320037ff7c54897148359","author":{"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},"date":"2021-05-04 09:40:24.000000000","message":"Patch Set 7:\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"3a74a5eabdf39d1ee1de05995faf53cdfc883c09","author":{"_account_id":23811,"name":"Oliver Walsh","email":"owalsh@redhat.com","username":"owalsh"},"date":"2021-05-04 11:18:08.000000000","message":"Patch Set 7:\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"54ce0d531f22c2b40f767da848a185999bec4327","author":{"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},"date":"2021-05-04 11:37:53.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"ab9b3adb110225e82a2ac3de7cdba49649488ee9","author":{"_account_id":23811,"name":"Oliver Walsh","email":"owalsh@redhat.com","username":"owalsh"},"date":"2021-05-04 11:46:13.000000000","message":"Patch Set 7: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"702c876e2e4fd6b50f6dff716f3b1bb18906d885","author":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"date":"2021-05-05 08:33:24.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"2610022bdee7231060114aee55052aa645170a4a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-05 08:34:55.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"324a1f9293d713e1c6563b43e597ad728e53b161","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-05 12:27:46.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aac6a4a2902a451b854a6dc4b62c83b6 : SUCCESS in 5m 21s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/13c10578a07f462a8226b46ff44465af : SUCCESS in 4m 31s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d02282ae9ecf4c67be47027425310e72 : SUCCESS in 4m 39s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/98114164a5944eeda6586e89935feebe : SUCCESS in 23m 39s\n- puppet-openstack-unit-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/ff8ae3b79a554e5194a0f9fed776d02b : SUCCESS in 28m 02s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d665c72b26b545c5a02b094676edeade : SUCCESS in 25m 28s\n- puppet-openstack-unit-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cd59cd5970874885bf6476af3e416f2f : SUCCESS in 23m 40s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/a78c8becb60b4d9f96a58fbef26ee041 : SUCCESS in 3h 47m 13s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/1c194beeb51c4a37900d914fb73243c2 : SUCCESS in 1h 12m 20s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/05d7ef05ed894ad5985971b39f7ba417 : SUCCESS in 2h 01m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c2fe647c026f4c9bac882d59f4d4d8fa : SUCCESS in 5m 44s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/3104a3c81719474ab5ce4459abb339f3 : SUCCESS in 1h 51m 36s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/eda3b94d7ca0404ea0574c7bc6a40e89 : SUCCESS in 1h 56m 13s","accounts_in_message":[],"_revision_number":7},{"id":"b2d4c8f5ae64eaafc594d973ab8686d37d53b2be","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-05 12:27:46.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":7},{"id":"2144cbf6362bfea3c4f8056709fc3293c4f36f16","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-05 12:29:18.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7f210c2f21f84039ad8a9b742dbb37ef : SUCCESS in 49s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"f09aa48d95bc0fada01b1a011bc00e8c9b1e87b0","revisions":{"038bd5d90ee72113fc655526ed901feb8cd852f1":{"kind":"REWORK","_number":1,"created":"2021-04-23 15:11:10.000000000","uploader":{"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},"ref":"refs/changes/71/787771/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tripleo","ref":"refs/changes/71/787771/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/71/787771/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/71/787771/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/71/787771/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tripleo refs/changes/71/787771/1"}}},"commit":{"parents":[{"commit":"98f00851c42f2fed2b09aba5d7dd37da4bfd03cd","subject":"Merge \"Remove redundant import of barbican::keystone::authtoken\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/98f00851c42f2fed2b09aba5d7dd37da4bfd03cd"}]}],"author":{"name":"Martin Schuppert","email":"mschuppert@redhat.com","date":"2021-04-23 14:52:45.000000000","tz":120},"committer":{"name":"Martin Schuppert","email":"mschuppert@redhat.com","date":"2021-04-23 14:52:45.000000000","tz":120},"subject":"Introduce nova virtlogd wrapper","message":"Introduce nova virtlogd wrapper\n\nWhen nova_virtlogd container gets restarted the instance console auth files\nwill not be reopened again by virtlogd. As a result either instances need\nto be restarted or live migrated to a different compute node to get new\nconsole logs messages logged again.\nUsually on receipt of SIGUSR1, virtlogd will re-exec() its binary, while\nmaintaining all current logs and clients. This allows for live upgrades of\nthe virtlogd service on non containerized environments where updates just\nby doing an RPM update.\nTo reduce the likelihood in a containerized environment virtlogd should\nonly be restarted on manual request, or on compute node reboot. It should\nnot be restarted on a minor update without migration off instances.\nThis introduces a nova_virtlogd_wrapper container and virtlogd wrapper\nscript, to only restart virtlogd on either manual or compute node restart.\n\nChange-Id: I5192f8c306422d0966eef7544f719cc21172bd92\nCo-Authored-By: Rajesh Tailor \u003cratailor@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/038bd5d90ee72113fc655526ed901feb8cd852f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/038bd5d90ee72113fc655526ed901feb8cd852f1"}]},"branch":"refs/heads/master"},"962730e919b071bf16b37f7cb976b5ccc6e0abd9":{"kind":"REWORK","_number":2,"created":"2021-04-26 13:30:14.000000000","uploader":{"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},"ref":"refs/changes/71/787771/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tripleo","ref":"refs/changes/71/787771/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/71/787771/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/71/787771/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/71/787771/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tripleo refs/changes/71/787771/2"}}},"commit":{"parents":[{"commit":"98f00851c42f2fed2b09aba5d7dd37da4bfd03cd","subject":"Merge \"Remove redundant import of barbican::keystone::authtoken\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/98f00851c42f2fed2b09aba5d7dd37da4bfd03cd"}]}],"author":{"name":"Martin Schuppert","email":"mschuppert@redhat.com","date":"2021-04-23 14:52:45.000000000","tz":120},"committer":{"name":"Martin Schuppert","email":"mschuppert@redhat.com","date":"2021-04-26 13:30:00.000000000","tz":120},"subject":"Introduce nova virtlogd wrapper","message":"Introduce nova virtlogd wrapper\n\nWhen nova_virtlogd container gets restarted the instance console auth files\nwill not be reopened again by virtlogd. As a result either instances need\nto be restarted or live migrated to a different compute node to get new\nconsole logs messages logged again.\nUsually on receipt of SIGUSR1, virtlogd will re-exec() its binary, while\nmaintaining all current logs and clients. This allows for live upgrades of\nthe virtlogd service on non containerized environments where updates just\nby doing an RPM update.\nTo reduce the likelihood in a containerized environment virtlogd should\nonly be restarted on manual request, or on compute node reboot. It should\nnot be restarted on a minor update without migration off instances.\nThis introduces a nova_virtlogd_wrapper container and virtlogd wrapper\nscript, to only restart virtlogd on either manual or compute node restart.\n\nChange-Id: I5192f8c306422d0966eef7544f719cc21172bd92\nCo-Authored-By: Rajesh Tailor \u003cratailor@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/962730e919b071bf16b37f7cb976b5ccc6e0abd9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/962730e919b071bf16b37f7cb976b5ccc6e0abd9"}]},"branch":"refs/heads/master"},"016af3762adfc80859f1076c673d8ac1355fc309":{"kind":"REWORK","_number":3,"created":"2021-04-27 05:49:45.000000000","uploader":{"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},"ref":"refs/changes/71/787771/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tripleo","ref":"refs/changes/71/787771/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/71/787771/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/71/787771/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/71/787771/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tripleo refs/changes/71/787771/3"}}},"commit":{"parents":[{"commit":"98f00851c42f2fed2b09aba5d7dd37da4bfd03cd","subject":"Merge \"Remove redundant import of barbican::keystone::authtoken\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/98f00851c42f2fed2b09aba5d7dd37da4bfd03cd"}]}],"author":{"name":"Martin Schuppert","email":"mschuppert@redhat.com","date":"2021-04-23 14:52:45.000000000","tz":120},"committer":{"name":"Martin Schuppert","email":"mschuppert@redhat.com","date":"2021-04-27 05:48:58.000000000","tz":120},"subject":"Introduce nova virtlogd wrapper","message":"Introduce nova virtlogd wrapper\n\nWhen nova_virtlogd container gets restarted the instance console auth files\nwill not be reopened again by virtlogd. As a result either instances need\nto be restarted or live migrated to a different compute node to get new\nconsole logs messages logged again.\nUsually on receipt of SIGUSR1, virtlogd will re-exec() its binary, while\nmaintaining all current logs and clients. This allows for live upgrades of\nthe virtlogd service on non containerized environments where updates just\nby doing an RPM update.\nTo reduce the likelihood in a containerized environment virtlogd should\nonly be restarted on manual request, or on compute node reboot. It should\nnot be restarted on a minor update without migration off instances.\nThis introduces a nova_virtlogd_wrapper container and virtlogd wrapper\nscript, to only restart virtlogd on either manual or compute node restart.\n\nChange-Id: I5192f8c306422d0966eef7544f719cc21172bd92\nCo-Authored-By: Rajesh Tailor \u003cratailor@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/016af3762adfc80859f1076c673d8ac1355fc309"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/016af3762adfc80859f1076c673d8ac1355fc309"}]},"branch":"refs/heads/master"},"62066b0ce21d0e26ac158516ee59fe3077b85984":{"kind":"REWORK","_number":4,"created":"2021-04-27 09:01:33.000000000","uploader":{"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},"ref":"refs/changes/71/787771/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tripleo","ref":"refs/changes/71/787771/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/71/787771/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/71/787771/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/71/787771/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tripleo refs/changes/71/787771/4"}}},"commit":{"parents":[{"commit":"98f00851c42f2fed2b09aba5d7dd37da4bfd03cd","subject":"Merge \"Remove redundant import of barbican::keystone::authtoken\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/98f00851c42f2fed2b09aba5d7dd37da4bfd03cd"}]}],"author":{"name":"Martin Schuppert","email":"mschuppert@redhat.com","date":"2021-04-23 14:52:45.000000000","tz":120},"committer":{"name":"Martin Schuppert","email":"mschuppert@redhat.com","date":"2021-04-27 09:00:57.000000000","tz":120},"subject":"Introduce nova virtlogd wrapper","message":"Introduce nova virtlogd wrapper\n\nWhen nova_virtlogd container gets restarted the instance console auth files\nwill not be reopened again by virtlogd. As a result either instances need\nto be restarted or live migrated to a different compute node to get new\nconsole logs messages logged again.\nUsually on receipt of SIGUSR1, virtlogd will re-exec() its binary, while\nmaintaining all current logs and clients. This allows for live upgrades of\nthe virtlogd service on non containerized environments where updates just\nby doing an RPM update.\nTo reduce the likelihood in a containerized environment virtlogd should\nonly be restarted on manual request, or on compute node reboot. It should\nnot be restarted on a minor update without migration off instances.\nThis introduces a nova_virtlogd_wrapper container and virtlogd wrapper\nscript, to only restart virtlogd on either manual or compute node restart.\n\nChange-Id: I5192f8c306422d0966eef7544f719cc21172bd92\nCo-Authored-By: Rajesh Tailor \u003cratailor@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/62066b0ce21d0e26ac158516ee59fe3077b85984"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/62066b0ce21d0e26ac158516ee59fe3077b85984"}]},"branch":"refs/heads/master"},"8f41413283f5568c21b2df35b8db8bded7af9a6f":{"kind":"REWORK","_number":5,"created":"2021-04-28 10:51:01.000000000","uploader":{"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},"ref":"refs/changes/71/787771/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tripleo","ref":"refs/changes/71/787771/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/71/787771/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/71/787771/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/71/787771/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tripleo refs/changes/71/787771/5"}}},"commit":{"parents":[{"commit":"98f00851c42f2fed2b09aba5d7dd37da4bfd03cd","subject":"Merge \"Remove redundant import of barbican::keystone::authtoken\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/98f00851c42f2fed2b09aba5d7dd37da4bfd03cd"}]}],"author":{"name":"Martin Schuppert","email":"mschuppert@redhat.com","date":"2021-04-23 14:52:45.000000000","tz":120},"committer":{"name":"Martin Schuppert","email":"mschuppert@redhat.com","date":"2021-04-28 10:02:35.000000000","tz":120},"subject":"Introduce nova virtlogd wrapper","message":"Introduce nova virtlogd wrapper\n\nWhen nova_virtlogd container gets restarted the instance console auth files\nwill not be reopened again by virtlogd. As a result either instances need\nto be restarted or live migrated to a different compute node to get new\nconsole logs messages logged again.\nUsually on receipt of SIGUSR1, virtlogd will re-exec() its binary, while\nmaintaining all current logs and clients. This allows for live upgrades of\nthe virtlogd service on non containerized environments where updates just\nby doing an RPM update.\nTo reduce the likelihood in a containerized environment virtlogd should\nonly be restarted on manual request, or on compute node reboot. It should\nnot be restarted on a minor update without migration off instances.\nThis introduces a nova_virtlogd_wrapper container and virtlogd wrapper\nscript, to only restart virtlogd on either manual or compute node restart.\n\nChange-Id: I5192f8c306422d0966eef7544f719cc21172bd92\nCo-Authored-By: Rajesh Tailor \u003cratailor@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/8f41413283f5568c21b2df35b8db8bded7af9a6f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/8f41413283f5568c21b2df35b8db8bded7af9a6f"}]},"branch":"refs/heads/master"},"a250c9fa383bcb5f547243982c3bc43b5602c686":{"kind":"REWORK","_number":6,"created":"2021-04-28 16:20:08.000000000","uploader":{"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},"ref":"refs/changes/71/787771/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tripleo","ref":"refs/changes/71/787771/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/71/787771/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/71/787771/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/71/787771/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tripleo refs/changes/71/787771/6"}}},"commit":{"parents":[{"commit":"98f00851c42f2fed2b09aba5d7dd37da4bfd03cd","subject":"Merge \"Remove redundant import of barbican::keystone::authtoken\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/98f00851c42f2fed2b09aba5d7dd37da4bfd03cd"}]}],"author":{"name":"Martin Schuppert","email":"mschuppert@redhat.com","date":"2021-04-23 14:52:45.000000000","tz":120},"committer":{"name":"Martin Schuppert","email":"mschuppert@redhat.com","date":"2021-04-28 14:40:51.000000000","tz":120},"subject":"Introduce nova virtlogd wrapper","message":"Introduce nova virtlogd wrapper\n\nWhen nova_virtlogd container gets restarted the instance console auth files\nwill not be reopened again by virtlogd. As a result either instances need\nto be restarted or live migrated to a different compute node to get new\nconsole logs messages logged again.\nUsually on receipt of SIGUSR1, virtlogd will re-exec() its binary, while\nmaintaining all current logs and clients. This allows for live upgrades of\nthe virtlogd service on non containerized environments where updates just\nby doing an RPM update.\nTo reduce the likelihood in a containerized environment virtlogd should\nonly be restarted on manual request, or on compute node reboot. It should\nnot be restarted on a minor update without migration off instances.\nThis introduces a nova_virtlogd_wrapper container and virtlogd wrapper\nscript, to only restart virtlogd on either manual or compute node restart.\n\nChange-Id: I5192f8c306422d0966eef7544f719cc21172bd92\nCo-Authored-By: Rajesh Tailor \u003cratailor@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/a250c9fa383bcb5f547243982c3bc43b5602c686"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/a250c9fa383bcb5f547243982c3bc43b5602c686"}]},"branch":"refs/heads/master"},"f09aa48d95bc0fada01b1a011bc00e8c9b1e87b0":{"kind":"NO_CODE_CHANGE","_number":7,"created":"2021-04-29 08:14:12.000000000","uploader":{"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},"ref":"refs/changes/71/787771/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tripleo","ref":"refs/changes/71/787771/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/71/787771/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/71/787771/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/71/787771/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tripleo refs/changes/71/787771/7"}}},"commit":{"parents":[{"commit":"98f00851c42f2fed2b09aba5d7dd37da4bfd03cd","subject":"Merge \"Remove redundant import of barbican::keystone::authtoken\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/98f00851c42f2fed2b09aba5d7dd37da4bfd03cd"}]}],"author":{"name":"Martin Schuppert","email":"mschuppert@redhat.com","date":"2021-04-23 14:52:45.000000000","tz":120},"committer":{"name":"Martin Schuppert","email":"mschuppert@redhat.com","date":"2021-04-29 08:11:29.000000000","tz":120},"subject":"Introduce nova virtlogd wrapper","message":"Introduce nova virtlogd wrapper\n\nWhen nova_virtlogd container gets restarted the instance console auth files\nwill not be reopened again by virtlogd. As a result either instances need\nto be restarted or live migrated to a different compute node to get new\nconsole logs messages logged again.\nUsually on receipt of SIGUSR1, virtlogd will re-exec() its binary, while\nmaintaining all current logs and clients. This allows for live upgrades of\nthe virtlogd service on non containerized environments where updates just\nby doing an RPM update.\nTo reduce the likelihood in a containerized environment virtlogd should\nonly be restarted on manual request, or on compute node reboot. It should\nnot be restarted on a minor update without migration off instances.\nThis introduces a nova_virtlogd_wrapper container and virtlogd wrapper\nscript, to only restart virtlogd on either manual or compute node restart.\n\nCloses-Bug: #1838272\n\nChange-Id: I5192f8c306422d0966eef7544f719cc21172bd92\nCo-Authored-By: Rajesh Tailor \u003cratailor@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/f09aa48d95bc0fada01b1a011bc00e8c9b1e87b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/f09aa48d95bc0fada01b1a011bc00e8c9b1e87b0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
