)]}'
{"id":"openstack%2Fpuppet-tripleo~795155","triplet_id":"openstack%2Fpuppet-tripleo~master~I97442478d77d222fdd428eebd270c08df08237f6","project":"openstack/puppet-tripleo","branch":"master","topic":"bug/1931133","hashtags":[],"change_id":"I97442478d77d222fdd428eebd270c08df08237f6","subject":"Add support for ovn clustered databases","status":"MERGED","created":"2021-06-07 16:21:54.000000000","updated":"2021-06-09 01:58:46.000000000","submitted":"2021-06-09 01:58:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"795155-bug/1931133","meta_rev_id":"6d40d91c3655cb8c93a9e8c4649eadcfe5d96a6a","_number":795155,"virtual_id_number":795155,"owner":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":20778,"name":"Damien Ciabrini","email":"dciabrin@redhat.com","username":"dciabrin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-09 01:58:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"value":1,"date":"2021-06-08 14:58:08.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":20778,"name":"Damien Ciabrini","email":"dciabrin@redhat.com","username":"dciabrin"},"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":2,"date":"2021-06-08 17:08:50.000000000","_account_id":20778,"name":"Damien Ciabrini","email":"dciabrin@redhat.com","username":"dciabrin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-06-08 14:04:50.000000000","_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":20778,"name":"Damien Ciabrini","email":"dciabrin@redhat.com","username":"dciabrin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-06-08 19:41:06.000000000","_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"_account_id":20778,"name":"Damien Ciabrini","email":"dciabrin@redhat.com","username":"dciabrin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-07 16:33:43.000000000","updated_by":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"reviewer":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"state":"REVIEWER"},{"updated":"2021-06-07 18:04:24.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-07 19:53:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-08 09:14:51.000000000","updated_by":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"reviewer":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"state":"CC"},{"updated":"2021-06-08 09:15:31.000000000","updated_by":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"reviewer":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"state":"REVIEWER"},{"updated":"2021-06-08 09:28:01.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2021-06-08 17:08:50.000000000","updated_by":{"_account_id":20778,"name":"Damien Ciabrini","email":"dciabrin@redhat.com","username":"dciabrin"},"reviewer":{"_account_id":20778,"name":"Damien Ciabrini","email":"dciabrin@redhat.com","username":"dciabrin"},"state":"REVIEWER"}],"messages":[{"id":"d26e78c0ce0d4aa55deae45da78255d387e906b1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-06-07 16:21:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"833049229c51a56b330daa04131c8b554359e30a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-06-07 16:23:56.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"43b38cc1165d0e59cb589708ec5b0ddb69efa97c","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2021-06-07 16:33:43.000000000","message":"Patch Set 2: Code-Review+1\n\nthank you for spotting and fixing that ovn_dbs_vip / ovn_db_node_ips thing!","accounts_in_message":[],"_revision_number":2},{"id":"a5c8f2114d32335fecdf2e43f5ffcf04ec0dcfce","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-06-07 18:04:24.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/71b204847ca74ba4aba63fe2a91752c3 : RETRY_LIMIT in 10m 09s","accounts_in_message":[],"_revision_number":2},{"id":"81d6d38dd61c6ed82bf7f6d2eff9c898f6bcf9ab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-07 19:53:18.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d799216f284d4ceea26cafeb0161709f : FAILURE in 6m 00s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/77d41bd202514239950287f00171daf9 : FAILURE in 4m 28s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0c212bbc7b5946038d40840e0aa4200a : FAILURE in 4m 42s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/c5e18e5d516641b68a0ba8f926e394ab : SUCCESS in 22m 16s\n- puppet-openstack-unit-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/213d6453510d405fb7cd99314b37ae2a : SUCCESS in 21m 30s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/debf2a1e5c2140719a63ae861266d9ff : SUCCESS in 21m 27s\n- puppet-openstack-unit-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7f130e6f28f44adcbe16f18dcb2dd14c : SUCCESS in 27m 52s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ee29ef925a694ec486679150b59d62f5 : SUCCESS in 23m 25s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/215944a095d94682a0264860889b2dcf : SUCCESS in 3h 06m 11s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/512de9ad3dbc4723904cf8f038f3fea5 : SUCCESS in 1h 10m 50s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/a24399a0828b49d192b83923c0a2ddb5 : FAILURE in 1h 24m 20s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/4da56b27a1df464eae4bedf0f2773555 : FAILURE in 46m 55s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/723690eba1134f038a5b74fb29a492cd : SUCCESS in 1h 37m 40s\n- tripleo-ci-centos-8-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/15d502845afb4212bbc1b07a03070d38 : SUCCESS in 1h 36m 10s","accounts_in_message":[],"_revision_number":2},{"id":"ff9e52b67fca806d26322e027d2baddf969490cb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-06-07 20:07:36.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"978459fb0b6a424aeef232c6aec581740a5a8faf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-06-07 20:08:24.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"f0fe5986a1e0e99bdf24b99435b9bfba4195d1fc","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-06-07 21:52:32.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/25fff1ffff9e4d37855c0480f0c40542 : RETRY_LIMIT in 19m 39s","accounts_in_message":[],"_revision_number":4},{"id":"3096408160571163c4ba2d7b380a5cd31c9e8b4c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-07 23:17:11.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cd645a76d91843468dca76896601fd41 : SUCCESS in 4m 04s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8dbda0c041674b86883d291a5db59fda : FAILURE in 3m 58s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b3494c44ffd74f32889e503ca50cb4eb : FAILURE in 3m 53s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/3d3d7408edda4e6a887b68a1af547989 : SUCCESS in 24m 36s\n- puppet-openstack-unit-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/d19c993d19de463cab32693540ab7ded : SUCCESS in 25m 29s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6661c922fdbb45bbba43bdfa0d298cee : SUCCESS in 22m 41s\n- puppet-openstack-unit-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6bde84d6fdcd48b9aea86a8f383b32e9 : SUCCESS in 22m 17s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/53c07d0c03244a7fb7ca2afdd1ca260a : SUCCESS in 21m 54s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/d22ad3649f304d88a42c89e6751bbd6b : SUCCESS in 3h 02m 56s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/a5ae63950ca04573b0a7430d3e2b3fdf : SUCCESS in 1h 12m 22s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/ce6c26d8a72541c9a2674f339b5428c0 : FAILURE in 1h 13m 47s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/c0d84b2949294c7e804c16b3fd2a59f3 : FAILURE in 50m 04s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/6fc3ab2c88e14fdaaed37545637c7a52 : SUCCESS in 1h 37m 17s\n- tripleo-ci-centos-8-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/86fc0bac0130491bbdec78ac360ba4dc : SUCCESS in 1h 39m 02s","accounts_in_message":[],"_revision_number":4},{"id":"3f68fe9b0f1bb70a0abffa5d7edf568c51ca9caf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-06-07 23:42:33.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"ebec5d1de4881ab1623040f7d249711a07e659a1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-08 02:26:55.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3bb29456e315427ca1e7a33473a5e866 : SUCCESS in 4m 10s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5636d398d09a43079dd90e607d01b132 : SUCCESS in 4m 15s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/65d5d9ddf122452d8ceb2bce9adabf33 : SUCCESS in 4m 07s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/1b4d49931ac240e7a193963a992c39e6 : SUCCESS in 20m 36s\n- puppet-openstack-unit-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/5f68378113574ed9874297379a026e7a : SUCCESS in 23m 23s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/203f46b741c848bb99bd0f2283d7d103 : SUCCESS in 20m 13s\n- puppet-openstack-unit-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ec2e1873aa4946b49f26037d4f792500 : SUCCESS in 20m 10s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6eef25d2e69f4def8c98ab38bb0dd8c5 : SUCCESS in 21m 06s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/31c1357b9b2a42cc85f25fc1a1acdc11 : SUCCESS in 2h 38m 48s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/44075e6f761d447c92453715e674b3ec : SUCCESS in 1h 04m 08s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/44f444cdc3c04487817fab3b684067c4 : FAILURE in 1h 11m 09s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/326d91d5d64f484d836a69d1ef2beca4 : FAILURE in 40m 28s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/c538bffc47324deea3a23b2acb539184 : SUCCESS in 1h 28m 39s\n- tripleo-ci-centos-8-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/b0caebf0924e41af96f24a16e8fd37a2 : SUCCESS in 1h 29m 48s","accounts_in_message":[],"_revision_number":5},{"id":"14d80ee52d23bbef393b75c340532c4203607ae6","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-06-08 04:08:25.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/77b624055d8a416abc5dc06450800e2f : SUCCESS in 3h 07m 33s","accounts_in_message":[],"_revision_number":5},{"id":"e24f9db4e5479b19634942ed86b704c975f9553c","author":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"date":"2021-06-08 09:14:51.000000000","message":"Patch Set 5:\n\nError: Evaluation Error: Resource type not found: False (file: /etc/puppet/modules/tripleo/manifests/profile/base/neutron/plugins/ml2/ovn.pp, line: 80, column: 57)\n\nI\u0027ll lowercase those and retest","accounts_in_message":[],"_revision_number":5},{"id":"0c8430ccff9992aec661102387d56e166c744a2b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"date":"2021-06-08 09:15:31.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"29187cee0464398f85eb7e4f975e8356ed735721","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-06-08 09:15:56.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/None : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":6},{"id":"5b0ce7ec3a95bd84798b67176c888faeb51c83a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-08 10:40:15.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f1d14e916866447693d414cce94fcc56 : SUCCESS in 4m 14s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9f0da7893be64c629708992650587457 : SUCCESS in 4m 26s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/350b3b62635b40a8a801163b0b3c02e8 : SUCCESS in 4m 23s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/2fda6562cdfc4750b26b42f46e066c74 : SUCCESS in 24m 34s\n- puppet-openstack-unit-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/3ca8caf653d5416d885aadf121fd3812 : SUCCESS in 23m 46s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bbe86781cc1a48a6bde8a4c77bed7bce : SUCCESS in 21m 39s\n- puppet-openstack-unit-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/44d311b0c43c46548fdd3ad1d0f63138 : SUCCESS in 21m 10s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3a896eb666c74b87aab362db27bc8757 : SUCCESS in 23m 55s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/43c0c3610f92453ead3c414134be1d2e : SUCCESS in 1h 22m 58s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/7cee31908c7c4817a21f0ce135f80353 : FAILURE in 11m 18s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/4e0c1a0e0ed14aa98f3d1171eb0daf43 : FAILURE in 14m 19s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/e56cb633b78644818d6e2e600ac4cb63 : FAILURE in 12m 25s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/8c0a4f3dd258437b90f27443b8ab15ee : FAILURE in 13m 33s\n- tripleo-ci-centos-8-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/f5fd505b25234267a7dd3767ede3c109 : FAILURE in 14m 38s","accounts_in_message":[],"_revision_number":6},{"id":"b3ab73ef8170afa1e4f933597d12107216c450c5","author":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"date":"2021-06-08 11:22:24.000000000","message":"Patch Set 6:\n\nThis version works fine in my tests. The failures are related to pip stuff I presume:\nERROR: Could not find a version that satisfies the requirement selinux (from versions: none)\nERROR: No matching distribution found for selinux\n\nand\n\nERROR: Could not find a version that satisfies the requirement clif (from versions: none)\nERROR: No matching distribution found for clif","accounts_in_message":[],"_revision_number":6},{"id":"67ea642fa466e3c4bd0266c719e53a5f642c352b","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2021-06-08 11:23:44.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"64e4d6e6ac677d82dbc70e28d7d5fdc7d4adc82c","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2021-06-08 11:24:10.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"d42c517e6e5578b94a960028d47d46f561736f73","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2021-06-08 11:30:35.000000000","message":"Patch Set 6:\n\nnit: ovn_db_*host* takes a value from hiera ovn_dbs_*vip*, which perhaps has a notion of host IP *or* FQDN as well? If so, we should call it ovn_db_node_hosts insteadm that takes either a list of FQDNs or IPs from ovn_dbs_node_ips maybe... But that\u0027s just in case in future we\u0027ll want a cluster assembled by hostnames instead of IPs.","accounts_in_message":[],"_revision_number":6},{"id":"e3df746fe7d810d734dfff5f5c6d435d971b0499","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2021-06-08 11:51:34.000000000","message":"Patch Set 6: -Code-Review\n\n2021-06-08 11:48:29.059 58729 ERROR ovsdbapp.backend.ovs_idl.idlutils [-] Unable to open stream to ssl:172.16.2.182:6642ssl:172.16.2.249:6642ssl:172.16.2.118:6642 to retrieve schema: Unknown error -2","accounts_in_message":[],"_revision_number":6},{"id":"b561125061b191db6a44555dc29398b289d28f67","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"date":"2021-06-08 11:55:23.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"208e8c97aaff5a4b603b7b9752e42e49cae25488","author":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"date":"2021-06-08 11:59:12.000000000","message":"Patch Set 7:\n\nfixed a typo in ovn_metadata, rerunning tests now","accounts_in_message":[],"_revision_number":7},{"id":"aab33b08a32265d5ab5b28ffabe43e24d1ce7c8a","author":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"date":"2021-06-08 14:04:50.000000000","message":"Patch Set 7: Code-Review+2\n\nTempest now passed in my env \"- Failed: 0\"","accounts_in_message":[],"_revision_number":7},{"id":"4cdef87c6521e5f5802c2a6b9d983198ea359fda","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-06-08 14:58:08.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/571346d6c6a94c1e814653e191b4a800 : SUCCESS in 3h 00m 02s","accounts_in_message":[],"_revision_number":7},{"id":"dc26e158100d86c14e89c71bf4ee26d391b6c6dd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-08 15:08:22.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4a23d5cad27e403babd7b3615ffb3b4e : SUCCESS in 4m 21s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8a9c6d8d611e403b8f61e340d9a1a750 : SUCCESS in 4m 18s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/668b911fa720446897bf295285ce753a : SUCCESS in 3m 44s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/29e150cc24834dfd9159011dfbad5f77 : SUCCESS in 22m 46s\n- puppet-openstack-unit-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/4c12b211649447a1ba87a457ad43dcaf : SUCCESS in 24m 52s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8f4ea13f65ac4cc294bbd6a9f438e589 : SUCCESS in 21m 31s\n- puppet-openstack-unit-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/737346e10cd242a38ca44e50cc6731ca : SUCCESS in 22m 11s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/35a36692cedd431fb43e5c44899b92e9 : SUCCESS in 23m 23s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/c2a0d47dceb44185811f0976ebc33445 : SUCCESS in 2h 56m 09s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/5ade6ddd8acd4aeebfcb1ff7b7e99723 : SUCCESS in 1h 04m 52s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/8667611565e94934b777a535784ede54 : SUCCESS in 1h 36m 16s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/57206d51c075490d84421e0c270c666f : SUCCESS in 1h 18m 38s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/844b20f005a447fd8af5ed9218fc04cc : SUCCESS in 1h 32m 56s\n- tripleo-ci-centos-8-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/44e9d0b51e644b4ca2c3d113bc02217e : SUCCESS in 1h 35m 54s","accounts_in_message":[],"_revision_number":7},{"id":"c601f9a1abd998dc7c4152318c963885ee6540dd","author":{"_account_id":20778,"name":"Damien Ciabrini","email":"dciabrin@redhat.com","username":"dciabrin"},"date":"2021-06-08 17:08:50.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"30021d400ea540379b6a45b88d29842b30b3c684","author":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"date":"2021-06-08 19:41:06.000000000","message":"Patch Set 7: Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"5756c9af5ca30d87a0158e5a4c96dc5b958e9d0e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-08 20:50:06.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"c6681235c34e4a0e70e274aa84c0b081c293443a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-09 01:58:45.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1298ffcf2e4a440c86ec1e649fa0a5df : SUCCESS in 4m 08s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/92dbf6db7fa34a77b3a79ff6e2db91b3 : SUCCESS in 4m 26s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/eb21daa0e13e4b88a1b1ffd7b2861dd7 : SUCCESS in 3m 57s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/7b83f312a57642d097fa1f132d4b9900 : SUCCESS in 22m 28s\n- puppet-openstack-unit-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/2901194cd77e4ab194124f35dfcd1963 : SUCCESS in 26m 25s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cae794975b294ff89f3f45d563639511 : SUCCESS in 21m 15s\n- puppet-openstack-unit-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/95afc03fabd04f159cb1446694519625 : SUCCESS in 22m 01s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/da97905d5d6f47a2b8e9afbfdd7dc8ab : SUCCESS in 3h 39m 07s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/152ac48ff3804e84995bc4bbe39ffd89 : SUCCESS in 1h 15m 35s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/9b9b5d7bbf5846339a2b48505f2fcf8f : SUCCESS in 1h 52m 09s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/5f2c90781b0d49ec97fa7a9d914b4553 : SUCCESS in 1h 25m 23s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/6a79fa8f58bc447bb832e761d18184af : SUCCESS in 1h 48m 33s","accounts_in_message":[],"_revision_number":7},{"id":"6d40d91c3655cb8c93a9e8c4649eadcfe5d96a6a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-09 01:58:46.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"dee8091c8904105df2747a6639fe2779726b28df","revisions":{"ccd1bf6800c137b81982a6bbe60dc6b8ab378fd7":{"kind":"REWORK","_number":1,"created":"2021-06-07 16:21:54.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/55/795155/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tripleo","ref":"refs/changes/55/795155/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/55/795155/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/55/795155/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/55/795155/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tripleo refs/changes/55/795155/1"}}},"commit":{"parents":[{"commit":"fad7e3fb90f2f03e011fd245e5b60d510726d9e3","subject":"Revert \"Allow ovn_db_host to be of type array\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/fad7e3fb90f2f03e011fd245e5b60d510726d9e3"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2021-06-07 16:17:48.000000000","tz":-360},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2021-06-07 16:21:46.000000000","tz":-360},"subject":"Add support for ovn clustered databases","message":"Add support for ovn clustered databases\n\nThis patch adds support for managing the ovn connection correctly when\novn is handling clustering instead of pacemaker. By default if the\nparameter is not enabled, the pacemaker managed way is assumed.\n\nChange-Id: I97442478d77d222fdd428eebd270c08df08237f6\nRelated-Bug: #1931133\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/ccd1bf6800c137b81982a6bbe60dc6b8ab378fd7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/ccd1bf6800c137b81982a6bbe60dc6b8ab378fd7"}]},"branch":"refs/heads/master"},"ee1a0250305fbc0789cdfbf6cf51affd157f2183":{"kind":"REWORK","_number":2,"created":"2021-06-07 16:23:56.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/55/795155/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tripleo","ref":"refs/changes/55/795155/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/55/795155/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/55/795155/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/55/795155/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tripleo refs/changes/55/795155/2"}}},"commit":{"parents":[{"commit":"fad7e3fb90f2f03e011fd245e5b60d510726d9e3","subject":"Revert \"Allow ovn_db_host to be of type array\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/fad7e3fb90f2f03e011fd245e5b60d510726d9e3"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2021-06-07 16:17:48.000000000","tz":-360},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2021-06-07 16:23:44.000000000","tz":-360},"subject":"Add support for ovn clustered databases","message":"Add support for ovn clustered databases\n\nThis patch adds support for managing the ovn connection correctly when\novn is handling clustering instead of pacemaker. By default if the\nparameter is not enabled, the pacemaker managed way is assumed.\n\nChange-Id: I97442478d77d222fdd428eebd270c08df08237f6\nRelated-Bug: #1931133\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/ee1a0250305fbc0789cdfbf6cf51affd157f2183"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/ee1a0250305fbc0789cdfbf6cf51affd157f2183"}]},"branch":"refs/heads/master"},"05b041ff934c90f3c053666dd909680690595570":{"kind":"REWORK","_number":3,"created":"2021-06-07 20:07:36.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/55/795155/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tripleo","ref":"refs/changes/55/795155/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/55/795155/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/55/795155/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/55/795155/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tripleo refs/changes/55/795155/3"}}},"commit":{"parents":[{"commit":"fad7e3fb90f2f03e011fd245e5b60d510726d9e3","subject":"Revert \"Allow ovn_db_host to be of type array\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/fad7e3fb90f2f03e011fd245e5b60d510726d9e3"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2021-06-07 16:17:48.000000000","tz":-360},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2021-06-07 20:07:32.000000000","tz":-360},"subject":"Add support for ovn clustered databases","message":"Add support for ovn clustered databases\n\nThis patch adds support for managing the ovn connection correctly when\novn is handling clustering instead of pacemaker. By default if the\nparameter is not enabled, the pacemaker managed way is assumed.\n\nChange-Id: I97442478d77d222fdd428eebd270c08df08237f6\nRelated-Bug: #1931133\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/05b041ff934c90f3c053666dd909680690595570"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/05b041ff934c90f3c053666dd909680690595570"}]},"branch":"refs/heads/master"},"d6b47db4d371447c6087e8d81914e89a7eec04cb":{"kind":"REWORK","_number":4,"created":"2021-06-07 20:08:24.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/55/795155/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tripleo","ref":"refs/changes/55/795155/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/55/795155/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/55/795155/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/55/795155/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tripleo refs/changes/55/795155/4"}}},"commit":{"parents":[{"commit":"fad7e3fb90f2f03e011fd245e5b60d510726d9e3","subject":"Revert \"Allow ovn_db_host to be of type array\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/fad7e3fb90f2f03e011fd245e5b60d510726d9e3"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2021-06-07 16:17:48.000000000","tz":-360},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2021-06-07 20:08:20.000000000","tz":-360},"subject":"Add support for ovn clustered databases","message":"Add support for ovn clustered databases\n\nThis patch adds support for managing the ovn connection correctly when\novn is handling clustering instead of pacemaker. By default if the\nparameter is not enabled, the pacemaker managed way is assumed.\n\nChange-Id: I97442478d77d222fdd428eebd270c08df08237f6\nRelated-Bug: #1931133\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/d6b47db4d371447c6087e8d81914e89a7eec04cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/d6b47db4d371447c6087e8d81914e89a7eec04cb"}]},"branch":"refs/heads/master"},"31a603dde9a89fce9f617769cc9c9116bde5f126":{"kind":"REWORK","_number":5,"created":"2021-06-07 23:42:33.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/55/795155/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tripleo","ref":"refs/changes/55/795155/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/55/795155/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/55/795155/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/55/795155/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tripleo refs/changes/55/795155/5"}}},"commit":{"parents":[{"commit":"fad7e3fb90f2f03e011fd245e5b60d510726d9e3","subject":"Revert \"Allow ovn_db_host to be of type array\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/fad7e3fb90f2f03e011fd245e5b60d510726d9e3"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2021-06-07 16:17:48.000000000","tz":-360},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2021-06-07 23:42:19.000000000","tz":-360},"subject":"Add support for ovn clustered databases","message":"Add support for ovn clustered databases\n\nThis patch adds support for managing the ovn connection correctly when\novn is handling clustering instead of pacemaker. By default if the\nparameter is not enabled, the pacemaker managed way is assumed.\n\nChange-Id: I97442478d77d222fdd428eebd270c08df08237f6\nRelated-Bug: #1931133\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/31a603dde9a89fce9f617769cc9c9116bde5f126"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/31a603dde9a89fce9f617769cc9c9116bde5f126"}]},"branch":"refs/heads/master"},"81e52e03981e0ef60d7b14448fa3ad592c149116":{"kind":"REWORK","_number":6,"created":"2021-06-08 09:15:31.000000000","uploader":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"ref":"refs/changes/55/795155/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tripleo","ref":"refs/changes/55/795155/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/55/795155/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/55/795155/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/55/795155/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tripleo refs/changes/55/795155/6"}}},"commit":{"parents":[{"commit":"fad7e3fb90f2f03e011fd245e5b60d510726d9e3","subject":"Revert \"Allow ovn_db_host to be of type array\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/fad7e3fb90f2f03e011fd245e5b60d510726d9e3"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2021-06-07 16:17:48.000000000","tz":-360},"committer":{"name":"Michele Baldessari","email":"michele@acksyn.org","date":"2021-06-08 09:13:11.000000000","tz":120},"subject":"Add support for ovn clustered databases","message":"Add support for ovn clustered databases\n\nThis patch adds support for managing the ovn connection correctly when\novn is handling clustering instead of pacemaker. By default if the\nparameter is not enabled, the pacemaker managed way is assumed.\n\nChange-Id: I97442478d77d222fdd428eebd270c08df08237f6\nRelated-Bug: #1931133\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/81e52e03981e0ef60d7b14448fa3ad592c149116"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/81e52e03981e0ef60d7b14448fa3ad592c149116"}]},"branch":"refs/heads/master"},"dee8091c8904105df2747a6639fe2779726b28df":{"kind":"REWORK","_number":7,"created":"2021-06-08 11:55:23.000000000","uploader":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"ref":"refs/changes/55/795155/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tripleo","ref":"refs/changes/55/795155/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/55/795155/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/55/795155/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/55/795155/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tripleo refs/changes/55/795155/7"}}},"commit":{"parents":[{"commit":"fad7e3fb90f2f03e011fd245e5b60d510726d9e3","subject":"Revert \"Allow ovn_db_host to be of type array\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/fad7e3fb90f2f03e011fd245e5b60d510726d9e3"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2021-06-07 16:17:48.000000000","tz":-360},"committer":{"name":"Michele Baldessari","email":"michele@acksyn.org","date":"2021-06-08 11:54:51.000000000","tz":120},"subject":"Add support for ovn clustered databases","message":"Add support for ovn clustered databases\n\nThis patch adds support for managing the ovn connection correctly when\novn is handling clustering instead of pacemaker. By default if the\nparameter is not enabled, the pacemaker managed way is assumed.\n\nChange-Id: I97442478d77d222fdd428eebd270c08df08237f6\nRelated-Bug: #1931133\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/dee8091c8904105df2747a6639fe2779726b28df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/dee8091c8904105df2747a6639fe2779726b28df"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":20778,"name":"Damien Ciabrini","email":"dciabrin@redhat.com","username":"dciabrin"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
