)]}'
{"id":"openstack%2Fpuppet-tripleo~804045","triplet_id":"openstack%2Fpuppet-tripleo~master~I483484c58ab703f3a4dcede636d733e23b051f63","project":"openstack/puppet-tripleo","branch":"master","hashtags":[],"change_id":"I483484c58ab703f3a4dcede636d733e23b051f63","subject":"Allow tuning of Galera gcache size","status":"MERGED","created":"2021-08-10 07:31:54.000000000","updated":"2021-08-10 20:33:54.000000000","submitted":"2021-08-10 20:33:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"804045","meta_rev_id":"26ac5a7f62ee979a927afadd5c1f2ebbcc0a0df9","_number":804045,"virtual_id_number":804045,"owner":{"_account_id":30126,"name":"Luca Miccini","email":"lmiccini@redhat.com","username":"lmiccini2"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":20778,"name":"Damien Ciabrini","email":"dciabrin@redhat.com","username":"dciabrin"},{"value":0,"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-10 20:33:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-08-10 13:09:55.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":30126,"name":"Luca Miccini","email":"lmiccini@redhat.com","username":"lmiccini2"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2021-08-10 14:27:09.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2021-08-10 10:42:06.000000000","_account_id":20778,"name":"Damien Ciabrini","email":"dciabrin@redhat.com","username":"dciabrin"},{"value":0,"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":30126,"name":"Luca Miccini","email":"lmiccini@redhat.com","username":"lmiccini2"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":1,"date":"2021-08-10 14:27:09.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":20778,"name":"Damien Ciabrini","email":"dciabrin@redhat.com","username":"dciabrin"},{"value":0,"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":30126,"name":"Luca Miccini","email":"lmiccini@redhat.com","username":"lmiccini2"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"_account_id":20778,"name":"Damien Ciabrini","email":"dciabrin@redhat.com","username":"dciabrin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":30126,"name":"Luca Miccini","email":"lmiccini@redhat.com","username":"lmiccini2"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-10 09:48:20.000000000","updated_by":{"_account_id":20778,"name":"Damien Ciabrini","email":"dciabrin@redhat.com","username":"dciabrin"},"reviewer":{"_account_id":20778,"name":"Damien Ciabrini","email":"dciabrin@redhat.com","username":"dciabrin"},"state":"REVIEWER"},{"updated":"2021-08-10 09:51:51.000000000","updated_by":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"reviewer":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"state":"REVIEWER"},{"updated":"2021-08-10 13:09:55.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-10 13:34:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-10 14:27:09.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"ca7a877169019ada6a92bea72105378b2d184ba8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30126,"name":"Luca Miccini","email":"lmiccini@redhat.com","username":"lmiccini2"},"date":"2021-08-10 07:31:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"56ce99eecb77249d240fe54b8754308bf6c26e5b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30126,"name":"Luca Miccini","email":"lmiccini@redhat.com","username":"lmiccini2"},"date":"2021-08-10 07:56:18.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a0c80f44d057f4d1000c1ec57af712ec0930aa33","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30126,"name":"Luca Miccini","email":"lmiccini@redhat.com","username":"lmiccini2"},"date":"2021-08-10 07:57:37.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"7b0c65fd0bc86fb9349abfe3fad5f2b6858a165d","author":{"_account_id":30126,"name":"Luca Miccini","email":"lmiccini@redhat.com","username":"lmiccini2"},"date":"2021-08-10 08:23:50.000000000","message":"Patch Set 3: Workflow-1","accounts_in_message":[],"_revision_number":3},{"id":"b7de82234eaa33e5d38a08cd1a2161f803cb33dd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30126,"name":"Luca Miccini","email":"lmiccini@redhat.com","username":"lmiccini2"},"date":"2021-08-10 09:25:37.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"74507e90778bf8aa42502f44fa778c1971345eee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30126,"name":"Luca Miccini","email":"lmiccini@redhat.com","username":"lmiccini2"},"date":"2021-08-10 09:40:25.000000000","message":"Uploaded patch set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"9888b4df2f560db66a8edb5a0d66f9bd1014b732","author":{"_account_id":20778,"name":"Damien Ciabrini","email":"dciabrin@redhat.com","username":"dciabrin"},"date":"2021-08-10 09:48:20.000000000","message":"Patch Set 5: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"c22062badd4b47834220c2596c67f0dc3ba32cda","author":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"date":"2021-08-10 09:51:51.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"518cad71d004d300ac6953809a5ac271f5857711","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30126,"name":"Luca Miccini","email":"lmiccini@redhat.com","username":"lmiccini2"},"date":"2021-08-10 10:36:03.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"54f2ad0adccd2662dd10f7ee7cbe6037f1019957","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30126,"name":"Luca Miccini","email":"lmiccini@redhat.com","username":"lmiccini2"},"date":"2021-08-10 10:37:06.000000000","message":"Uploaded patch set 7: Commit message was updated.","accounts_in_message":[],"_revision_number":7},{"id":"2860ca669e5c866138fdc14e63e01e5bd84ea313","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30126,"name":"Luca Miccini","email":"lmiccini@redhat.com","username":"lmiccini2"},"date":"2021-08-10 10:38:30.000000000","message":"Uploaded patch set 8: Commit message was updated.","accounts_in_message":[],"_revision_number":8},{"id":"02ebcb81420e935b20dc6eb795606002e8f8689a","author":{"_account_id":20778,"name":"Damien Ciabrini","email":"dciabrin@redhat.com","username":"dciabrin"},"date":"2021-08-10 10:42:06.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"67eedb50f22c3a62e9489d4c0089160f6a2b90e9","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-08-10 13:09:55.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/0af4cf4f43b246cca1480997b617258d : SUCCESS in 2h 30m 04s","accounts_in_message":[],"_revision_number":8},{"id":"7297a78d598c98df6e4fb0eb0fb4732b0829854a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-10 13:34:41.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b3a001bc12ae4b5bafba53708266a937 : SUCCESS in 4m 15s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2c9a6c0f3c6543f0a98d921611382c32 : SUCCESS in 4m 19s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/25ecde18d4714d07bf8b69342feca2f4 : SUCCESS in 4m 25s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/2c6db645412d4e37a1af4de3f4ac40fd : SUCCESS in 25m 43s\n- puppet-openstack-unit-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/15e0948f9cd3427a99c11e4803c1c5ba : SUCCESS in 23m 00s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2b4651d3b52f490188bdfd7d2eb29d4a : SUCCESS in 21m 08s\n- puppet-openstack-unit-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8b85fc63f3f04c5ebd89f95f6f57cbf8 : SUCCESS in 21m 12s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/94746f08584f4882910b767f6499fd5a : SUCCESS in 21m 15s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/e66e02cfd9dc4b868fac780ca7939a4a : SUCCESS in 2h 54m 04s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/c169e1aa36d94a01bca8b91d79fd7981 : SUCCESS in 43m 19s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/0decebde4ea842d381e8c9feb67f573a : SUCCESS in 1h 31m 29s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/daf37bdcba564123bd45f316f30e26ea : SUCCESS in 1h 11m 11s\n- tripleo-ci-centos-8-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/6abe37ca013349ff88e791f87e3a1f86 : SUCCESS in 2h 25m 57s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/2947caee7535463294feef5f2ca0f023 : SUCCESS in 1h 25m 10s\n- tripleo-ci-centos-8-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/04a704afcf5b42d1a6b9ba40de665f20 : SUCCESS in 1h 07m 21s","accounts_in_message":[],"_revision_number":8},{"id":"692551bde94617804f4910c7bdfc002055c259de","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-08-10 14:27:09.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"79f860fc4cca6bb703a720ffa4c1424504649229","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-10 14:28:38.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"8e92495162e26a395fe3d971aed35aceef5351ca","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-10 20:33:54.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/edfddd2c8eff491f8e72e452243feb9e : SUCCESS in 5m 36s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ccdd69d10f004069a514438d93d6991f : SUCCESS in 5m 46s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c1d2a04fa59d4bc9a5143f2d5b69d8c6 : SUCCESS in 4m 20s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/0b14ccb3689c46d7a955ed437e7e3dc1 : SUCCESS in 28m 13s\n- puppet-openstack-unit-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/4558773cda40489fae19b11c3d82a802 : SUCCESS in 27m 16s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0d07e7907da94b4a936737b07ddaef61 : SUCCESS in 23m 21s\n- puppet-openstack-unit-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f6c513d7dccc400eb3553e1bd7c969f4 : SUCCESS in 24m 37s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/89cdb9c04b6a48fba3209474c4d14b59 : SUCCESS in 3h 19m 48s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/69b93283791a4227a6f25933c7c55d1e : SUCCESS in 54m 00s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/d68819018b5c40c4b83ba866ba7c6cb4 : SUCCESS in 1h 39m 37s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/7be594c73f664963927f9fd3b70534f0 : SUCCESS in 1h 22m 14s\n- tripleo-ci-centos-8-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/d15b3a50fe3e41779fd0c8c653c20c77 : SUCCESS in 2h 55m 19s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/71fdce882e784c38971e17239eab7f9e : SUCCESS in 1h 27m 15s\n- tripleo-ci-centos-8-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/f1874ce3b91a4eb78200f3ca0092cc3d : SUCCESS in 1h 13m 09s","accounts_in_message":[],"_revision_number":8},{"id":"26ac5a7f62ee979a927afadd5c1f2ebbcc0a0df9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-10 20:33:54.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"feed67dcc45eb3da818cc2ceac8cb3682a2b6364","revisions":{"b65f93d7e3f9873efc066bf6b97108dc9649d58f":{"kind":"REWORK","_number":1,"created":"2021-08-10 07:31:54.000000000","uploader":{"_account_id":30126,"name":"Luca Miccini","email":"lmiccini@redhat.com","username":"lmiccini2"},"ref":"refs/changes/45/804045/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tripleo","ref":"refs/changes/45/804045/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/45/804045/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/45/804045/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/45/804045/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tripleo refs/changes/45/804045/1"}}},"commit":{"parents":[{"commit":"9bc231913773665f9e84e934a4d17d8bd0bf3f36","subject":"Merge \"Enable configuration of swift-recon-cron job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/9bc231913773665f9e84e934a4d17d8bd0bf3f36"}]}],"author":{"name":"Luca Miccini","email":"lmiccini@redhat.com","date":"2021-08-10 06:25:00.000000000","tz":120},"committer":{"name":"Luca Miccini","email":"lmiccini@redhat.com","date":"2021-08-10 07:03:48.000000000","tz":120},"subject":"Allow tuning of Galera Cluster cache","message":"Allow tuning of Galera Cluster cache\n\nSizing the gcache size properly allows a cluster node to perform IST\ninstead of SST if the amount of writes occurred during its downtime do\nnot exceed the size of the cache itself.\nThis is especially beneficial during maintenance windows.\n\nChange-Id: I483484c58ab703f3a4dcede636d733e23b051f63\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/b65f93d7e3f9873efc066bf6b97108dc9649d58f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/b65f93d7e3f9873efc066bf6b97108dc9649d58f"}]},"branch":"refs/heads/master"},"9b24b616fb7bba41dc98f0f86d5e780a3ca61a44":{"kind":"REWORK","_number":2,"created":"2021-08-10 07:56:18.000000000","uploader":{"_account_id":30126,"name":"Luca Miccini","email":"lmiccini@redhat.com","username":"lmiccini2"},"ref":"refs/changes/45/804045/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tripleo","ref":"refs/changes/45/804045/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/45/804045/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/45/804045/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/45/804045/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tripleo refs/changes/45/804045/2"}}},"commit":{"parents":[{"commit":"9bc231913773665f9e84e934a4d17d8bd0bf3f36","subject":"Merge \"Enable configuration of swift-recon-cron job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/9bc231913773665f9e84e934a4d17d8bd0bf3f36"}]}],"author":{"name":"Luca Miccini","email":"lmiccini@redhat.com","date":"2021-08-10 06:25:00.000000000","tz":120},"committer":{"name":"Luca Miccini","email":"lmiccini@redhat.com","date":"2021-08-10 07:55:56.000000000","tz":120},"subject":"Allow tuning of Galera Cluster cache","message":"Allow tuning of Galera Cluster cache\n\nSizing the gcache size properly allows a cluster node to perform IST\ninstead of SST if the amount of writes occurred during its downtime do\nnot exceed the size of the cache itself.\nThis is especially beneficial during maintenance windows.\n\nChange-Id: I483484c58ab703f3a4dcede636d733e23b051f63\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/9b24b616fb7bba41dc98f0f86d5e780a3ca61a44"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/9b24b616fb7bba41dc98f0f86d5e780a3ca61a44"}]},"branch":"refs/heads/master"},"e00ff17409ecd9fc626d25c99748f56701fcb91a":{"kind":"REWORK","_number":3,"created":"2021-08-10 07:57:37.000000000","uploader":{"_account_id":30126,"name":"Luca Miccini","email":"lmiccini@redhat.com","username":"lmiccini2"},"ref":"refs/changes/45/804045/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tripleo","ref":"refs/changes/45/804045/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/45/804045/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/45/804045/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/45/804045/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tripleo refs/changes/45/804045/3"}}},"commit":{"parents":[{"commit":"9bc231913773665f9e84e934a4d17d8bd0bf3f36","subject":"Merge \"Enable configuration of swift-recon-cron job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/9bc231913773665f9e84e934a4d17d8bd0bf3f36"}]}],"author":{"name":"Luca Miccini","email":"lmiccini@redhat.com","date":"2021-08-10 06:25:00.000000000","tz":120},"committer":{"name":"Luca Miccini","email":"lmiccini@redhat.com","date":"2021-08-10 07:57:22.000000000","tz":120},"subject":"Allow tuning of Galera Cluster cache","message":"Allow tuning of Galera Cluster cache\n\nSizing the gcache size properly allows a cluster node to perform IST\ninstead of SST if the amount of writes occurred during its downtime do\nnot exceed the size of the cache itself.\nThis is especially beneficial during maintenance windows.\n\nChange-Id: I483484c58ab703f3a4dcede636d733e23b051f63\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/e00ff17409ecd9fc626d25c99748f56701fcb91a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/e00ff17409ecd9fc626d25c99748f56701fcb91a"}]},"branch":"refs/heads/master"},"989e039804c327c6c1f20dd232fd480e1b23a6de":{"kind":"REWORK","_number":4,"created":"2021-08-10 09:25:37.000000000","uploader":{"_account_id":30126,"name":"Luca Miccini","email":"lmiccini@redhat.com","username":"lmiccini2"},"ref":"refs/changes/45/804045/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tripleo","ref":"refs/changes/45/804045/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/45/804045/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/45/804045/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/45/804045/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tripleo refs/changes/45/804045/4"}}},"commit":{"parents":[{"commit":"9bc231913773665f9e84e934a4d17d8bd0bf3f36","subject":"Merge \"Enable configuration of swift-recon-cron job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/9bc231913773665f9e84e934a4d17d8bd0bf3f36"}]}],"author":{"name":"Luca Miccini","email":"lmiccini@redhat.com","date":"2021-08-10 06:25:00.000000000","tz":120},"committer":{"name":"Luca Miccini","email":"lmiccini@redhat.com","date":"2021-08-10 09:25:28.000000000","tz":120},"subject":"Allow tuning of Galera Cluster cache","message":"Allow tuning of Galera Cluster cache\n\nSizing the gcache size properly allows a cluster node to perform IST\ninstead of SST if the amount of writes occurred during its downtime do\nnot exceed the size of the cache itself.\nThis is especially beneficial during maintenance windows.\n\nChange-Id: I483484c58ab703f3a4dcede636d733e23b051f63\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/989e039804c327c6c1f20dd232fd480e1b23a6de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/989e039804c327c6c1f20dd232fd480e1b23a6de"}]},"branch":"refs/heads/master"},"f19c5fd94fb5e1e229816096cc631f806babe673":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2021-08-10 09:40:25.000000000","uploader":{"_account_id":30126,"name":"Luca Miccini","email":"lmiccini@redhat.com","username":"lmiccini2"},"ref":"refs/changes/45/804045/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tripleo","ref":"refs/changes/45/804045/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/45/804045/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/45/804045/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/45/804045/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tripleo refs/changes/45/804045/5"}}},"commit":{"parents":[{"commit":"9bc231913773665f9e84e934a4d17d8bd0bf3f36","subject":"Merge \"Enable configuration of swift-recon-cron job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/9bc231913773665f9e84e934a4d17d8bd0bf3f36"}]}],"author":{"name":"Luca Miccini","email":"lmiccini@redhat.com","date":"2021-08-10 06:25:00.000000000","tz":120},"committer":{"name":"Luca Miccini","email":"lmiccini@redhat.com","date":"2021-08-10 09:39:56.000000000","tz":120},"subject":"Allow tuning of Galera Cluster cache","message":"Allow tuning of Galera Cluster cache\n\nAppropriate gcache.size values allows a cluster node to perform IST\ninstead of SST if the writes occurred during its downtime do not exceed\nthe size of the cache itself.\nThis is especially beneficial during maintenance windows.\n\nChange-Id: I483484c58ab703f3a4dcede636d733e23b051f63\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/f19c5fd94fb5e1e229816096cc631f806babe673"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/f19c5fd94fb5e1e229816096cc631f806babe673"}]},"branch":"refs/heads/master"},"e3b9eed0dd7d7e37ebd1b903f00955c41de50d80":{"kind":"REWORK","_number":6,"created":"2021-08-10 10:36:03.000000000","uploader":{"_account_id":30126,"name":"Luca Miccini","email":"lmiccini@redhat.com","username":"lmiccini2"},"ref":"refs/changes/45/804045/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tripleo","ref":"refs/changes/45/804045/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/45/804045/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/45/804045/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/45/804045/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tripleo refs/changes/45/804045/6"}}},"commit":{"parents":[{"commit":"9bc231913773665f9e84e934a4d17d8bd0bf3f36","subject":"Merge \"Enable configuration of swift-recon-cron job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/9bc231913773665f9e84e934a4d17d8bd0bf3f36"}]}],"author":{"name":"Luca Miccini","email":"lmiccini@redhat.com","date":"2021-08-10 06:25:00.000000000","tz":120},"committer":{"name":"Luca Miccini","email":"lmiccini@redhat.com","date":"2021-08-10 10:35:36.000000000","tz":120},"subject":"Allow tuning of Galera Cluster cache","message":"Allow tuning of Galera Cluster cache\n\nAppropriate gcache.size values allows a cluster node to perform IST\ninstead of SST if the writes occurred during its downtime do not exceed\nthe size of the cache itself.\nThis is especially beneficial during maintenance windows.\n\nChange-Id: I483484c58ab703f3a4dcede636d733e23b051f63\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/e3b9eed0dd7d7e37ebd1b903f00955c41de50d80"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/e3b9eed0dd7d7e37ebd1b903f00955c41de50d80"}]},"branch":"refs/heads/master"},"3a5f4a11ee10076f38f2fb46a8b4b8c9726d4b36":{"kind":"NO_CODE_CHANGE","_number":7,"created":"2021-08-10 10:37:06.000000000","uploader":{"_account_id":30126,"name":"Luca Miccini","email":"lmiccini@redhat.com","username":"lmiccini2"},"ref":"refs/changes/45/804045/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tripleo","ref":"refs/changes/45/804045/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/45/804045/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/45/804045/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/45/804045/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tripleo refs/changes/45/804045/7"}}},"commit":{"parents":[{"commit":"9bc231913773665f9e84e934a4d17d8bd0bf3f36","subject":"Merge \"Enable configuration of swift-recon-cron job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/9bc231913773665f9e84e934a4d17d8bd0bf3f36"}]}],"author":{"name":"Luca Miccini","email":"lmiccini@redhat.com","date":"2021-08-10 06:25:00.000000000","tz":120},"committer":{"name":"Luca Miccini","email":"lmiccini@redhat.com","date":"2021-08-10 10:36:34.000000000","tz":120},"subject":"Allow tuning of Galera gcache size","message":"Allow tuning of Galera gcache size\n\nAppropriate gcache.size values allows a cluster node to perform IST\ninstead of SST if the writes occurred during its downtime do not exceed\nthe size of the cache itself.\nThis is especially beneficial during maintenance windows.\n\nChange-Id: I483484c58ab703f3a4dcede636d733e23b051f63\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/3a5f4a11ee10076f38f2fb46a8b4b8c9726d4b36"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/3a5f4a11ee10076f38f2fb46a8b4b8c9726d4b36"}]},"branch":"refs/heads/master"},"feed67dcc45eb3da818cc2ceac8cb3682a2b6364":{"kind":"NO_CODE_CHANGE","_number":8,"created":"2021-08-10 10:38:30.000000000","uploader":{"_account_id":30126,"name":"Luca Miccini","email":"lmiccini@redhat.com","username":"lmiccini2"},"ref":"refs/changes/45/804045/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tripleo","ref":"refs/changes/45/804045/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/45/804045/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/45/804045/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/45/804045/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tripleo refs/changes/45/804045/8"}}},"commit":{"parents":[{"commit":"9bc231913773665f9e84e934a4d17d8bd0bf3f36","subject":"Merge \"Enable configuration of swift-recon-cron job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/9bc231913773665f9e84e934a4d17d8bd0bf3f36"}]}],"author":{"name":"Luca Miccini","email":"lmiccini@redhat.com","date":"2021-08-10 06:25:00.000000000","tz":120},"committer":{"name":"Luca Miccini","email":"lmiccini@redhat.com","date":"2021-08-10 10:37:47.000000000","tz":120},"subject":"Allow tuning of Galera gcache size","message":"Allow tuning of Galera gcache size\n\nAppropriate gcache.size values allow a cluster node to perform IST\ninstead of SST if the writes occurred during its downtime do not exceed\nthe size of the cache itself.\nThis is especially beneficial during maintenance windows.\n\nChange-Id: I483484c58ab703f3a4dcede636d733e23b051f63\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/feed67dcc45eb3da818cc2ceac8cb3682a2b6364"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/feed67dcc45eb3da818cc2ceac8cb3682a2b6364"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
