)]}'
{"id":"openstack%2Fpuppet-tripleo~808774","triplet_id":"openstack%2Fpuppet-tripleo~master~Ic3e427156fc82c524c763d1896937fcc3c49fabb","project":"openstack/puppet-tripleo","branch":"master","topic":"selinux/missing-flags","hashtags":[],"change_id":"Ic3e427156fc82c524c763d1896937fcc3c49fabb","subject":"Add missing \"z\" flag for specific mounts","status":"MERGED","created":"2021-09-13 13:58:02.000000000","updated":"2021-09-15 07:20:47.000000000","submitted":"2021-09-15 07:05:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"808774-selinux/missing-flags","meta_rev_id":"d04f70af8e62fd58203bec872f959ce009a9a54d","_number":808774,"virtual_id_number":808774,"owner":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"value":0,"_account_id":20778,"name":"Damien Ciabrini","email":"dciabrin@redhat.com","username":"dciabrin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-15 07:05:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2021-09-14 13:35:58.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"all":[{"value":2,"date":"2021-09-14 20:02:18.000000000","_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"value":0,"_account_id":20778,"name":"Damien Ciabrini","email":"dciabrin@redhat.com","username":"dciabrin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":2,"date":"2021-09-14 14:57:35.000000000","_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"all":[{"value":1,"date":"2021-09-14 20:02:18.000000000","_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"value":0,"_account_id":20778,"name":"Damien Ciabrini","email":"dciabrin@redhat.com","username":"dciabrin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"_account_id":11166,"name":"Lukas Bezdicka","email":"social@v3.sk","username":"xbezdick"}],"reviewers":{"REVIEWER":[{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"_account_id":11166,"name":"Lukas Bezdicka","email":"social@v3.sk","username":"xbezdick"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"_account_id":20778,"name":"Damien Ciabrini","email":"dciabrin@redhat.com","username":"dciabrin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-13 14:07:48.000000000","updated_by":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"reviewer":{"_account_id":11166,"name":"Lukas Bezdicka","email":"social@v3.sk","username":"xbezdick"},"state":"REVIEWER"},{"updated":"2021-09-13 14:07:48.000000000","updated_by":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"reviewer":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"state":"REVIEWER"},{"updated":"2021-09-13 14:07:48.000000000","updated_by":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2021-09-13 14:07:48.000000000","updated_by":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"reviewer":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"state":"REVIEWER"},{"updated":"2021-09-13 14:07:48.000000000","updated_by":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"reviewer":{"_account_id":20778,"name":"Damien Ciabrini","email":"dciabrin@redhat.com","username":"dciabrin"},"state":"REVIEWER"},{"updated":"2021-09-13 16:53:47.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-13 17:05:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-14 08:30:14.000000000","updated_by":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"reviewer":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"state":"REVIEWER"},{"updated":"2021-09-14 20:02:18.000000000","updated_by":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"reviewer":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"state":"REVIEWER"}],"messages":[{"id":"5a3b0d13585563d49e78b1c5d59f55745566631f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2021-09-13 13:58:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"29fe91bf852c5962a595f7e59138d3124a1208c4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2021-09-13 14:13:40.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"37b3261f8e949b32bbe885b70bd2faa95cc473aa","author":{"_account_id":20778,"name":"Damien Ciabrini","email":"dciabrin@redhat.com","username":"dciabrin"},"date":"2021-09-13 14:14:58.000000000","message":"Patch Set 2: Code-Review-1\n\nLet\u0027s make sure we test it thoroughly and everybody understands the impact of changing pacemaker configs. First thing that comes to mind is that this requires a full restart of the resource everywhere, so full service outage.","accounts_in_message":[],"_revision_number":2},{"id":"95e7d3ed878455dd763b5faffc3cc9de5fd250c4","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2021-09-13 14:17:20.000000000","message":"Patch Set 2:\n\n\u003e Patch Set 2: Code-Review-1\n\u003e \n\u003e Let\u0027s make sure we test it thoroughly and everybody understands the impact of changing pacemaker configs. First thing that comes to mind is that this requires a full restart of the resource everywhere, so full service outage.\n\nwell..... we can\u0027t either keep containers with the wrong options, really. Until now, it was working by chance, because most of the things are done in tripleo-heat-templates.\nBut if, for any reason, something is tempering the labels, even a container restart won\u0027t do anything.","accounts_in_message":[],"_revision_number":2},{"id":"4dd377af7df83cde26b9fc3925169e29a328be9e","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2021-09-13 14:18:33.000000000","message":"Patch Set 2: Workflow-1\n\nLet\u0027s get proper reviews on that one, since it apparently has potential huge impact.","accounts_in_message":[],"_revision_number":2},{"id":"76f7da58cebf0e9f08301962b886cb8e058023bf","author":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"date":"2021-09-13 15:15:53.000000000","message":"Patch Set 2:\n\n\u003e Patch Set 2: Code-Review-1\n\u003e \n\u003e Let\u0027s make sure we test it thoroughly and everybody understands the impact of changing pacemaker configs. First thing that comes to mind is that this requires a full restart of the resource everywhere, so full service outage.\n\nIf we go from without this patch to with this patch, does it take immediate effect or does it only affect it once the restarts are done? Will it affect the running environment?\n\nDo we have something in THT which is setting this anyway, so we currently have a race between the two?","accounts_in_message":[],"_revision_number":2},{"id":"f34886d530f4f0810d6f33820cf77ae13fc2a81d","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-09-13 16:53:47.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/4e206ac5218849c698a8155a61056a31 : SUCCESS in 2h 35m 35s\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset035 https://review.rdoproject.org/zuul/build/490efbb9c0834316851ee3c8c0bf9bcb : FAILURE in 1h 28m 44s","accounts_in_message":[],"_revision_number":2},{"id":"1483a2f67492c4582301b301aaace74967547b2f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-13 17:05:37.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/faf6f520aa354058bd0092b73028acd3 : SUCCESS in 4m 46s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8116440f60b1438d8ef5fb9421a7e83e : SUCCESS in 4m 50s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e70258bd8b5a4e2380938590794d20f5 : SUCCESS in 4m 03s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/5c33f987db84494fa4f1244cfd2c89bc : SUCCESS in 26m 02s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/08b32e887e564dc68bd8791d558ab2ce : SUCCESS in 22m 42s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d36c93b45aa44077ace5735d47763731 : SUCCESS in 21m 41s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6967cedac67e4ae8a83dc6bb695a4e14 : SUCCESS in 24m 35s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/59ec5c6748834291baa9a5d9e82ba62a : SUCCESS in 21m 39s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/b09a0fa9c21b47afae17a45c9476d1d8 : SUCCESS in 2h 46m 34s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/28fcdede4103434dbc662ea5a68f20c4 : SUCCESS in 46m 02s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/b6dca2c5ad064c36898e494229f9fe59 : SUCCESS in 1h 32m 19s\n- tripleo-ci-centos-8-scenario001-standalone https://zuul.opendev.org/t/openstack/build/36c74df0f9b343518616805ec769f754 : SUCCESS in 1h 31m 14s\n- tripleo-ci-centos-8-scenario002-standalone https://zuul.opendev.org/t/openstack/build/976a1fd43f114cd69a5eeb5b21628783 : SUCCESS in 1h 18m 59s\n- tripleo-ci-centos-8-scenario004-standalone https://zuul.opendev.org/t/openstack/build/ed7ffdb7711940a78d6042069dc973c6 : SUCCESS in 1h 14m 48s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/f3c5913c268b45868bda31e12c9f111d : SUCCESS in 1h 17m 34s\n- tripleo-ci-centos-8-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/bd8e8786399841f4a185cadf078100fa : SUCCESS in 2h 39m 15s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/eda2a10a56ed44b6b9114fadc250e1f7 : SUCCESS in 1h 23m 14s\n- tripleo-ci-centos-8-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/0d0317f53bc347c5be58f188aeeededf : SUCCESS in 1h 12m 50s\n- tripleo-ci-centos-8-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/664ce8005ade4cb185a239f48c4e51ea : SUCCESS in 1h 29m 18s","accounts_in_message":[],"_revision_number":2},{"id":"84a1ca97dbd36c3e1886214511b4546d1438b296","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2021-09-14 05:22:12.000000000","message":"Patch Set 2:\n\ncheck-rdo\n\nissue at node provisionning:\nhttps://logserver.rdoproject.org/74/808774/2/openstack-check/tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset035/490efbb/logs/undercloud/home/zuul/overcloud_node_provision.log.txt.gz","accounts_in_message":[],"_revision_number":2},{"id":"c1ab7e75e2aaa7a9cbd0ce3349ccaaa108214ce0","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-09-14 06:57:28.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/6f4b1aa311c149609563a484645b64c6 : FAILURE in 1h 34m 38s\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset035 https://review.rdoproject.org/zuul/build/9493e2f3571e4653b8b6c320a1c3ded2 : FAILURE in 1h 24m 16s","accounts_in_message":[],"_revision_number":2},{"id":"403038e7b8c20fbef1ab3fdc41742396d772d94f","author":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"date":"2021-09-14 08:16:23.000000000","message":"Patch Set 2:\n\nI suggest two improvements:\n1) Add a proper analysis as to why this is a problem now (given that this was the status quo since pike/osp12) and it was not before.\n2) Add a note in deployment/README.rst what service owner should do in general in regards to bind mounts. Currently this is totally cargo-culted or done via trial-error all over tht with very little guidance.","accounts_in_message":[],"_revision_number":2},{"id":"8f9f08587f274feb654c670e35de3f73498f6ca5","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2021-09-14 08:30:14.000000000","message":"Patch Set 2: Code-Review+2\n\nwell, that is a must have flag, unless there is other options?\nalthough upgrade impact is restarting everything on controllers (since that\u0027s HA only components) - wouldn\u0027t it be handled just fine, in a 1-by-1 serial fashion?","accounts_in_message":[],"_revision_number":2},{"id":"cd78ab73dedb44218bf00078910690e5096de573","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2021-09-14 09:14:46.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"09f0941726ede1c4c24392696226368c25e19ca1","author":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"date":"2021-09-14 09:52:53.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"e1a456945aeffa955eef4a7e1f9a0c84ac15793f","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2021-09-14 10:58:48.000000000","message":"Patch Set 3:\n\n(1 comment)\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":3},{"id":"bfe1d7374bfb6d93e9e9e2098f389a3446c0e02c","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2021-09-14 10:59:37.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"1b39c1fcd3d814cd6170dd44c80446e69c93dd19","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-09-14 13:35:58.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/9093c631ee00430da2486cbed86a2f15 : FAILURE in 2h 35m 51s\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset035 https://review.rdoproject.org/zuul/build/9952f15b2b3a4d1d8d7fc89e8e30dd58 : FAILURE in 1h 19m 10s","accounts_in_message":[],"_revision_number":4},{"id":"753df17d4d2428bb01371202b1032be6bd108fb1","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2021-09-14 13:37:09.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":4},{"id":"e63d42f731872e7941be1dcdbe170c35a4bb4e91","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-14 14:07:09.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/00b71b6827364e598bbdf56089eb2ad7 : SUCCESS in 4m 42s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4aff686c728c47a8bbcc97d3823cc1ba : SUCCESS in 4m 25s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6a111c25e2ea401c802b4a4c692a4cf2 : SUCCESS in 4m 23s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/6263df7ddcdd46f58b32634ea143448b : SUCCESS in 24m 46s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/82b7e73e7f1943afb114a860a0f02f60 : SUCCESS in 22m 34s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0dd570bae5094dff97f90a79a26b51c4 : SUCCESS in 20m 44s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d19a0bc94a574c188ddd4d792f7f2a73 : SUCCESS in 21m 56s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/483cd51391cd4c87b83958a62267fe04 : SUCCESS in 22m 07s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/54fc4744907a475d825c17ce868e0ac0 : SUCCESS in 3h 03m 51s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/c28fd448aad64a049f651f1571dcbd97 : SUCCESS in 45m 17s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/64125270dbec410b9b8aca44a6e6b5c6 : SUCCESS in 1h 36m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7b3fb4e1218e4998acc70d18b3d86ea5 : SUCCESS in 6m 08s\n- tripleo-ci-centos-8-scenario001-standalone https://zuul.opendev.org/t/openstack/build/d787e980935341f089ce08d1f94ce66c : SUCCESS in 1h 34m 44s\n- tripleo-ci-centos-8-scenario002-standalone https://zuul.opendev.org/t/openstack/build/29b60228565642baa4bddd3d47a31f6b : SUCCESS in 1h 15m 54s\n- tripleo-ci-centos-8-scenario004-standalone https://zuul.opendev.org/t/openstack/build/fb353adfeae949b597cc6a7ed861c116 : SUCCESS in 1h 17m 00s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/ba2e999ace0646d181dfd4e0dfb5d158 : SUCCESS in 1h 08m 01s\n- tripleo-ci-centos-8-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/3fe5118ee6684fd3947cb72aa1005ba7 : SUCCESS in 2h 43m 36s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/f43b51f9248943ee9dcb4879e4ea42f8 : SUCCESS in 1h 29m 05s\n- tripleo-ci-centos-8-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/c36beb8eddae4666934697113c2f6239 : SUCCESS in 1h 16m 36s\n- tripleo-ci-centos-8-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/037467e6ace347f886b1f7b5d6db264f : SUCCESS in 1h 29m 18s","accounts_in_message":[],"_revision_number":4},{"id":"d4f039d0e0fed57965601de09a1eb6a43dd832fb","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2021-09-14 14:57:35.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"b371979cb708349efe32732502446a0ba338d8b7","author":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"date":"2021-09-14 20:02:18.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"9b2cbcaf4d4e9b01c06388211ddca37238848e68","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-14 20:04:24.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"2c8c271c58079f37070591e9a51f5a708ebf9774","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-15 07:05:41.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6f06e73a8a6341f8bbd33301088ae6bd : SUCCESS in 5m 06s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/97d0134342b74c0881b53f957e8e0600 : SUCCESS in 4m 06s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5b45eb2e9806418bace849bae0f27327 : SUCCESS in 4m 35s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/272aada2dc394b63ac7ec4672ddf0f94 : SUCCESS in 23m 35s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/6c5842e5120a471a96e8ecb60a78fee1 : SUCCESS in 22m 41s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c40559a4dd634e28a983c068a7281eb9 : SUCCESS in 24m 08s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6160c4f60ff34aa3b378873b884693e9 : SUCCESS in 24m 14s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/20305ae311f844158447d19661083393 : SUCCESS in 3h 15m 56s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/b323749061e64868b9a6855071f71081 : SUCCESS in 47m 31s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/51cd926b74aa43dbaec1a6cbcfd8495c : SUCCESS in 1h 38m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/38d2d58f33244ecd92e2e5133510c6a7 : SUCCESS in 6m 06s\n- tripleo-ci-centos-8-scenario001-standalone https://zuul.opendev.org/t/openstack/build/7a05228b9b2e4acaae99457b03fafa55 : SUCCESS in 1h 31m 00s\n- tripleo-ci-centos-8-scenario002-standalone https://zuul.opendev.org/t/openstack/build/6ccb1313bbc8408ea3fb47693ed345b0 : SUCCESS in 1h 17m 12s\n- tripleo-ci-centos-8-scenario004-standalone https://zuul.opendev.org/t/openstack/build/5026d3cea2a94b229bddd2c36d88fe74 : SUCCESS in 1h 15m 49s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/80efadd016ac49fc8e3724bc438d1398 : SUCCESS in 1h 18m 05s\n- tripleo-ci-centos-8-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/b01030e117c541c6a6fa4481e5d1b150 : SUCCESS in 2h 52m 34s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/cd10bb9bf7f34b139f82cd601289118c : SUCCESS in 1h 29m 46s\n- tripleo-ci-centos-8-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/f82273ce1500473293fda1403e072247 : SUCCESS in 1h 07m 18s","accounts_in_message":[],"_revision_number":4},{"id":"f77ad5acda9e9c3769aec3507e6987780e87c481","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-15 07:05:41.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"88a0afdb672a995b2634d36540ef671849f29282","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-15 07:08:50.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/97c471a6cfe645d0a7ce761994c1b3b4 : SUCCESS in 1m 57s","accounts_in_message":[],"_revision_number":4},{"id":"d04f70af8e62fd58203bec872f959ce009a9a54d","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2021-09-15 07:20:47.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"e91aac28229f7b4eb5b054f3b88d9839ffc50fdb","revisions":{"a3edcc5332300e7132f6c8ab682e45c82fbba149":{"kind":"REWORK","_number":1,"created":"2021-09-13 13:58:02.000000000","uploader":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"ref":"refs/changes/74/808774/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tripleo","ref":"refs/changes/74/808774/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/74/808774/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/74/808774/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/74/808774/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tripleo refs/changes/74/808774/1"}}},"commit":{"parents":[{"commit":"71bc89e33091441c5b12c07dc5f6bf031c1284bb","subject":"Merge \"Horizon: Enable heat-dashboard configurations\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/71bc89e33091441c5b12c07dc5f6bf031c1284bb"}]}],"author":{"name":"Cédric Jeanneret","email":"cjeanner@redhat.com","date":"2021-09-13 13:54:34.000000000","tz":120},"committer":{"name":"Cédric Jeanneret","email":"cjeanner@redhat.com","date":"2021-09-13 13:54:34.000000000","tz":120},"subject":"Add missing \"z\" flag for specific mounts","message":"Add missing \"z\" flag for specific mounts\n\nIn order to ensure consistency with everything linked to SELinux and\nbind-mounts, we must ensure we relabel locations/files when we mount\nthem, especially with read-write rights.\n\nChange-Id: Ic3e427156fc82c524c763d1896937fcc3c49fabb\nCloses-Bug: #1943459\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/a3edcc5332300e7132f6c8ab682e45c82fbba149"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/a3edcc5332300e7132f6c8ab682e45c82fbba149"}]},"branch":"refs/heads/master"},"250fb294160d9475637d1836c85a3638c6c3e72b":{"kind":"REWORK","_number":2,"created":"2021-09-13 14:13:40.000000000","uploader":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"ref":"refs/changes/74/808774/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tripleo","ref":"refs/changes/74/808774/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/74/808774/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/74/808774/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/74/808774/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tripleo refs/changes/74/808774/2"}}},"commit":{"parents":[{"commit":"71bc89e33091441c5b12c07dc5f6bf031c1284bb","subject":"Merge \"Horizon: Enable heat-dashboard configurations\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/71bc89e33091441c5b12c07dc5f6bf031c1284bb"}]}],"author":{"name":"Cédric Jeanneret","email":"cjeanner@redhat.com","date":"2021-09-13 13:54:34.000000000","tz":120},"committer":{"name":"Cédric Jeanneret","email":"cjeanner@redhat.com","date":"2021-09-13 14:13:34.000000000","tz":120},"subject":"Add missing \"z\" flag for specific mounts","message":"Add missing \"z\" flag for specific mounts\n\nIn order to ensure consistency with everything linked to SELinux and\nbind-mounts, we must ensure we relabel locations/files when we mount\nthem, especially with read-write rights.\n\nChange-Id: Ic3e427156fc82c524c763d1896937fcc3c49fabb\nCloses-Bug: #1943459\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/250fb294160d9475637d1836c85a3638c6c3e72b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/250fb294160d9475637d1836c85a3638c6c3e72b"}]},"branch":"refs/heads/master"},"e8c4e9304fa803288b38bb81a96c6183e88fdd93":{"kind":"REWORK","_number":3,"created":"2021-09-14 09:14:46.000000000","uploader":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"ref":"refs/changes/74/808774/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tripleo","ref":"refs/changes/74/808774/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/74/808774/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/74/808774/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/74/808774/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tripleo refs/changes/74/808774/3"}}},"commit":{"parents":[{"commit":"71bc89e33091441c5b12c07dc5f6bf031c1284bb","subject":"Merge \"Horizon: Enable heat-dashboard configurations\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/71bc89e33091441c5b12c07dc5f6bf031c1284bb"}]}],"author":{"name":"Cédric Jeanneret","email":"cjeanner@redhat.com","date":"2021-09-13 13:54:34.000000000","tz":120},"committer":{"name":"Cédric Jeanneret","email":"cjeanner@redhat.com","date":"2021-09-14 09:14:32.000000000","tz":120},"subject":"Add missing \"z\" flag for specific mounts","message":"Add missing \"z\" flag for specific mounts\n\nDepending on the host history, it may happen some directory content\ndon\u0027t have the correct SELinux type. This has been seen with OVN\nservice, during a Queens -\u003e Train FFU:\n\nwhile the /var/lib/openvswitch/ovn directory had the correct\ncontainer_file_t type, some files in this location were typed with\nopenvswitch_var_lib_t, leading to errors during the deploy part of the\nupgrade (after the OS upgrade, when the deploy is running on the cleaned\nhost).\nThe specific issue depends on the actual files with the wrong label, but\nusually it involves a container crash/error, leading to a deploy error,\nand a manual intervention in order to correct the SELinux type in the\nlocation.\n\nThis situation may happen when first deployed on Queens, since it was\nusing Docker. For the records, back then Docker Daemon was configured in\norder to disable the SELinux support, so it didn\u0027t really care about\nlabels; but the situation is different with Podman, and we have a full\nSELinux support at all levels on the OS, leading to the issue.\n\nFor the records, tripleo-heat-templates as well as tripleo-ansible are\nsetting the \"setype: container_file_t\" on the directories, but we don\u0027t\nuse the \"recurse: true\" in order to avoid performance issues - some\nlocations might be huge, and it would take too much time to relabel\neverything via ansible.\n\nThis patch aims to converge all the mounts to the same options, and\nensure no SELinux denial can prevent the actual container startup and\nfunction.\n\nChange-Id: Ic3e427156fc82c524c763d1896937fcc3c49fabb\nCloses-Bug: #1943459\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/e8c4e9304fa803288b38bb81a96c6183e88fdd93"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/e8c4e9304fa803288b38bb81a96c6183e88fdd93"}]},"branch":"refs/heads/master"},"e91aac28229f7b4eb5b054f3b88d9839ffc50fdb":{"kind":"REWORK","_number":4,"created":"2021-09-14 10:59:37.000000000","uploader":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"ref":"refs/changes/74/808774/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tripleo","ref":"refs/changes/74/808774/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/74/808774/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/74/808774/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/74/808774/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tripleo refs/changes/74/808774/4"}}},"commit":{"parents":[{"commit":"71bc89e33091441c5b12c07dc5f6bf031c1284bb","subject":"Merge \"Horizon: Enable heat-dashboard configurations\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/71bc89e33091441c5b12c07dc5f6bf031c1284bb"}]}],"author":{"name":"Cédric Jeanneret","email":"cjeanner@redhat.com","date":"2021-09-13 13:54:34.000000000","tz":120},"committer":{"name":"Cédric Jeanneret","email":"cjeanner@redhat.com","date":"2021-09-14 10:59:31.000000000","tz":120},"subject":"Add missing \"z\" flag for specific mounts","message":"Add missing \"z\" flag for specific mounts\n\nDepending on the host history, it may happen some directory content\ndon\u0027t have the correct SELinux type. This has been seen with OVN\nservice, during a Queens -\u003e Train FFU:\n\nwhile the /var/lib/openvswitch/ovn directory had the correct\ncontainer_file_t type, some files in this location were typed with\nopenvswitch_var_lib_t, leading to errors during the deploy part of the\nupgrade (after the OS upgrade, when the deploy is running on the cleaned\nhost).\nThe specific issue depends on the actual files with the wrong label, but\nusually it involves a container crash/error, leading to a deploy error,\nand a manual intervention in order to correct the SELinux type in the\nlocation.\n\nThis situation may happen when first deployed on Queens, since it was\nusing Docker. For the records, back then Docker Daemon was configured in\norder to disable the SELinux support, so it didn\u0027t really care about\nlabels; but the situation is different with Podman, and we have a full\nSELinux support at all levels on the OS, leading to the issue.\n\nFor the records, tripleo-heat-templates as well as tripleo-ansible are\nsetting the \"setype: container_file_t\" on the directories, but we don\u0027t\nuse the \"recurse: true\" in order to avoid performance issues - some\nlocations might be huge, and it would take too much time to relabel\neverything via ansible.\n\nThis patch aims to converge all the mounts to the same options, and\nensure no SELinux denial can prevent the actual container startup and\nfunction.\n\nChange-Id: Ic3e427156fc82c524c763d1896937fcc3c49fabb\nCloses-Bug: #1943459\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/e91aac28229f7b4eb5b054f3b88d9839ffc50fdb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/e91aac28229f7b4eb5b054f3b88d9839ffc50fdb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
