)]}'
{"id":"openstack%2Fpuppet-tripleo~809132","triplet_id":"openstack%2Fpuppet-tripleo~stable%2Fwallaby~Ic3e427156fc82c524c763d1896937fcc3c49fabb","project":"openstack/puppet-tripleo","branch":"stable/wallaby","topic":"selinux/missing-flags-stable/wallaby","hashtags":[],"change_id":"Ic3e427156fc82c524c763d1896937fcc3c49fabb","subject":"Add missing \"z\" flag for specific mounts","status":"MERGED","created":"2021-09-15 07:21:04.000000000","updated":"2021-09-16 19:14:34.000000000","submitted":"2021-09-16 19:12:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"809132-selinux/missing-flags-stable/wallaby","cherry_pick_of_change":808774,"cherry_pick_of_patch_set":4,"meta_rev_id":"244f6ed954911851c901078df227e8ea235d176c","_number":809132,"virtual_id_number":809132,"owner":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-16 19:12:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-09-16 09:20:16.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"all":[{"value":2,"date":"2021-09-16 13:21:53.000000000","_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":1,"date":"2021-09-16 07:00:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":2,"date":"2021-09-16 07:42:05.000000000","_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"},{"value":1,"date":"2021-09-16 13:57:23.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-15 10:00:52.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-15 11:08:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-16 07:00:22.000000000","updated_by":{"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"},"reviewer":{"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"},"state":"REVIEWER"},{"updated":"2021-09-16 07:42:05.000000000","updated_by":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"reviewer":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"state":"REVIEWER"},{"updated":"2021-09-16 13:21:53.000000000","updated_by":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"reviewer":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"state":"REVIEWER"},{"updated":"2021-09-16 13:57:23.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"e69213b9093bf12585ae888d748b7125be2566e9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2021-09-15 07:21:04.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"f5c958215c3d89d29dec89965a27a70b037a8a52","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-09-15 10:00:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/34000cb55dfb4905a4a48f026fb28448 : SUCCESS in 2h 17m 09s\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset035 https://review.rdoproject.org/zuul/build/6065b36c6e944640a29a3a83a6841807 : SUCCESS in 2h 38m 35s","accounts_in_message":[],"_revision_number":1},{"id":"2b00d2e464ff54bd2896f709c11cb73f85e45fe8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-15 11:08:37.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c14c8ec32f124b99adf4ceeaa1c814db : SUCCESS in 4m 38s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f2284caa375343348d0541bdaf64c614 : SUCCESS in 6m 10s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/bbaaf7d6407c4ca686991f00c3f62472 : SUCCESS in 26m 21s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6425fd26e6c84512baf7abbc9867c5f0 : SUCCESS in 22m 53s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/01b8c7474c4e41c9b40b79785b963be4 : SUCCESS in 22m 03s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/485b2fc70eed461fa5e8e62ea7f8b97b : SUCCESS in 3h 37m 21s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/65e155b7664a4dc3a36668166d3c4313 : SUCCESS in 1h 12m 40s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/69a1548f2b6b45288dfdd2a150070b5a : SUCCESS in 2h 03m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8d4b2f651cc747f5bac47c3731f8bbf0 : SUCCESS in 5m 04s\n- tripleo-ci-centos-8-scenario001-standalone https://zuul.opendev.org/t/openstack/build/d2a0da5e83624482858ce8db7a404c47 : SUCCESS in 1h 47m 26s\n- tripleo-ci-centos-8-scenario002-standalone https://zuul.opendev.org/t/openstack/build/fc92db2eb9a54bac9c1015f4ae1d7466 : SUCCESS in 1h 30m 03s\n- tripleo-ci-centos-8-scenario004-standalone https://zuul.opendev.org/t/openstack/build/9ff4ac3c233b4f6db1b3473c7e7c559e : FAILURE in 20m 11s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/d523c91236394875afb4562780c84955 : FAILURE in 20m 07s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/a76bae4e600a4ed2a410d88f825cb617 : SUCCESS in 3h 09m 09s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-wallaby https://zuul.opendev.org/t/openstack/build/a4c264fc4026401895bd5a233a5f2499 : FAILURE in 18m 49s\n- tripleo-ci-centos-8-undercloud-upgrade-wallaby https://zuul.opendev.org/t/openstack/build/0907ed441c5f4ebf97e468638e2b0f8f : SUCCESS in 1h 29m 52s\n- tripleo-ci-centos-8-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/76a5cd05fbbd4fa9bd261c8e545c1577 : FAILURE in 19m 07s","accounts_in_message":[],"_revision_number":1},{"id":"0084d659ef1065471559b7570d3686e7833a54f9","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2021-09-15 12:49:35.000000000","message":"Patch Set 1:\n\nrecheck\n\nNot related to the patch I\u0027d say:\n\"No matching distribution found for shyaml\"","accounts_in_message":[],"_revision_number":1},{"id":"c583946a6bec552e8341321c5e7e2ab3f0760ca8","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-09-15 15:52:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/00bc3e5fa9514be297665f7002ef81e3 : SUCCESS in 3h 02m 41s\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset035 https://review.rdoproject.org/zuul/build/ec9bfcb1657a4eaf88c066f731649468 : SUCCESS in 2h 56m 28s","accounts_in_message":[],"_revision_number":1},{"id":"084f3befac557eb3f72eee4f65a68a3181093d71","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-15 16:05:25.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b1b5cefbf7e34df8aef3042626a88a09 : SUCCESS in 4m 11s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b410d322f3e74c2aaf0b09eefb6180c1 : SUCCESS in 4m 05s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/bd7625b8018842c88ae195333d65e181 : SUCCESS in 24m 00s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e4d0090265f04c5eb832f02a8612335d : SUCCESS in 21m 07s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/47c797e9e49e49e8ada2b17f7c397d83 : SUCCESS in 23m 17s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/b14cad453fc7438da9c465a8ea6f34f4 : SUCCESS in 3h 10m 19s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/51b044113377484cb0c6e3cc498a102f : SUCCESS in 1h 15m 16s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/15cb28f1c1ae48f4b27fd609dd24113c : SUCCESS in 1h 45m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/99e4667bfb954aa6bb6995949eef0b07 : SUCCESS in 4m 17s\n- tripleo-ci-centos-8-scenario001-standalone https://zuul.opendev.org/t/openstack/build/7cba4b2ce44f40199cff70c3b4351d32 : SUCCESS in 1h 37m 24s\n- tripleo-ci-centos-8-scenario002-standalone https://zuul.opendev.org/t/openstack/build/b58ffbf224544333b843a0c0f3581f14 : SUCCESS in 1h 25m 34s\n- tripleo-ci-centos-8-scenario004-standalone https://zuul.opendev.org/t/openstack/build/9ebc602c0a4a430e8dd7ac27782dbab7 : SUCCESS in 1h 27m 44s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/0d961550f4f04ad79b966888abdd6da3 : SUCCESS in 1h 35m 54s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/b7094c50950645a1975dd436463a6823 : SUCCESS in 2h 46m 46s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-wallaby https://zuul.opendev.org/t/openstack/build/68eb124c88ad477fa3384c04cc81f3c4 : FAILURE in 48m 30s\n- tripleo-ci-centos-8-undercloud-upgrade-wallaby https://zuul.opendev.org/t/openstack/build/6ad9ffa088554374ad071b1ded3719e5 : SUCCESS in 1h 23m 26s\n- tripleo-ci-centos-8-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/6a213358f5724ef0b5ac697563a5e50c : SUCCESS in 1h 35m 30s","accounts_in_message":[],"_revision_number":1},{"id":"a51661a2b15a1b21b88a53a6e03108d9f3284b47","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2021-09-16 06:44:41.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"a0e3ea772aecadf5064c78d0dd4a063a979b2af7","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2021-09-16 06:44:59.000000000","message":"Patch Set 2:\n\nlast failure wasn\u0027t due to the change, again:\n\n2021-09-15 14:56:22 | 2021-09-15 14:56:22.545409 | bc764e10-1be7-cd47-5d81-000000000dc4 |      FATAL | Container image prepare | undercloud | error\u003d{\"changed\": false, \"error\": \"Not found image: http://104.130.207.112:5001/v2/tripleowallaby/openstack-neutron-server/manifests/2bb20e8908ea60d8285dee886ef0be66\", \"msg\": \"Error running container image prepare: Not found image: http://104.130.207.112:5001/v2/tripleowallaby/openstack-neutron-server/manifests/2bb20e8908ea60d8285dee886ef0be66\", \"params\": {}, \"success\": false}","accounts_in_message":[],"_revision_number":2},{"id":"7f4a72647a4df4e432338bccddfc5d9375eb1c11","author":{"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"},"date":"2021-09-16 07:00:22.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"f29e9c0401f52fe220e947e79059f0751303436f","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2021-09-16 07:42:05.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"790b4a6864dbf9e17dcace3b0b3461e56b7ec83f","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-09-16 09:20:16.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/08a95bbf7abd439a8f582cf606a43350 : SUCCESS in 2h 23m 14s\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset035 https://review.rdoproject.org/zuul/build/185e16a2fb5c43a8a561bde5f1175c3d : SUCCESS in 2h 34m 28s","accounts_in_message":[],"_revision_number":2},{"id":"ad0139eba44ed2a59a763e3bde4d6a350cd66246","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-16 10:39:19.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5be897f5d2c04a35befa0d2278c1e173 : SUCCESS in 4m 45s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/575ab03c791b4084a8cead46f90e34ca : SUCCESS in 5m 24s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/40bb74afeab84585ab585a013dc99545 : SUCCESS in 23m 14s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cce4ec1adfc942a98f26f294486a4b49 : SUCCESS in 21m 35s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1bc3586eb7fc44fdaaf73f49c93fe249 : SUCCESS in 21m 47s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/bd4882a844ed448f8342369bd8d61c23 : SUCCESS in 3h 19m 37s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/342c96330ad1493d9e110b9febca3539 : SUCCESS in 1h 07m 28s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/2ed33c56c311468b87e8c1cb9099d714 : SUCCESS in 1h 48m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/60a6016f5ec246c0b2d6a1a2c65f9741 : SUCCESS in 4m 30s\n- tripleo-ci-centos-8-scenario001-standalone https://zuul.opendev.org/t/openstack/build/02c2dc12b2f04fdf87cb5b328da8d3e4 : SUCCESS in 1h 38m 19s\n- tripleo-ci-centos-8-scenario002-standalone https://zuul.opendev.org/t/openstack/build/e781a4e861394edda32d4c0a770f5eed : SUCCESS in 1h 30m 39s\n- tripleo-ci-centos-8-scenario004-standalone https://zuul.opendev.org/t/openstack/build/f473d8549ab049d597a1c13a18c49daf : SUCCESS in 1h 31m 12s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/9bf685425de848b3bd4bd76ab9ca658e : SUCCESS in 1h 26m 05s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/b4a29f4a7b0b40ca9e0f66dbc4e409b1 : SUCCESS in 3h 08m 19s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-wallaby https://zuul.opendev.org/t/openstack/build/b3661657b637441d9feae877f902b0c6 : SUCCESS in 1h 38m 35s\n- tripleo-ci-centos-8-undercloud-upgrade-wallaby https://zuul.opendev.org/t/openstack/build/d5868001ffe5405baaa21b3315957a17 : SUCCESS in 1h 34m 43s\n- tripleo-ci-centos-8-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/aacca9228a3a4c5f9baa39ae1718d2a1 : SUCCESS in 1h 36m 38s","accounts_in_message":[],"_revision_number":2},{"id":"5ce03f1a40870f2f022e6813b6de58e2c14466b6","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2021-09-16 13:21:53.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"220348a8101435d5c4be33ff6fbf046a1f411fb9","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-09-16 13:57:23.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"a7bbb65c63ef4a5b863297fbd8e2c11f41d7402d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-16 14:03:55.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"031dee30381dfffdf2221308b73d9c117bb44460","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-16 19:12:57.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5985229f6ff54c048271e99df7c80129 : SUCCESS in 5m 18s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7c5b24e2b1be4bb4a56cd77348eb535e : SUCCESS in 4m 45s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/b97eb1e198a74237b250b8c2f7d28c39 : SUCCESS in 23m 41s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0600c19c0357473c801848435b026b46 : SUCCESS in 19m 04s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/df41f6521aae42ba89fe43e86aaf9fb1 : SUCCESS in 3h 21m 46s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/761dd3dd415c4f76a9259008b012afaa : SUCCESS in 1h 09m 46s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/4000e9c7d09947979daa57b36a7db4ae : SUCCESS in 1h 49m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7fcb56f529da4124b4aac87c1b349e84 : SUCCESS in 5m 30s\n- tripleo-ci-centos-8-scenario001-standalone https://zuul.opendev.org/t/openstack/build/09583cea75144ed7b6764fb24e8ad363 : SUCCESS in 1h 39m 54s\n- tripleo-ci-centos-8-scenario002-standalone https://zuul.opendev.org/t/openstack/build/d5206884cc754b50a74421e09613be45 : SUCCESS in 1h 27m 41s\n- tripleo-ci-centos-8-scenario004-standalone https://zuul.opendev.org/t/openstack/build/c10385055ae9492bbd5eafc66cdbc57a : SUCCESS in 1h 26m 50s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/134e090c52b24b46ad6f2d0a4dbb697a : SUCCESS in 1h 29m 34s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/97b7458dac694bda804e57d70ff9d770 : SUCCESS in 3h 02m 24s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-wallaby https://zuul.opendev.org/t/openstack/build/be11cfbf9f2249ebb600efc31c39d1e8 : SUCCESS in 1h 35m 00s\n- tripleo-ci-centos-8-undercloud-upgrade-wallaby https://zuul.opendev.org/t/openstack/build/42ef791fb2aa439daaf9b15f0291b179 : SUCCESS in 1h 29m 42s","accounts_in_message":[],"_revision_number":2},{"id":"4a7e682e77a01e567e39577996e95c9099c3754e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-16 19:12:57.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"244f6ed954911851c901078df227e8ea235d176c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-16 19:14:34.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/377b40183db74c9690e74da560fc7cc1 : SUCCESS in 1m 08s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"848f2acd5b176dc4d0d82bc2518a9e0e330f9c56","revisions":{"fef5d1573ddc088e21a0d1384150be79be95929b":{"kind":"REWORK","_number":1,"created":"2021-09-15 07:21:04.000000000","uploader":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"ref":"refs/changes/32/809132/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tripleo","ref":"refs/changes/32/809132/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/32/809132/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/32/809132/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/32/809132/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tripleo refs/changes/32/809132/1"}}},"commit":{"parents":[{"commit":"9827b88ca4fed013d5f1a0b7385d1fdd4e9ea92e","subject":"Merge \"Replace deprecated is_string/is_array/is_hash method\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/9827b88ca4fed013d5f1a0b7385d1fdd4e9ea92e"}]}],"author":{"name":"Cédric Jeanneret","email":"cjeanner@redhat.com","date":"2021-09-13 13:54:34.000000000","tz":120},"committer":{"name":"Cedric Jeanneret","email":"cjeanner@redhat.com","date":"2021-09-15 07:21:04.000000000","tz":0},"subject":"Add missing \"z\" flag for specific mounts","message":"Add missing \"z\" flag for specific mounts\n\nDepending on the host history, it may happen some directory content\ndon\u0027t have the correct SELinux type. This has been seen with OVN\nservice, during a Queens -\u003e Train FFU:\n\nwhile the /var/lib/openvswitch/ovn directory had the correct\ncontainer_file_t type, some files in this location were typed with\nopenvswitch_var_lib_t, leading to errors during the deploy part of the\nupgrade (after the OS upgrade, when the deploy is running on the cleaned\nhost).\nThe specific issue depends on the actual files with the wrong label, but\nusually it involves a container crash/error, leading to a deploy error,\nand a manual intervention in order to correct the SELinux type in the\nlocation.\n\nThis situation may happen when first deployed on Queens, since it was\nusing Docker. For the records, back then Docker Daemon was configured in\norder to disable the SELinux support, so it didn\u0027t really care about\nlabels; but the situation is different with Podman, and we have a full\nSELinux support at all levels on the OS, leading to the issue.\n\nFor the records, tripleo-heat-templates as well as tripleo-ansible are\nsetting the \"setype: container_file_t\" on the directories, but we don\u0027t\nuse the \"recurse: true\" in order to avoid performance issues - some\nlocations might be huge, and it would take too much time to relabel\neverything via ansible.\n\nThis patch aims to converge all the mounts to the same options, and\nensure no SELinux denial can prevent the actual container startup and\nfunction.\n\nChange-Id: Ic3e427156fc82c524c763d1896937fcc3c49fabb\nCloses-Bug: #1943459\n(cherry picked from commit e8c4e9304fa803288b38bb81a96c6183e88fdd93)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/fef5d1573ddc088e21a0d1384150be79be95929b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/fef5d1573ddc088e21a0d1384150be79be95929b"}]},"branch":"refs/heads/stable/wallaby"},"848f2acd5b176dc4d0d82bc2518a9e0e330f9c56":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-09-16 06:44:41.000000000","uploader":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"ref":"refs/changes/32/809132/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tripleo","ref":"refs/changes/32/809132/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/32/809132/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/32/809132/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/32/809132/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tripleo refs/changes/32/809132/2"}}},"commit":{"parents":[{"commit":"b67596f1b58ed6bd07ba36739ad2e223b4434a16","subject":"[c9] wrappers should start containers with host\u0027s cgroupns","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/b67596f1b58ed6bd07ba36739ad2e223b4434a16"}]}],"author":{"name":"Cédric Jeanneret","email":"cjeanner@redhat.com","date":"2021-09-13 13:54:34.000000000","tz":120},"committer":{"name":"Cedric Jeanneret","email":"cjeanner@redhat.com","date":"2021-09-16 06:44:41.000000000","tz":0},"subject":"Add missing \"z\" flag for specific mounts","message":"Add missing \"z\" flag for specific mounts\n\nDepending on the host history, it may happen some directory content\ndon\u0027t have the correct SELinux type. This has been seen with OVN\nservice, during a Queens -\u003e Train FFU:\n\nwhile the /var/lib/openvswitch/ovn directory had the correct\ncontainer_file_t type, some files in this location were typed with\nopenvswitch_var_lib_t, leading to errors during the deploy part of the\nupgrade (after the OS upgrade, when the deploy is running on the cleaned\nhost).\nThe specific issue depends on the actual files with the wrong label, but\nusually it involves a container crash/error, leading to a deploy error,\nand a manual intervention in order to correct the SELinux type in the\nlocation.\n\nThis situation may happen when first deployed on Queens, since it was\nusing Docker. For the records, back then Docker Daemon was configured in\norder to disable the SELinux support, so it didn\u0027t really care about\nlabels; but the situation is different with Podman, and we have a full\nSELinux support at all levels on the OS, leading to the issue.\n\nFor the records, tripleo-heat-templates as well as tripleo-ansible are\nsetting the \"setype: container_file_t\" on the directories, but we don\u0027t\nuse the \"recurse: true\" in order to avoid performance issues - some\nlocations might be huge, and it would take too much time to relabel\neverything via ansible.\n\nThis patch aims to converge all the mounts to the same options, and\nensure no SELinux denial can prevent the actual container startup and\nfunction.\n\nChange-Id: Ic3e427156fc82c524c763d1896937fcc3c49fabb\nCloses-Bug: #1943459\n(cherry picked from commit e8c4e9304fa803288b38bb81a96c6183e88fdd93)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/848f2acd5b176dc4d0d82bc2518a9e0e330f9c56"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/848f2acd5b176dc4d0d82bc2518a9e0e330f9c56"}]},"branch":"refs/heads/stable/wallaby","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
