)]}'
{"id":"openstack%2Fpuppet-tripleo~822244","triplet_id":"openstack%2Fpuppet-tripleo~stable%2Ftrain~I48658413f69a68dcad0a2f24ea66fe027e987f26","project":"openstack/puppet-tripleo","branch":"stable/train","topic":"certmonger-track-keys-stable/train","attention_set":{},"removed_from_attention_set":{"14250":{"account":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"last_update":"2022-01-26 15:34:40.000000000","reason":"Change was submitted"},"20778":{"account":{"_account_id":20778,"name":"Damien Ciabrini","email":"dciabrin@redhat.com","username":"dciabrin"},"last_update":"2022-01-25 08:25:43.000000000","reason":"Damien Ciabrini replied on the change"}},"hashtags":[],"change_id":"I48658413f69a68dcad0a2f24ea66fe027e987f26","subject":"[train-only] certmonger: track change in service\u0027s private keys","status":"MERGED","created":"2021-12-20 12:36:27.000000000","updated":"2022-01-26 15:34:40.000000000","submitted":"2022-01-26 15:34:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"822244-certmonger-track-keys-stable/train","cherry_pick_of_change":771227,"cherry_pick_of_patch_set":1,"meta_rev_id":"500adbc1118e4fbea9156d6fca6f2e789db67e5a","_number":822244,"virtual_id_number":822244,"owner":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":20778,"name":"Damien Ciabrini","email":"dciabrin@redhat.com","username":"dciabrin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-26 15:34:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-01-19 18:34:49.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2022-01-25 18:00:37.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2022-01-25 08:25:43.000000000","_account_id":20778,"name":"Damien Ciabrini","email":"dciabrin@redhat.com","username":"dciabrin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":1,"date":"2022-01-25 18:00:37.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":20778,"name":"Damien Ciabrini","email":"dciabrin@redhat.com","username":"dciabrin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":20778,"name":"Damien Ciabrini","email":"dciabrin@redhat.com","username":"dciabrin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-20 12:36:27.000000000","updated_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"reviewer":{"_account_id":20778,"name":"Damien Ciabrini","email":"dciabrin@redhat.com","username":"dciabrin"},"state":"REVIEWER"},{"updated":"2021-12-20 15:10:49.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-12-20 16:46:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-25 18:00:37.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"d73a7048d713debad35a00a498459786e9a2349a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2021-12-20 12:36:27.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"e508cc881ff34852340a519385b644046b6a0230","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-12-20 15:10:49.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/cab8d0cfd22e4420b03f1cefe94c5f27 : FAILURE in 2h 12m 52s","accounts_in_message":[],"_revision_number":1},{"id":"94b947f0da6f488a399a15420041dad7c9d07915","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-20 16:46:02.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/dd7e5be339c546fab5c69ee45cc3910f : FAILURE in 5m 49s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b1854853bf2f4bcb9440b1cf87c52e4f : SUCCESS in 5m 17s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e172cbbd86b8467590bb5ab501e6e308 : SUCCESS in 5m 44s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/71ebed97eb17448d926405512bcb54f7 : FAILURE in 34m 43s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6386503ba0b341d3ba4b82794a250d25 : FAILURE in 21m 27s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/a06cb16e8ed44f9bbea99d8c6dfd9e72 : SUCCESS in 3h 38m 16s\n- tripleo-ci-centos-7-content-provider https://zuul.opendev.org/t/openstack/build/fb88cc9e1d8f4e4c8aab0c286dc3186e : SUCCESS in 4h 04m 27s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/586055a89351481d94bf74909828afec : SUCCESS in 1h 36m 25s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/a8fd6f64034d4b3b9f3a10fea37eb39d : SUCCESS in 2h 21m 11s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/2891c22825554217b2cb2f67a0ee5a78 : SUCCESS in 2h 27m 24s\n- tripleo-ci-centos-8-scenario001-standalone https://zuul.opendev.org/t/openstack/build/b828443e65d14c54a55f71cc698ef29c : SUCCESS in 1h 39m 27s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/00b82f37c26e4a39a49a2acbfd597294 : SUCCESS in 1h 20m 43s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-train https://zuul.opendev.org/t/openstack/build/2af7a9842f12451c82de65b1702f2bc4 : SUCCESS in 2h 02m 51s","accounts_in_message":[],"_revision_number":1},{"id":"c6fdeca998fec60b063a51574bd48dafcb020d23","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2022-01-13 08:06:40.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"cf5942f015e297ae7fa00aa57bbe4b625fbf3845","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-01-13 11:10:17.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/133471bfef1f4d92a3fe51e5c6301e3c : SUCCESS in 2h 54m 56s","accounts_in_message":[],"_revision_number":2},{"id":"771330f7deda3fbfca9a0bd707a29e573cafeb5d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-13 11:45:32.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/986746a6b2cf486c9939fe8980992322 : SUCCESS in 5m 21s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a075e483d5c14b5aa0bdac6cca63a497 : SUCCESS in 4m 41s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/82ae0e05925a489e85493ca647284613 : SUCCESS in 4m 48s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/4c763a54194d4851baccfe46e65073b4 : FAILURE in 45m 36s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ee23e1d6e1dd4d0097fdf0804b35e3f5 : FAILURE in 17m 27s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/48f55b74775c4bb0845733d81ce46de1 : SUCCESS in 3h 36m 14s\n- tripleo-ci-centos-7-content-provider https://zuul.opendev.org/t/openstack/build/1e68da3d688d4abba10dcf3b3dcf21a4 : SUCCESS in 2h 17m 55s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/4fe420c028154b88a520d95729fde8f9 : SUCCESS in 1h 38m 44s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/e5434f4e1b804321b20741f75eb261cd : SUCCESS in 2h 16m 07s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/fa4f685f50004e4eb9706c66d0353cf7 : FAILURE in 30m 41s\n- tripleo-ci-centos-8-scenario001-standalone https://zuul.opendev.org/t/openstack/build/02ec8619a46d4e079ef496b192f72c42 : SUCCESS in 1h 33m 22s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/60a703f4b5f54bd18905a6f158f40009 : SUCCESS in 1h 22m 38s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-train https://zuul.opendev.org/t/openstack/build/326d218e0fb8451e81441637d66360e4 : SUCCESS in 1h 59m 00s","accounts_in_message":[],"_revision_number":2},{"id":"4697d8be8b55cad887852352d99e14205c5cbfed","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2022-01-19 08:54:28.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"d78ca55e3d41c21bf1c206b2a99bb77a7312ce01","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-01-19 12:27:15.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/8dba741eeb41416295e14d82c6de68bc : SUCCESS in 3h 31m 46s","accounts_in_message":[],"_revision_number":3},{"id":"6da650dba271add7377b3e18f25b65e697d0945e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-19 13:24:10.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b30f106800864235a1d509fc3e1a3264 : SUCCESS in 5m 16s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fa7cac26a00249f5be408d847c330f81 : SUCCESS in 4m 24s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0a59014ff7f8400585456e3308045a7e : SUCCESS in 4m 29s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/330804827c27495c998d5614d332071e : FAILURE in 45m 41s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ea6c4dbde5e14270b2f7919ba2001ddd : FAILURE in 21m 59s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/5b48043b9340467b854d0cc789540fab : SUCCESS in 3h 32m 05s\n- tripleo-ci-centos-7-content-provider https://zuul.opendev.org/t/openstack/build/241cb7349c3649a1b7adb53c85f1fea4 : SUCCESS in 4h 23m 58s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/142eb97fb64b40eea2673b517bde3cda : SUCCESS in 2h 03m 32s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/60e7b2522b2a43388b168ecd8c027cf8 : SUCCESS in 2h 08m 43s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/53eb808c76054a1482c551c9edc7ac6d : SUCCESS in 2h 33m 35s\n- tripleo-ci-centos-8-scenario001-standalone https://zuul.opendev.org/t/openstack/build/200526ef068e4c4992e80ebf26aecd51 : SUCCESS in 1h 35m 09s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/121601e02e2c4cbcb244fda109500529 : SUCCESS in 1h 24m 26s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-train https://zuul.opendev.org/t/openstack/build/2a59e9782143444d89e78537cb30e50d : SUCCESS in 1h 59m 07s","accounts_in_message":[],"_revision_number":3},{"id":"f215863f1033e60b3b6e5d0658f08b14820de847","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2022-01-19 15:41:07.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"b0271d91ef7599aa9d0293045e67b0b005601afa","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-01-19 18:34:49.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/60a8184feb3949759e48a7f834b812ba : SUCCESS in 2h 50m 27s\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset035 https://review.rdoproject.org/zuul/build/b051ac27d75e465f9a4918c87588d117 : SUCCESS in 2h 53m 02s","accounts_in_message":[],"_revision_number":4},{"id":"f5bf87b6bdf5be4d82a0a55ff1e8a4286349b425","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-19 20:04:26.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2761eaefbf84451eb2d212fe40983d04 : SUCCESS in 5m 26s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1a863876facb47f7a16847270325127f : SUCCESS in 5m 27s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b4c50ddb95f244bfb89cda64f2257eae : SUCCESS in 4m 19s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/6dfb2865974d4198ba956c3fc30b2f70 : SUCCESS in 43m 38s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9a8955692a54400cbf998fd7caa6d074 : SUCCESS in 16m 38s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/203d6b8df9bb41f0a0d7568604496adb : SUCCESS in 3h 16m 34s\n- tripleo-ci-centos-7-content-provider https://zuul.opendev.org/t/openstack/build/15326dd0d8744246977a655d3a48c485 : SUCCESS in 4h 18m 30s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/851074bddba14914ba916ee7bc896eb4 : SUCCESS in 1h 31m 31s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/7396e9e701e94b10932e1a2231cf7438 : SUCCESS in 2h 06m 06s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/9626e87fd14f41ba8df4515817c8779e : SUCCESS in 2h 25m 30s\n- tripleo-ci-centos-8-scenario001-standalone https://zuul.opendev.org/t/openstack/build/c216a652eb754a14b7ff2fe4ddea9181 : SUCCESS in 1h 29m 07s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/5db2f066f47c40e5becc54bc951fbc42 : SUCCESS in 1h 15m 46s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-train https://zuul.opendev.org/t/openstack/build/ff97da7bf55e49229c4ca5901b2e13d7 : SUCCESS in 1h 58m 03s","accounts_in_message":[],"_revision_number":4},{"id":"03e48e9ca156200d3389b429a87e7b9475f83c53","author":{"_account_id":20778,"name":"Damien Ciabrini","email":"dciabrin@redhat.com","username":"dciabrin"},"date":"2022-01-25 08:25:43.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"c5674f2c394f073ba323d543795300c4aab36673","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2022-01-25 18:00:37.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"232cad4e401bcf0f9f0ed5c35df68b2a4031b969","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-26 03:52:25.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"7183bb7322abe8b46ef8c7b93bdd84586b8a3c95","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-26 15:34:40.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/785ac2124678479f9a32cc37682b9dc5 : SUCCESS in 5m 38s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/afe5921eb62f48dba27d6ea1434279f9 : SUCCESS in 5m 00s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0e5c8172b45046918217cc6ea3dd50dc : SUCCESS in 4m 22s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/42497240f9ee4054b62a08e1b6732b47 : SUCCESS in 38m 19s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/1125386a02e04d5c9362127b1bf68f7c : SUCCESS in 3h 53m 08s\n- tripleo-ci-centos-7-content-provider https://zuul.opendev.org/t/openstack/build/042afd65bd4a4a1a886ed4b45f823a1c : SUCCESS in 4h 26m 59s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/76b0144244d5414cabcbde84e28a28f9 : SUCCESS in 1h 37m 58s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/57c11ea9befa4896be80a5dc8a7cb501 : SUCCESS in 2h 11m 50s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/64ccf05f52764f22b3d71578906d1d41 : SUCCESS in 2h 35m 26s\n- tripleo-ci-centos-8-scenario001-standalone https://zuul.opendev.org/t/openstack/build/dcc3717b67d940a9bbec5de2fd8d9e26 : SUCCESS in 1h 35m 43s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/310bbe37b7c1426b9c437bddddb09af3 : SUCCESS in 1h 24m 51s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-train https://zuul.opendev.org/t/openstack/build/1416fd532e7f42599a0a96605c39ec00 : SUCCESS in 1h 57m 50s","accounts_in_message":[],"_revision_number":4},{"id":"500adbc1118e4fbea9156d6fca6f2e789db67e5a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-26 15:34:40.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"93a93c166254070f54876aaeec81479a332651ee","revisions":{"edca1a8b52e8a2d0cd460db002cb76084e91df8f":{"kind":"REWORK","_number":1,"created":"2021-12-20 12:36:27.000000000","uploader":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"ref":"refs/changes/44/822244/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tripleo","ref":"refs/changes/44/822244/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/44/822244/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/44/822244/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/44/822244/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tripleo refs/changes/44/822244/1"}}},"commit":{"parents":[{"commit":"23a518abfdd0ac56b356ca1e2e87d4f5d7930806","subject":"Merge \"Run TripleO jobs on CentOS8 instead of CentOS7\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/23a518abfdd0ac56b356ca1e2e87d4f5d7930806"}]}],"author":{"name":"Damien Ciabrini","email":"dciabrin@redhat.com","date":"2021-01-14 12:44:05.000000000","tz":60},"committer":{"name":"Grzegorz Grasza","email":"xek@redhat.com","date":"2021-12-20 12:36:27.000000000","tz":0},"subject":"[train-only] certmonger: track change in service\u0027s private keys","message":"[train-only] certmonger: track change in service\u0027s private keys\n\nCertmonger_certificate resources are marked as changed\nby puppet one some of its properties changed, like e.g.\nthe filename of the service\u0027s private key.\n\nHowever puppet-certmonger has no property to track that\nthe content of the private key has changed, so puppet\ncannot trigger a certificate renewal when a user\nexplicitely regenerated this service\u0027s private key.\n\nAdd a file resource for tracking any content change in\nthe service\u0027s private key file, and make sure that puppet\nnotify puppet-certmonger to trigger a certificate renewal.\n\nNote: some services that already use a file resource for\ntheir private key are updated, because they configure a\nwrong/backward dependency that causes a dependency loop\nwhen we add a notify dependency to the file resource.\n\nRelated-Bug: #1906505\n\nChange-Id: I48658413f69a68dcad0a2f24ea66fe027e987f26\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/edca1a8b52e8a2d0cd460db002cb76084e91df8f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/edca1a8b52e8a2d0cd460db002cb76084e91df8f"}]},"branch":"refs/heads/stable/train"},"f9a6143d456ff627f644432d5375326027366d98":{"kind":"REWORK","_number":2,"created":"2022-01-13 08:06:40.000000000","uploader":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"ref":"refs/changes/44/822244/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tripleo","ref":"refs/changes/44/822244/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/44/822244/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/44/822244/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/44/822244/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tripleo refs/changes/44/822244/2"}}},"commit":{"parents":[{"commit":"23a518abfdd0ac56b356ca1e2e87d4f5d7930806","subject":"Merge \"Run TripleO jobs on CentOS8 instead of CentOS7\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/23a518abfdd0ac56b356ca1e2e87d4f5d7930806"}]}],"author":{"name":"Damien Ciabrini","email":"dciabrin@redhat.com","date":"2021-01-14 12:44:05.000000000","tz":60},"committer":{"name":"Grzegorz Grasza","email":"xek@redhat.com","date":"2022-01-13 08:06:32.000000000","tz":60},"subject":"[train-only] certmonger: track change in service\u0027s private keys","message":"[train-only] certmonger: track change in service\u0027s private keys\n\nCertmonger_certificate resources are marked as changed\nby puppet one some of its properties changed, like e.g.\nthe filename of the service\u0027s private key.\n\nHowever puppet-certmonger has no property to track that\nthe content of the private key has changed, so puppet\ncannot trigger a certificate renewal when a user\nexplicitely regenerated this service\u0027s private key.\n\nAdd a file resource for tracking any content change in\nthe service\u0027s private key file, and make sure that puppet\nnotify puppet-certmonger to trigger a certificate renewal.\n\nNote: some services that already use a file resource for\ntheir private key are updated, because they configure a\nwrong/backward dependency that causes a dependency loop\nwhen we add a notify dependency to the file resource.\n\nRelated-Bug: #1906505\n\nChange-Id: I48658413f69a68dcad0a2f24ea66fe027e987f26\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/f9a6143d456ff627f644432d5375326027366d98"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/f9a6143d456ff627f644432d5375326027366d98"}]},"branch":"refs/heads/stable/train"},"4b4d9577910906c9ae3660508b1e6926a1232e4e":{"kind":"REWORK","_number":3,"created":"2022-01-19 08:54:28.000000000","uploader":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"ref":"refs/changes/44/822244/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tripleo","ref":"refs/changes/44/822244/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/44/822244/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/44/822244/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/44/822244/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tripleo refs/changes/44/822244/3"}}},"commit":{"parents":[{"commit":"23a518abfdd0ac56b356ca1e2e87d4f5d7930806","subject":"Merge \"Run TripleO jobs on CentOS8 instead of CentOS7\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/23a518abfdd0ac56b356ca1e2e87d4f5d7930806"}]}],"author":{"name":"Damien Ciabrini","email":"dciabrin@redhat.com","date":"2021-01-14 12:44:05.000000000","tz":60},"committer":{"name":"Grzegorz Grasza","email":"xek@redhat.com","date":"2022-01-19 08:54:18.000000000","tz":60},"subject":"[train-only] certmonger: track change in service\u0027s private keys","message":"[train-only] certmonger: track change in service\u0027s private keys\n\nCertmonger_certificate resources are marked as changed\nby puppet one some of its properties changed, like e.g.\nthe filename of the service\u0027s private key.\n\nHowever puppet-certmonger has no property to track that\nthe content of the private key has changed, so puppet\ncannot trigger a certificate renewal when a user\nexplicitely regenerated this service\u0027s private key.\n\nAdd a file resource for tracking any content change in\nthe service\u0027s private key file, and make sure that puppet\nnotify puppet-certmonger to trigger a certificate renewal.\n\nNote: some services that already use a file resource for\ntheir private key are updated, because they configure a\nwrong/backward dependency that causes a dependency loop\nwhen we add a notify dependency to the file resource.\n\nRelated-Bug: #1906505\n\nChange-Id: I48658413f69a68dcad0a2f24ea66fe027e987f26\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/4b4d9577910906c9ae3660508b1e6926a1232e4e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/4b4d9577910906c9ae3660508b1e6926a1232e4e"}]},"branch":"refs/heads/stable/train"},"93a93c166254070f54876aaeec81479a332651ee":{"kind":"REWORK","_number":4,"created":"2022-01-19 15:41:07.000000000","uploader":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"ref":"refs/changes/44/822244/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tripleo","ref":"refs/changes/44/822244/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/44/822244/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/44/822244/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/44/822244/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tripleo refs/changes/44/822244/4"}}},"commit":{"parents":[{"commit":"23a518abfdd0ac56b356ca1e2e87d4f5d7930806","subject":"Merge \"Run TripleO jobs on CentOS8 instead of CentOS7\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/23a518abfdd0ac56b356ca1e2e87d4f5d7930806"}]}],"author":{"name":"Damien Ciabrini","email":"dciabrin@redhat.com","date":"2021-01-14 12:44:05.000000000","tz":60},"committer":{"name":"Grzegorz Grasza","email":"xek@redhat.com","date":"2022-01-19 15:41:00.000000000","tz":60},"subject":"[train-only] certmonger: track change in service\u0027s private keys","message":"[train-only] certmonger: track change in service\u0027s private keys\n\nCertmonger_certificate resources are marked as changed\nby puppet one some of its properties changed, like e.g.\nthe filename of the service\u0027s private key.\n\nHowever puppet-certmonger has no property to track that\nthe content of the private key has changed, so puppet\ncannot trigger a certificate renewal when a user\nexplicitely regenerated this service\u0027s private key.\n\nAdd a file resource for tracking any content change in\nthe service\u0027s private key file, and make sure that puppet\nnotify puppet-certmonger to trigger a certificate renewal.\n\nNote: some services that already use a file resource for\ntheir private key are updated, because they configure a\nwrong/backward dependency that causes a dependency loop\nwhen we add a notify dependency to the file resource.\n\nRelated-Bug: #1906505\n\nChange-Id: I48658413f69a68dcad0a2f24ea66fe027e987f26\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/93a93c166254070f54876aaeec81479a332651ee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/93a93c166254070f54876aaeec81479a332651ee"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
