)]}'
{"id":"openstack%2Fpuppet-tripleo~847237","triplet_id":"openstack%2Fpuppet-tripleo~stable%2Fwallaby~Ia90921a840a1dd0dad7d6617d123c59e79f803bf","project":"openstack/puppet-tripleo","branch":"stable/wallaby","topic":"octavia-driver-agent-puppet","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-07-27 18:56:44.000000000","reason":"Change was submitted"},"6926":{"account":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"last_update":"2022-07-27 18:56:44.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia90921a840a1dd0dad7d6617d123c59e79f803bf","subject":"Octavia: Add optional inclusion of driver_agent class","status":"MERGED","created":"2022-06-23 04:28:55.000000000","updated":"2022-07-27 18:56:44.000000000","submitted":"2022-07-27 18:56:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"847237-octavia-driver-agent-puppet","meta_rev_id":"8f0e638f9e9e453cfd6d3e0e39fd9c37373c9089","_number":847237,"virtual_id_number":847237,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},{"value":0,"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-27 18:56:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2022-06-23 06:14:34.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"recommended":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},"all":[{"value":1,"date":"2022-07-18 08:59:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},{"value":2,"date":"2022-07-19 09:26:53.000000000","_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},{"value":1,"date":"2022-07-19 14:23:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":1,"date":"2022-07-19 07:41:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},{"value":2,"date":"2022-07-19 14:26:21.000000000","_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"all":[{"value":0,"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},{"value":0,"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},{"value":1,"date":"2022-07-19 14:26:21.000000000","_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"}],"reviewers":{"REVIEWER":[{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-23 06:14:34.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-23 07:31:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-18 08:59:04.000000000","updated_by":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},"reviewer":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},"state":"REVIEWER"},{"updated":"2022-07-19 07:41:35.000000000","updated_by":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"reviewer":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"state":"REVIEWER"},{"updated":"2022-07-19 09:26:53.000000000","updated_by":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"reviewer":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"state":"REVIEWER"},{"updated":"2022-07-19 14:21:08.000000000","updated_by":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"reviewer":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"state":"REVIEWER"},{"updated":"2022-07-19 14:23:41.000000000","updated_by":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"reviewer":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"state":"REVIEWER"},{"updated":"2022-07-19 14:24:08.000000000","updated_by":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"reviewer":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"state":"CC"},{"updated":"2022-07-19 14:26:21.000000000","updated_by":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"reviewer":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"state":"REVIEWER"}],"messages":[{"id":"cafe2ff2057952efdd09811de489aa54ddfe9df9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-06-23 04:28:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3f2521df5e76f1749b6101c525d2f56218309506","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-06-23 06:14:34.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/b8d926277aca4969898708575a29fafe : FAILURE in 1h 02m 37s\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/4db9e648e1414073a638e48713757c07 : FAILURE in 1h 44m 25s\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset035 https://review.rdoproject.org/zuul/build/feb3afb2223e42aeb8d9a147260ae876 : FAILURE in 1h 01m 19s\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset035 https://review.rdoproject.org/zuul/build/72518fb0d21c4e32bb387fca0936bff4 : FAILURE in 1h 41m 46s","accounts_in_message":[],"_revision_number":1},{"id":"7dfbb8510455ed5f06a9e21b0c97534a3908b6ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-23 07:31:47.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/44d8e3dc53d94721b81e8751c40b6093 : SUCCESS in 3m 10s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/52560703c1e940a8bb7cdd2e42e7b2bb : SUCCESS in 3m 03s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/64e287ce4eef499e9d30c9a962bf0a19 : SUCCESS in 22m 26s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8c2b2d88a64b496c8cc6a380c1f4cf39 : SUCCESS in 17m 57s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ac21ea04575c4097b8a8c16a3116d137 : SUCCESS in 16m 58s (non-voting)\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/dd5c7ff07b8f422b98b659db8fb09498 : SUCCESS in 2h 16m 19s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/281a15d78cf547bf874e86fc7d5f5a40 : SUCCESS in 3h 00m 06s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/bc9cca3cd7804ac99a20d1f649fa1f9a : SUCCESS in 54m 52s\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/1e1bd8a6fc63457e8f32050ad2a0e30e : SUCCESS in 1h 18m 59s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/d8faafe79c3a48978b74a21b9172cc61 : SUCCESS in 1h 10m 04s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/5c5b6f8ce629411daa735d97562073b2 : SUCCESS in 1h 13m 41s\n- tripleo-ci-centos-9-scenario010-standalone https://zuul.opendev.org/t/openstack/build/a3ea9f5da5d74c55964baad861e7e8c7 : FAILURE in 39m 19s\n- tripleo-ci-centos-9-scenario010-ovn-provider-standalone https://zuul.opendev.org/t/openstack/build/db91e99038b1497a8a6c3dd44a2deda8 : FAILURE in 38m 19s\n- tripleo-ci-centos-9-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/536e623f2cb94e618fc37758e2134be0 : SUCCESS in 1h 11m 15s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/116b125eb60546bdaa4987846ca35f3a : SUCCESS in 2h 59m 09s\n- tripleo-ci-centos-8-undercloud-ffu-wallaby https://zuul.opendev.org/t/openstack/build/0dbeee3113784d93bf280f8e3e796755 : SUCCESS in 1h 36m 03s","accounts_in_message":[],"_revision_number":1},{"id":"7a695cb0b2df293b8fa15f659d6f9fc0b3782c14","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-01 00:01:38.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"298848615ee2de456f1e07b01a0d0137e3f827d8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-01 03:57:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8d12a9859298453ca863f9d71822e739 : SUCCESS in 3m 46s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e375e377516a44a2966cd828ee53c940 : SUCCESS in 3m 17s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/6de55ab02fc2484abac34559ff214449 : SUCCESS in 24m 53s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/99cfd0053a014e5c9fd895733b38f0a0 : SUCCESS in 21m 49s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9eb8711c22ce4c619a5c0438ea4dc7d6 : SUCCESS in 20m 26s (non-voting)\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/62d15fecccbc468d9f0ab223d4a8b078 : SUCCESS in 3h 04m 26s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/6cf26674b39543e6ba69cb7404c9c776 : SUCCESS in 3h 47m 26s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/297cc387b4334e488d2c71811974e27a : SUCCESS in 1h 14m 46s\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/f50bdad343fd4c91b8b55aa797049017 : SUCCESS in 1h 47m 39s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/37ad63e36c624dc3814b638d555ba125 : SUCCESS in 1h 38m 17s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/e7a9816e306246ecbc838fa2fc12c004 : SUCCESS in 1h 30m 17s\n- tripleo-ci-centos-9-scenario010-standalone https://zuul.opendev.org/t/openstack/build/4ce1a3d854004915b4ae469288de0101 : SUCCESS in 1h 39m 14s\n- tripleo-ci-centos-9-scenario010-ovn-provider-standalone https://zuul.opendev.org/t/openstack/build/1b6f8e6e99954b83b15d867a87a3bda6 : SUCCESS in 1h 28m 55s\n- tripleo-ci-centos-9-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/c88e3f3fd73048368c93e3a39e5a2387 : SUCCESS in 1h 47m 15s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/1c09cab28c7b4d299fc61fc858c90a73 : SUCCESS in 3h 48m 43s\n- tripleo-ci-centos-8-undercloud-ffu-wallaby https://zuul.opendev.org/t/openstack/build/ef9e63c90f67408eaccf7f5f476030e9 : SUCCESS in 2h 06m 04s","accounts_in_message":[],"_revision_number":1},{"id":"37f583be6297882fb31c4c267daa784061d3a01c","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-07-01 05:12:01.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/561eaa4bd4254f699de102b7cf113e31 : FAILURE in 53m 48s\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/9b4e69ae4136450cbdb4260218791277 : FAILURE in 1h 29m 56s\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset035 https://review.rdoproject.org/zuul/build/300380828da14677bb1bf752a9edd262 : FAILURE in 54m 34s\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset035 https://review.rdoproject.org/zuul/build/dcc9a0a3d5d0456c81760cece3347902 : FAILURE in 1h 53m 04s","accounts_in_message":[],"_revision_number":1},{"id":"66911bdc6af6afd19a97af08a963d11989d66f61","author":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},"date":"2022-07-18 08:59:04.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"21f57461f28916aa25ed28d3efc36234a0440496","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2022-07-19 07:41:35.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"5506446e95054ebd4b4b462e2cfb7df49c922d6f","author":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"date":"2022-07-19 09:26:53.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"29d28d6b58796fc71031502774614b02b5da6ec6","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-07-19 14:23:41.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"858000dca1adbafd49bc996fc8ff275f31b4d807","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2022-07-19 14:26:21.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"68056a094b82e4d7fe1c6cde18ac4e43ad02d0de","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-19 14:28:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"99582d538eacc617afde8ca46c9903f6394300ea","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-19 21:32:38.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/14b78e7a99d741cd92f3008c30662063\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4c59ff26ff0e4c4480622f0e6a34aa37 : SUCCESS in 3m 50s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8c96b112fa9f4caebfa1764f9e108486 : SUCCESS in 3m 24s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/ca28af86104b41d08a292706646c21de : SUCCESS in 30m 49s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/60010d012e7a423f9f320a1483f80b6f : SUCCESS in 26m 41s\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/0c4afeb134414734a0723611d73ba245 : SUCCESS in 3h 27m 23s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/c0486dc666874d1e974769894d6dd1fa : SUCCESS in 3h 56m 56s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/2d36d1e4216e4d22a3b11413051a7bee : FAILURE in 50m 10s\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/292707c4ef9546e5ac51084b81297627 : FAILURE in 2h 07m 59s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/353538756fb84234bfa6f2e91d2ec541 : SUCCESS in 1h 38m 28s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/11b9f5ebadad43ffb8fe8053b059cb08 : FAILURE in 2h 02m 23s\n- tripleo-ci-centos-9-scenario010-standalone https://zuul.opendev.org/t/openstack/build/435f65c3ea7449bbacefbda5f202150b : FAILURE in 1h 16m 49s\n- tripleo-ci-centos-9-scenario010-ovn-provider-standalone https://zuul.opendev.org/t/openstack/build/b1f35fb2da5c44aaafc81952cc1221dc : FAILURE in 1h 24m 45s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/374613262f814aaa8f8c42ef7b57255e : SUCCESS in 3h 57m 08s\n- tripleo-ci-centos-9-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/79e2483be65346d280d24e77ba9d5143 : SUCCESS in 1h 43m 58s\n- tripleo-ci-centos-8-undercloud-ffu-wallaby https://zuul.opendev.org/t/openstack/build/e7d7283696184366a9d30874d8762cc2 : SUCCESS in 2h 15m 10s","accounts_in_message":[],"_revision_number":1},{"id":"24aa15434478af7d52dc0f53922615b54ce52eb5","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2022-07-20 06:17:40.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8d76c3c99adc1ca8f517ddafe588bb0dcb46969a","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-07-20 08:45:53.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/60c07b589a4d40d494d0d2a9dc80a56b : SUCCESS in 2h 16m 09s\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/b2127d61eb4146139e032713ac90ec51 : FAILURE in 1h 32m 08s\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset035 https://review.rdoproject.org/zuul/build/e8036ef8ad4b4a708cd97d66675ed6df : SUCCESS in 2h 27m 10s\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset035 https://review.rdoproject.org/zuul/build/8a1ddc5ddeba4e6b8fb377c7b062e0f1 : FAILURE in 1h 49m 08s","accounts_in_message":[],"_revision_number":1},{"id":"b819b63e6903e98a4aac75a0255846d560386d53","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-20 10:24:18.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e68127883e304365831021f4fac74120\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f5ab13de5ea34735a5f63400e6942aea : SUCCESS in 3m 52s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/75f16c0febf146f59b2d2c865dce754d : SUCCESS in 3m 50s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/552c9e1bb8dc4008803821755b814825 : SUCCESS in 28m 15s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/761bd0c6b3cd4d2ea1228bff5564ce54 : SUCCESS in 24m 54s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0ddeaf4b980145bdaf33b2ef56679a12 : SUCCESS in 23m 51s (non-voting)\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/1be17139fdb34f3f80b86933bfe472c5 : SUCCESS in 3h 30m 19s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/93a735cc1ee848449760070e49f2b213 : SUCCESS in 3h 59m 27s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/e1bfc86a559e49e094a57ea68a063a86 : FAILURE in 50m 18s\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/84835cff89a64e10a9fd5307c98e1aed : FAILURE in 2h 01m 48s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/9d968e78a2cb4569be61a8bc1c667e4e : SUCCESS in 1h 32m 22s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/2b816cefe0ae47fc8adcba00deaffed0 : FAILURE in 1h 56m 42s\n- tripleo-ci-centos-9-scenario010-standalone https://zuul.opendev.org/t/openstack/build/5685ddaa69dc45bd896947178f580e37 : FAILURE in 1h 08m 48s\n- tripleo-ci-centos-9-scenario010-ovn-provider-standalone https://zuul.opendev.org/t/openstack/build/657165643ee04e72bc434250c7e8a382 : FAILURE in 1h 17m 08s\n- tripleo-ci-centos-9-standalone-fips https://zuul.opendev.org/t/openstack/build/27adfea3438a4be38550b536fc774a7b : FAILURE in 2h 17m 09s (non-voting)\n- tripleo-ci-centos-9-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/63743b33d33243d09b97bd15c4a808a3 : SUCCESS in 1h 35m 10s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/ef5cb2670e354bc4bae5188e72c462a1 : SUCCESS in 4h 00m 01s\n- tripleo-ci-centos-8-undercloud-ffu-wallaby https://zuul.opendev.org/t/openstack/build/8f77d456eb5b4ec19916982d957b7983 : FAILURE in 2h 05m 50s","accounts_in_message":[],"_revision_number":1},{"id":"0c6c76001eb5b3907ce0bfc5e3241ca8d487cede","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-21 02:01:01.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a657475caf5bdeb91ac4bfb5381afbd2dfa8439f","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-07-21 05:01:02.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/652c2d2ecf6c45b29ea45496783b9c18 : SUCCESS in 2h 32m 53s\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/413979b3bf1543d78da07c320e5094ff : FAILURE in 1h 21m 21s\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset035 https://review.rdoproject.org/zuul/build/70b4c27f6b794c94965e613a036b4bdd : SUCCESS in 2h 43m 48s\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset035 https://review.rdoproject.org/zuul/build/72b3955d7f4a4883864d513ca4b52571 : FAILURE in 1h 34m 55s","accounts_in_message":[],"_revision_number":1},{"id":"4b667187067a3d876a20911fdde8fa367197b860","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-21 05:53:29.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9b112b3e1ef7415d85793ed33ae9f7ea\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4b5c4db5b7bb4d2498d84941a92a8f6c : SUCCESS in 4m 05s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/630371b299b94b5cbde946afc7dc670d : SUCCESS in 3m 37s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/220ff40810bc440a8b3a8eb1c7072e71 : SUCCESS in 36m 03s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0f51dfe965d94b0093e166d65ed2b09a : SUCCESS in 20m 36s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4f2009ebd16d4f1682770454ab387c17 : SUCCESS in 30m 02s (non-voting)\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/dce2bc02051045b1888510b882cd49f3 : SUCCESS in 2h 41m 31s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/bc0a47a1d83d49699cd3ee556b844c53 : SUCCESS in 3h 40m 20s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/eee9b5f8a04d488ebb0f607607f441c2 : SUCCESS in 53m 21s\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/0edcde58e2ee41f0a5435cfeb37cccaf : SUCCESS in 1h 17m 31s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/46255ff5beb346a3a50045ea9b27b70f : FAILURE in 1h 43m 54s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/8a539088fcb24a7784e168360d61d07a : SUCCESS in 1h 18m 36s\n- tripleo-ci-centos-9-scenario010-standalone https://zuul.opendev.org/t/openstack/build/a9afb8dc4260498f818220d7888dcae5 : FAILURE in 1h 32m 52s\n- tripleo-ci-centos-9-scenario010-ovn-provider-standalone https://zuul.opendev.org/t/openstack/build/9feba20e06c54971a62348b4faa63c95 : FAILURE in 1h 09m 58s\n- tripleo-ci-centos-9-standalone-fips https://zuul.opendev.org/t/openstack/build/292db3efe7a6438a9c4232b9d36ff92f : SUCCESS in 1h 40m 11s (non-voting)\n- tripleo-ci-centos-9-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/9d1b4a2576f64404bf518a2bf921f1a8 : SUCCESS in 1h 22m 00s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/d78b82b31bfc45e88372c99fc9d73113 : SUCCESS in 3h 44m 00s\n- tripleo-ci-centos-8-undercloud-ffu-wallaby https://zuul.opendev.org/t/openstack/build/34196b2dd9d74094b0c51663a262646d : FAILURE in 2h 08m 27s","accounts_in_message":[],"_revision_number":1},{"id":"fade650f37b3eced2dc43f04ef805871d0640759","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-21 08:11:59.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"330f38aa4f68a3712edce4205e6929c15d84d572","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2022-07-22 06:55:43.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"730b9f7a1a83dbdf79f93fae36a19457611bf25b","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-07-22 09:26:33.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/644b5b37696e45a1ad2610f18eaac7fb : SUCCESS in 2h 30m 24s\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/46922b927422412a87e21124610dd6f5 : SUCCESS in 2h 15m 01s\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset035 https://review.rdoproject.org/zuul/build/6b61c4e2c897439fb13ecc10ab2458c9 : SUCCESS in 2h 28m 08s\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset035 https://review.rdoproject.org/zuul/build/470f09a054d64ae09a1099aa67a093eb : FAILURE in 2h 19m 27s","accounts_in_message":[],"_revision_number":1},{"id":"7d88020476d89e1c42975c32413990d8f4a9bb79","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-22 10:21:37.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e36a80acc99d4ba79d41202405cd04a4\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/83e19f3dda7840d099a72158e181a335 : SUCCESS in 3m 41s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4e829677a7c343058b6f91a5cd903c37 : SUCCESS in 3m 36s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/abc62c2be23948ccaa5fddaf50a2c2cb : SUCCESS in 28m 09s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9fc4699fe37340679eade177f0601f6f : SUCCESS in 24m 18s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/006a6bc764ec4aefbaa89adc3249454d : SUCCESS in 22m 46s (non-voting)\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/c03970ec76e6420a834d7c3bebbc3931 : SUCCESS in 3h 18m 29s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/7ef22e5b731b45a3aa225ee57145866f : SUCCESS in 3h 23m 34s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/166bb15bf63a436d9d6516b8966d4525 : SUCCESS in 1h 04m 23s\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/3abd11bac0c94fa48a3714ae7b37f692 : SUCCESS in 1h 55m 28s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/06179ca9b9c64f5c845ddcf0e5e4a5b9 : SUCCESS in 1h 25m 08s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/499cb3d664da44238af1377e2f18f69c : FAILURE in 1h 34m 16s\n- tripleo-ci-centos-9-scenario010-standalone https://zuul.opendev.org/t/openstack/build/3ed8debe3ea548d98dee92f7cf1c2a29 : SUCCESS in 1h 32m 11s\n- tripleo-ci-centos-9-scenario010-ovn-provider-standalone https://zuul.opendev.org/t/openstack/build/168406980089419daa7f68eb1bb4bdef : SUCCESS in 1h 28m 39s\n- tripleo-ci-centos-9-standalone-fips https://zuul.opendev.org/t/openstack/build/67852476328b4444b63e1d179ef4f56f : FAILURE in 1h 57m 28s (non-voting)\n- tripleo-ci-centos-9-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/96b0fa1616b5428d89b630f217956d94 : SUCCESS in 1h 40m 31s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/603c2862c30f41f4b2d599a22a98fc74 : SUCCESS in 3h 23m 38s\n- tripleo-ci-centos-8-undercloud-ffu-wallaby https://zuul.opendev.org/t/openstack/build/815d02a81e494969b450c3c98df4d700 : SUCCESS in 1h 55m 25s","accounts_in_message":[],"_revision_number":1},{"id":"4ff1fe5f80e825744512fa3cd19270ed0fbaaaa7","author":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},"date":"2022-07-27 06:46:26.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"08f0683125e59a44721e54151b6fcf6857a34c92","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-07-27 09:20:14.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/76b647c68d654d50a1188fbbaaa64fa3 : SUCCESS in 2h 23m 06s\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/1d73a254d4e147268e182bd31b9e52b5 : FAILURE in 1h 28m 33s\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset035 https://review.rdoproject.org/zuul/build/4cda66ea777e4c1f8b9d2bc33444657b : SUCCESS in 2h 33m 02s\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset035 https://review.rdoproject.org/zuul/build/616ee10fa23b4a89ba70121f159ccbb6 : FAILURE in 1h 34m 11s","accounts_in_message":[],"_revision_number":1},{"id":"b29c4c1e19584d2759ce03c9679195ec7b505e2d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-27 10:49:25.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3f253bf1febd46b4b261243e09a71303\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1193fb53677d49419487dacb81912d94 : SUCCESS in 3m 48s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/34ba4af688ca4584b8a5474234c3f372 : SUCCESS in 3m 30s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/592245e065624479bb4883ac233bf820 : SUCCESS in 27m 03s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9b9a758f481742ec8631f4d11b89df8d : SUCCESS in 23m 02s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cb1325a8c6f04708a3ecd1bea14f9e3d : SUCCESS in 21m 40s (non-voting)\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/5663911b8cdf4b7fb9dd8b298b7693a7 : SUCCESS in 3h 56m 13s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/65f2c296810f440f9c4e3fd36b466e23 : SUCCESS in 3h 54m 47s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/343760888fa74810b0b3ca6e8a8cf3c5 : SUCCESS in 1h 13m 04s\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/57d7da5e99e041cda1ba7963490c81fa : SUCCESS in 1h 54m 27s\n- tripleo-ci-centos-8-9-multinode-mixed-os https://zuul.opendev.org/t/openstack/build/6e23a5bbb77c490790fe275224e57e90 : SUCCESS in 2h 15m 27s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/1617cfdb6a0f4956be058a1fc8606642 : SUCCESS in 1h 42m 48s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/fe2d13926bf543ada8c8fb8ddf1780ef : SUCCESS in 1h 30m 22s\n- tripleo-ci-centos-9-scenario010-standalone https://zuul.opendev.org/t/openstack/build/863ea89359b3404fb9830f2d4403db43 : FAILURE in 1h 34m 54s\n- tripleo-ci-centos-9-scenario010-ovn-provider-standalone https://zuul.opendev.org/t/openstack/build/b7be3bb4a566431b852776438caafa26 : SUCCESS in 1h 29m 43s\n- tripleo-ci-centos-9-standalone-fips https://zuul.opendev.org/t/openstack/build/a4e605c22de14410b2c666aedfb28579 : SUCCESS in 1h 42m 33s (non-voting)\n- tripleo-ci-centos-9-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/949598091e634da3aca223474912aff9 : SUCCESS in 1h 49m 57s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/356f4fdd569c4fbe8ffe367df3be46d4 : SUCCESS in 3h 35m 00s\n- tripleo-ci-centos-8-undercloud-ffu-wallaby https://zuul.opendev.org/t/openstack/build/e14b104b23bb4aeab530b5c4dd59a55f : SUCCESS in 1h 55m 39s","accounts_in_message":[],"_revision_number":1},{"id":"745b93c1d19188ce5baa9124c6e95875f8254de6","author":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},"date":"2022-07-27 10:54:06.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5632a9ab371b8a482a66237f65ed298a6721f0e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-27 14:42:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4dd2f92d7d29469abf1d9213a45c3295\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9c223bf41edc4115bc81464ed4e6d208 : SUCCESS in 3m 15s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f621202b2bf24b70afd72c147b132afb : SUCCESS in 3m 14s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/3b6249756f2a49e49faa23d2181684fa : SUCCESS in 25m 28s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c241604dc8e94f2db8a6c5168e970541 : SUCCESS in 24m 21s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2fdd51e45d6641b6933a8c3d8560fa5d : SUCCESS in 19m 05s (non-voting)\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/083ccfe4cee94e81bbbf05280a827b7b : SUCCESS in 3h 43m 08s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/bafb6a66e9d0486590f926b72c4cd9b3 : SUCCESS in 3h 41m 34s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/001f6cb69e6c405e9e347d0b2a853fcf : SUCCESS in 1h 04m 53s\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/7a4ded589c094b14b1f3008f4806cef3 : SUCCESS in 1h 46m 34s\n- tripleo-ci-centos-8-9-multinode-mixed-os https://zuul.opendev.org/t/openstack/build/cdb57c25d4e04c79a68b7f54aaa09184 : SUCCESS in 2h 00m 30s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/c7f98790f7b7494cb05ecf298ade93d9 : SUCCESS in 1h 27m 40s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/75cf171dd7a14c338d4c84aa030da8c0 : SUCCESS in 1h 19m 31s\n- tripleo-ci-centos-9-scenario010-standalone https://zuul.opendev.org/t/openstack/build/8613c39f743c4dcc93693345a2232029 : SUCCESS in 1h 47m 36s\n- tripleo-ci-centos-9-scenario010-ovn-provider-standalone https://zuul.opendev.org/t/openstack/build/2f29ef0506a442918e23d05622b78b1b : SUCCESS in 1h 17m 36s\n- tripleo-ci-centos-9-standalone-fips https://zuul.opendev.org/t/openstack/build/6e89c7b74cf44ec3bf178cd87d0f52fe : SUCCESS in 1h 32m 58s (non-voting)\n- tripleo-ci-centos-9-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/ccc2a8c545db491a8e4691e0f3e362b7 : SUCCESS in 1h 32m 39s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/113f9ae049f94d41be049b5da7efab6c : SUCCESS in 3h 29m 36s\n- tripleo-ci-centos-8-undercloud-ffu-wallaby https://zuul.opendev.org/t/openstack/build/f3e9a4d2e7f245edab175bbdd9329643 : SUCCESS in 1h 49m 09s","accounts_in_message":[],"_revision_number":1},{"id":"0128bcf53bf8bcb817be6c45dd71087247942f27","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-27 14:44:14.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1118c36d4c0b43ae7dd327097acc804625c8a11c","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-07-27 14:54:13.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/8d2d8174048d4401bf1777a766d0033d : FAILURE in 2h 36m 26s\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/f9500ad5b4024c75b098ed7ed12d0d49 : FAILURE in 1h 49m 00s\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset035 https://review.rdoproject.org/zuul/build/fac84cefb40740eb994c6a4439bac62e : FAILURE in 3h 58m 29s\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset035 https://review.rdoproject.org/zuul/build/b3106e1700134b64a9d0de50a1c0960b : FAILURE in 1h 42m 26s","accounts_in_message":[],"_revision_number":1},{"id":"471a2bad88cee43601d5181b6f31eeb25d1b298e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-27 18:56:44.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fbe8e70460f84ca9b9027e999111fe3c\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ecbd116e6c204408ba37b17002d4f31e : SUCCESS in 4m 03s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8832c0431628466082e5890b430e639b : SUCCESS in 3m 32s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/97d067781cd34298a0e039378b9d07d7 : SUCCESS in 26m 21s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a2e396a755ed48dcbec0a998c01e0df6 : SUCCESS in 22m 37s\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/3bd704b4e0e34c3c9933c518c6d75b08 : SUCCESS in 3h 29m 04s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/4a5f27a500eb411fad572fe0a7505a78 : SUCCESS in 4h 06m 27s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/53a44d98556241ca8616334b06bdbe64 : SUCCESS in 57m 16s\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/72b75992d36a4ee3a0d5fdaacaa7a8b8 : SUCCESS in 1h 28m 06s\n- tripleo-ci-centos-8-9-multinode-mixed-os https://zuul.opendev.org/t/openstack/build/a572775ca2db4509b151c05e2873318c : SUCCESS in 1h 49m 32s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/ee0a8fb4382d4d0d8fcca888aac54fe7 : SUCCESS in 1h 37m 39s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/1b909c37a44b4b589a794ee90fa91692 : SUCCESS in 1h 01m 27s\n- tripleo-ci-centos-9-scenario010-standalone https://zuul.opendev.org/t/openstack/build/773b2fc41bee48a4a6e19549e30d83df : SUCCESS in 1h 22m 37s\n- tripleo-ci-centos-9-scenario010-ovn-provider-standalone https://zuul.opendev.org/t/openstack/build/c2a85c8c4d2a4e41ad8012ae767d13b5 : SUCCESS in 1h 10m 09s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/f8071be5f5fc49b9ae19325fea4a1a74 : SUCCESS in 4h 08m 12s\n- tripleo-ci-centos-9-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/ce836079e84144f6b58f9c45c6ae9944 : SUCCESS in 1h 18m 18s\n- tripleo-ci-centos-8-undercloud-ffu-wallaby https://zuul.opendev.org/t/openstack/build/2b04f1511aca45629a704ac9605d8d17 : SUCCESS in 2h 16m 01s","accounts_in_message":[],"_revision_number":1},{"id":"8f0e638f9e9e453cfd6d3e0e39fd9c37373c9089","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-27 18:56:44.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c5db68dc0dbbc784646ae8c15ae7dad695d233eb","revisions":{"c5db68dc0dbbc784646ae8c15ae7dad695d233eb":{"kind":"REWORK","_number":1,"created":"2022-06-23 04:28:55.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/37/847237/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tripleo","ref":"refs/changes/37/847237/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/37/847237/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/37/847237/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/37/847237/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tripleo refs/changes/37/847237/1"}}},"commit":{"parents":[{"commit":"bc63c9e2d82f608d1ffb11e46abfbb7e9748f6ef","subject":"Merge \"Remove manifests for Contrail deployment\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/bc63c9e2d82f608d1ffb11e46abfbb7e9748f6ef"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-06-29 03:23:22.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-06-23 04:28:12.000000000","tz":540},"subject":"Octavia: Add optional inclusion of driver_agent class","message":"Octavia: Add optional inclusion of driver_agent class\n\n... so that we can maintain all of class inclusions in puppet-tripleo.\n\nConflicts:\n\tspec/classes/tripleo_profile_base_octavia_spec.rb\n\nResolved conflicts caused by 7b3a8cd8919399e0309c6343894a594e528d2f9a.\n\nChange-Id: Ia90921a840a1dd0dad7d6617d123c59e79f803bf\n(cherry picked from commit 3d53beb6f6e2c26dabc564b0905997f1b413a2e9)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/c5db68dc0dbbc784646ae8c15ae7dad695d233eb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/c5db68dc0dbbc784646ae8c15ae7dad695d233eb"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}}]}],"submit_requirements":[]}
