)]}'
{"id":"openstack%2Fpuppet-tripleo~855155","triplet_id":"openstack%2Fpuppet-tripleo~master~I52f099b5ab198c8cd02bbe73a4fe491b86c6d442","project":"openstack/puppet-tripleo","branch":"master","topic":"bug/1988053","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-08-31 00:06:42.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"I52f099b5ab198c8cd02bbe73a4fe491b86c6d442","subject":"tls_proxy: Make sure port is an integer","status":"MERGED","created":"2022-08-30 09:31:08.000000000","updated":"2022-08-31 03:15:12.000000000","submitted":"2022-08-31 03:15:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"855155-bug/1988053","meta_rev_id":"9f12e04edfa4f5ad0b44c4bde6e8756cfcb40753","_number":855155,"virtual_id_number":855155,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-31 03:15:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2022-08-30 11:31:38.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"all":[{"value":2,"date":"2022-08-30 13:00:07.000000000","_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":2,"date":"2022-08-30 13:41:52.000000000","_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":0,"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"value":1,"date":"2022-08-31 00:06:42.000000000","_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-30 11:31:38.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-30 11:53:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-30 13:00:07.000000000","updated_by":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"reviewer":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"state":"REVIEWER"},{"updated":"2022-08-30 13:41:52.000000000","updated_by":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"reviewer":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"state":"REVIEWER"}],"messages":[{"id":"10bad475f1547ce8d5d3b52043d45c9a9046f32e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-30 09:31:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"288a971131f06285168acc5d1403cc0d0edd02dc","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-08-30 11:31:38.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/889c48480d744f60bbcd6d2cec7f3e93 : FAILURE in 1h 53m 02s\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset035 https://review.rdoproject.org/zuul/build/39375d8d4c3f4f6691080fcca0ead0ae : POST_FAILURE in 1h 55m 02s","accounts_in_message":[],"_revision_number":1},{"id":"e124d344f98308ca1054fd86e7eabaf4c89db9c8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-30 11:53:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f792bac7515b4ec397dd1c527ed0964e\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7a7d2b09ecb04cd489bcce8217120035 : SUCCESS in 4m 33s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2f514e3daebf435a8e6a32a9358ba762 : SUCCESS in 5m 22s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/54f201779e244f0999d62994486f8dc0 : SUCCESS in 5m 25s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/2c7105a4a61c462d9ee989fa539c68a3 : SUCCESS in 22m 51s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/672dbf0bb0d94dc69460688df984a65f : SUCCESS in 21m 32s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9ff01e11555f487bacd589667d3c3faa : SUCCESS in 28m 07s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7f834ab4b39f48d5b654116026d565a8 : SUCCESS in 18m 27s (non-voting)\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/e273ed5cc6f1401c81489514c911f2d7 : SUCCESS in 2h 19m 42s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/39f5fc6e06514a93a0c76ba1967ddb58 : SUCCESS in 34m 02s\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/4042939240464995bea6fac99feca9d0 : SUCCESS in 1h 22m 56s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/fa25ebc0f3524f6cb40a7b242208347a : SUCCESS in 1h 04m 47s\n- tripleo-ci-centos-9-standalone-fips https://zuul.opendev.org/t/openstack/build/2fe30af511f24f8db9f51b215c1bbdd4 : SUCCESS in 1h 27m 17s (non-voting)\n- tripleo-ci-centos-9-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/a09a9ed34c0b46149f6fba860ee032d7 : SUCCESS in 1h 52m 34s\n- tripleo-ci-centos-9-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/524c3a2ebaf94e12af96b7a460d32d8e : SUCCESS in 1h 17m 05s\n- tripleo-ci-centos-9-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/a04138f03c7c40dd9830149ff546e53e : SUCCESS in 1h 01m 44s","accounts_in_message":[],"_revision_number":1},{"id":"48d5d6257d0102466bdda437cba41fa0131d5311","author":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"date":"2022-08-30 13:00:07.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"339803c39a26c2d4aa2de87abf3c4b37b6382cc2","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2022-08-30 13:41:52.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"61bf3b6c9251f2d175ba14e2e302571b0c5d20cd","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-31 00:06:42.000000000","message":"Patch Set 1: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"07817afb77344b40c12bcad7e8b75d322c1043ee","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-31 00:08:08.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"51a76f1b457da7b1c0e0f4622299866fd892936a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-31 03:15:12.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/36db04286a8a4d71ac5088f35a07e66c\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6bac3bf711b944859cceed27022f7518 : SUCCESS in 3m 47s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/694c516d88734a2b8071c967975322b5 : SUCCESS in 3m 39s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/57ca340c16e144edab29b79c33093bdd : SUCCESS in 3m 33s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/4d446de308e9404a94d48717c6b91aaf : SUCCESS in 30m 34s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/49bd201adbfc49a48e0147a0bc69aa31 : SUCCESS in 24m 36s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ed72240421414505a775db4acda413e9 : SUCCESS in 23m 20s\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/a9d1f83b42004545bc667d42a27ee74a : SUCCESS in 3h 03m 15s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/05df0d2b7a4f4f758745566058f1bdf9 : SUCCESS in 54m 35s\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/2db2c8744a5241d89069bdc6dfe6c631 : SUCCESS in 1h 48m 40s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/f99d7b0a58f1474fb2ef105caf564c42 : SUCCESS in 1h 34m 25s\n- tripleo-ci-centos-9-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/d490686c80f24f06a539970ae434ebb5 : SUCCESS in 2h 38m 00s\n- tripleo-ci-centos-9-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/29c5d4161d874a14ba00e185e566b292 : SUCCESS in 1h 22m 42s\n- tripleo-ci-centos-9-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/80fb9833564341a2bd102ced703418e7 : SUCCESS in 1h 38m 16s","accounts_in_message":[],"_revision_number":1},{"id":"9f12e04edfa4f5ad0b44c4bde6e8756cfcb40753","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-31 03:15:12.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"136f7a16c5f2526731ec9bfab8e97f2eda80e9d2","revisions":{"136f7a16c5f2526731ec9bfab8e97f2eda80e9d2":{"kind":"REWORK","_number":1,"created":"2022-08-30 09:31:08.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/55/855155/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tripleo","ref":"refs/changes/55/855155/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/55/855155/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/55/855155/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/55/855155/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tripleo refs/changes/55/855155/1"}}},"commit":{"parents":[{"commit":"51d77f99d3756bfabefc1f3bb6c3363c7142e5bd","subject":"Merge \"Remove unused class for cell v2 setup\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/51d77f99d3756bfabefc1f3bb6c3363c7142e5bd"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-30 09:29:16.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-30 09:30:51.000000000","tz":540},"subject":"tls_proxy: Make sure port is an integer","message":"tls_proxy: Make sure port is an integer\n\nThe apache::vhost::port parameter no longer accepts a string value and\nnow an integer value is required. This change makes sure the port is\nconverted to an integer value even if a string value is passed.\n\nPartial-Bug: #1988053\nChange-Id: I52f099b5ab198c8cd02bbe73a4fe491b86c6d442\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/136f7a16c5f2526731ec9bfab8e97f2eda80e9d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/136f7a16c5f2526731ec9bfab8e97f2eda80e9d2"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"51d77f99d3756bfabefc1f3bb6c3363c7142e5bd","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}]}],"submit_requirements":[]}
