)]}'
{"id":"openstack%2Fpuppet-tripleo~866817","triplet_id":"openstack%2Fpuppet-tripleo~master~I5592967b3a7bb928265e2ca909cf88a1f92964be","project":"openstack/puppet-tripleo","branch":"master","attention_set":{},"removed_from_attention_set":{"32240":{"account":{"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"},"last_update":"2023-02-14 04:36:59.000000000","reason":"\u003cGERRIT_ACCOUNT_32240\u003e replied on the change","reason_account":{"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"}},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2023-02-14 11:54:59.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I5592967b3a7bb928265e2ca909cf88a1f92964be","subject":"collectd: Allow duplicate plugins","status":"MERGED","created":"2022-12-07 02:12:55.000000000","updated":"2023-02-14 11:54:59.000000000","submitted":"2023-02-14 11:54:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"866817","meta_rev_id":"29248be02c518ea538a1eec2f4d1718f0b753515","_number":866817,"virtual_id_number":866817,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"value":0,"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-14 11:54:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"all":[{"value":2,"date":"2023-02-07 05:06:43.000000000","_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"value":1,"date":"2023-02-13 19:33:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-02-03 13:22:53.000000000","_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"all":[{"value":1,"date":"2023-02-07 05:06:43.000000000","_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"value":0,"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"CC":[{"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-12-07 05:42:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-03 11:56:25.000000000","updated_by":{"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"},"reviewer":{"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"},"state":"CC"},{"updated":"2023-02-03 13:22:53.000000000","updated_by":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"reviewer":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"state":"REVIEWER"},{"updated":"2023-02-07 05:06:43.000000000","updated_by":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"reviewer":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"state":"REVIEWER"},{"updated":"2023-02-13 19:33:18.000000000","updated_by":{"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},"reviewer":{"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},"state":"REVIEWER"}],"messages":[{"id":"5c9635db171ce9bf93d02fe619fbddd16f047e0c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-12-07 02:12:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"67483bf9a042288457aed500d2d76b4f865ec61f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-07 05:42:27.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ed8f10f812284c839668b85e40638179\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f9a90c652f5c45f1a395905d24fb1209 : SUCCESS in 3m 07s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/65b4e84194514ea6b5e2179de9e656b0 : SUCCESS in 2m 50s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c5718b42df1d4ad381185c4b80bc3610 : SUCCESS in 3m 25s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/4ecf5d4064ac40cbbaaec3d7c5d7be1a : FAILURE in 20m 24s\n- puppet-openstack-unit-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/39d42418db2e49c8b27502a464e29cb8 : FAILURE in 15m 32s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ab42f6b09fef430c9f361ec9ff91741b : FAILURE in 15m 17s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dcb895df7cf94acfaf57b543e01f0eec : FAILURE in 18m 57s (non-voting)\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/5a469ed9e5e34b7394c7355f497fcb66 : SUCCESS in 2h 58m 36s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/7573785a824241dd84b446c1028da8bb : SUCCESS in 45m 06s\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/bcf082d899224e198b50781d475bff7a : SUCCESS in 1h 34m 34s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/531440c02c494c8aa1aa296ee0b937a0 : SUCCESS in 1h 15m 28s\n- tripleo-ci-centos-9-scenario001-standalone https://zuul.opendev.org/t/openstack/build/7a43ae47e5414761a573d563bdbf06c1 : FAILURE in 28m 41s\n- tripleo-ci-centos-9-standalone-fips https://zuul.opendev.org/t/openstack/build/c46e439ee8ad429baf335ce2a143331b : FAILURE in 1h 48m 41s (non-voting)\n- tripleo-ci-centos-9-content-provider-zed https://zuul.opendev.org/t/openstack/build/3c0fb340d1c44669b91c65bab343e43b : SUCCESS in 2h 07m 27s\n- tripleo-ci-centos-9-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/c2559ccd50cf471d8f12b10d6d08b63f : SUCCESS in 1h 23m 36s\n- tripleo-ci-centos-9-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/dfe0cb8ffd49451386398942e4efedcd : SUCCESS in 59m 47s","accounts_in_message":[],"_revision_number":1},{"id":"85d859eeac02a9f21c42772257518f37a483ac41","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-12-07 09:35:41.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"a0202253efcbc7155b839d9eb50c19105ff8a343","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-07 10:22:19.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3de7126fef9f47cb9b56282f8a909a5f\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e9c07adea70442abb91c1e7e6d3d6cfc : SUCCESS in 3m 36s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f0886ca74bee4b75baf4799397f4d7b1 : SUCCESS in 2m 48s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b3d82fac3d36426bab669ecec1df3b21 : SUCCESS in 3m 38s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/480d4534e9a04a919157828b0a53412e : SUCCESS in 17m 44s\n- puppet-openstack-unit-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7edc10b812db46238b50fda1ce1bfb1d : SUCCESS in 21m 45s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/12d1d2d98e474fa3b8f5f9e5a3c1f412 : SUCCESS in 16m 09s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/db4cfd1b3e43478a8beff00f691401f4 : SUCCESS in 17m 40s (non-voting)\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/6c32b8159d2a433abcef297861782a08 : FAILURE in 35m 16s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/ad59ef78edea462388a0b5a7e3fbe052 : SKIPPED Skipped due to failed job tripleo-ci-centos-9-content-provider\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/dd7418864f684b4bbfbfc1a6f3d3c819 : SKIPPED Skipped due to failed job tripleo-ci-centos-9-content-provider\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/ff0d52a31303431d9418d2972e711810 : SKIPPED Skipped due to failed job tripleo-ci-centos-9-content-provider\n- tripleo-ci-centos-9-scenario001-standalone https://zuul.opendev.org/t/openstack/build/82ee390157304292ada226582fc75916 : SKIPPED Skipped due to failed job tripleo-ci-centos-9-content-provider\n- tripleo-ci-centos-9-standalone-fips https://zuul.opendev.org/t/openstack/build/08bb783725c14b528e90d46dd0943981 : SKIPPED Skipped due to failed job tripleo-ci-centos-9-content-provider (non-voting)\n- tripleo-ci-centos-9-content-provider-zed https://zuul.opendev.org/t/openstack/build/ff1c12c74a1e4cdbb2a76325d6d33e0d : FAILURE in 36m 49s\n- tripleo-ci-centos-9-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/bf52ffd4f990425d8aaaa82e63fee47d : SKIPPED Skipped due to failed job tripleo-ci-centos-9-content-provider\n- tripleo-ci-centos-9-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/e0c2bafa9b5c40f1a2df39b4cb46b106 : SKIPPED Skipped due to failed job tripleo-ci-centos-9-content-provider-zed","accounts_in_message":[],"_revision_number":2},{"id":"c9cfe13e51b825951cbaf7e23198fbd9b4025bb7","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-12-08 06:36:34.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"cb2f6a7171a56752b9545f33c6e0a61f8f23a49f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-08 09:50:49.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5cb653130a884c2cafd255d0cb706154\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/60d79f2a972846bea54594102a6ed568 : SUCCESS in 3m 02s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/570ff157f0f048bfb637cce2bb5bc0a2 : SUCCESS in 2m 43s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/48c5f1cd82f046b08098c17a26456fa2 : SUCCESS in 3m 00s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/e4d27c57ffc14d6d853569e87ca32668 : SUCCESS in 21m 35s\n- puppet-openstack-unit-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/46bfb7712a764af8b432954096332b28 : SUCCESS in 12m 48s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f87f8eec77064ba48ed529624274bbff : SUCCESS in 12m 37s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d9bbcfb71ae5428bab6df33bf898bda0 : SUCCESS in 12m 51s (non-voting)\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/8a9df7fa54f440848507de1bbb11c731 : SUCCESS in 3h 09m 00s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/4fb49db341f241719107ad95c7b242ff : SUCCESS in 52m 01s\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/3497a33f7ca24b0b819cdb427a13c069 : SUCCESS in 1h 44m 44s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/da3dc20e742d44699cf74980f5129ee9 : SUCCESS in 1h 33m 11s\n- tripleo-ci-centos-9-scenario001-standalone https://zuul.opendev.org/t/openstack/build/ea23d837884b418fbfafa2b023f98531 : SUCCESS in 1h 45m 21s\n- tripleo-ci-centos-9-standalone-fips https://zuul.opendev.org/t/openstack/build/66fc9da39cb74f9bb3dc202919efc725 : FAILURE in 1h 51m 28s (non-voting)\n- tripleo-ci-centos-9-content-provider-zed https://zuul.opendev.org/t/openstack/build/c7afbb08c5c748539a3253237f2faa20 : SUCCESS in 2h 09m 04s\n- tripleo-ci-centos-9-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/a6461e8b9e494c2c9bd92ec1d2859db4 : SUCCESS in 1h 29m 00s\n- tripleo-ci-centos-9-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/b99085f4198a447e981add5414fe5dd9 : SUCCESS in 52m 20s","accounts_in_message":[],"_revision_number":2},{"id":"7452b3bcc9a22f9dcef9bad8becea05ea895fd95","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"},"date":"2023-01-12 04:53:41.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"58ba740f4741dde7c70d426ed43d626504e582df","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-12 07:36:52.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/09260605a67841e3bbdbd40e5cfb5bfa\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b7784fab83ec4ac4a5b01f2102e510f0 : SUCCESS in 2m 58s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e14a4cf7c9b547f89af613c757d8abea : SUCCESS in 3m 03s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c3e099b9f79f4db7a43b551b0d746137 : SUCCESS in 3m 31s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/f3ba06d6ab894f5fbc8313c26190ca29 : SUCCESS in 18m 03s\n- puppet-openstack-unit-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/763af24a12344655ac04545e962d82d5 : SUCCESS in 19m 52s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8ef38ae48c5446719d45d976471f07df : SUCCESS in 17m 48s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a0f9a20cdecc4f7cb47f34ea34d9e5ab : SUCCESS in 16m 25s (non-voting)\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/9672f2f17c384e2b88aad2e5165f5c85 : SUCCESS in 2h 39m 41s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/f87b831867564f59a060e70e7f2f68d9 : SUCCESS in 32m 21s\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/404b646fc488439dba9b2ceff7f9f7db : SUCCESS in 1h 16m 45s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/146415a79d614538aec791c28342ca8b : SUCCESS in 1h 02m 49s\n- tripleo-ci-centos-9-scenario001-standalone https://zuul.opendev.org/t/openstack/build/b0b2ff85de53487a9408b0ee5acd3804 : SUCCESS in 1h 16m 08s\n- tripleo-ci-centos-9-standalone-fips https://zuul.opendev.org/t/openstack/build/380b9246d069400b8dd77252025467d1 : FAILURE in 1h 36m 58s (non-voting)\n- tripleo-ci-centos-9-content-provider-zed https://zuul.opendev.org/t/openstack/build/f09ef237413e4d439ea515d9b00946b8 : SUCCESS in 2h 29m 30s\n- tripleo-ci-centos-9-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/d9711f61f0ce475fae2c38c468c36183 : SUCCESS in 1h 16m 01s\n- tripleo-ci-centos-9-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/da6191941604422aa09df4d218952f3b : SUCCESS in 1h 12m 57s","accounts_in_message":[],"_revision_number":3},{"id":"97374e3e90e16332e45e4a3c26deb4e3023ca001","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2023-02-03 13:22:53.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"9a5d3229f1d6d6bf0ec6439bcf82f3e27218aeea","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2023-02-07 05:06:43.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"9e04991b4e50017c6e4ab7f18e4f1eb1e4ecbc65","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-07 05:08:07.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"4f028bfcbd3e5e8e89d7cd2e4dfed00359b9b10d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-07 08:55:11.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/648f52e4521646359315222206b04dbd\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fd7f2dc91afb4c068d15822e4b4cefa5 : FAILURE in 2m 37s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4ddc81df3aa04305bb02c174fc18ab6e : SUCCESS in 2m 54s\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/654c24f955ba4a288649987f1c7c0aa2 : SUCCESS in 23m 50s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bd4cdccb136c43519713e726d595c39b : SUCCESS in 19m 06s\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/61cf4d16c6ef4ce892d25194be7aca80 : SUCCESS in 3h 41m 10s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/bba80ad4a2ab4b839a02792c630cbb9b : SUCCESS in 52m 54s\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/de932e2d863442c6a1ea1d0a27c2d22c : SUCCESS in 2h 20m 44s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/421754532434477b97c243161f5d8bb4 : SUCCESS in 1h 33m 43s\n- tripleo-ci-centos-9-scenario001-standalone https://zuul.opendev.org/t/openstack/build/dc1e0b34ad4243e3be9a59a731d2cb66 : SUCCESS in 1h 54m 27s\n- tripleo-ci-centos-9-content-provider-zed https://zuul.opendev.org/t/openstack/build/630d4f1aef5542aab1484a8df155da26 : SUCCESS in 2h 24m 58s\n- tripleo-ci-centos-9-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/e023a283ae53403d8fc05794a8a6c9f0 : SUCCESS in 1h 05m 23s\n- tripleo-ci-centos-9-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/d92ccca4732e4c8195513491c4d3c430 : SUCCESS in 1h 29m 10s","accounts_in_message":[],"_revision_number":3},{"id":"9c59d0db1e6e22e9d7eb1c5e4ae63081e58b8513","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-02-07 09:00:22.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"607286a324afcae073ee0e88a0582fc6a8563f17","author":{"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},"date":"2023-02-13 19:33:18.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"8157e4d81c9d4e893fdce76a4bb33a47a77d4fbd","author":{"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"},"date":"2023-02-14 04:36:59.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"b647244a6db1125593a5ab37f15b89876f263195","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-14 08:08:59.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/99190de0d9254178bad884e10ac1bf08\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ee91bd2ba0a8464d815e14e0e66e5550 : SUCCESS in 2m 51s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d458ee37addc459fb45e44b041cf0336 : SUCCESS in 3m 15s\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/06feedb17155476b86733af304df38d9 : SUCCESS in 23m 58s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/000ff9c9ce11464fa28ba4033db60d76 : SUCCESS in 20m 04s\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f27afeb07f4d4557abf2ac4b2e769ffa : SUCCESS in 21m 38s (non-voting)\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/98a020b276984e019b60f376cad6e157 : SUCCESS in 3h 22m 55s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/1421ca4a01864d028c12907e4adf8d0d : SUCCESS in 59m 41s\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/c9dce4dc388645269c13cff7da626440 : SUCCESS in 1h 43m 13s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/fa1f6114ad7a43348b1370a67c41e810 : SUCCESS in 1h 17m 44s\n- tripleo-ci-centos-9-scenario001-standalone https://zuul.opendev.org/t/openstack/build/54d70c9845454e92aa0ee328438dc854 : SUCCESS in 1h 50m 15s\n- tripleo-ci-centos-9-standalone-fips https://zuul.opendev.org/t/openstack/build/7f3d894182bb4b93a5be7fbda2ef8292 : FAILURE in 2h 00m 57s (non-voting)\n- tripleo-ci-centos-9-content-provider-zed https://zuul.opendev.org/t/openstack/build/02257cb8233e452abfcdeb50c0e2c6a3 : SUCCESS in 2h 43m 15s\n- tripleo-ci-centos-9-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/2499a2a5d2344803ae8d24731524790d : SUCCESS in 1h 39m 05s\n- tripleo-ci-centos-9-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/aac2b42b26a44983a11e839f5cf3e1bc : SUCCESS in 1h 17m 34s","accounts_in_message":[],"_revision_number":3},{"id":"4aba829632e4186d24d718c5c601ed1558277d8c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-14 08:10:52.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"c6810b886cb235af96cb7b867c81a5f290bc7901","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-14 11:54:59.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4e3fe6aff15f4557aa6c718743d88973\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9715af3f19a746bb9dafff35967ec071 : SUCCESS in 3m 54s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a701592c5791426f886b9796c77cd548 : SUCCESS in 3m 39s\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/6f44f30e862a42188f3220195e9faac1 : SUCCESS in 24m 33s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1215f9f619604b3f9979b7c21eb4bed7 : SUCCESS in 19m 01s\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/ce2213e10c5348309cde9af860ffcc57 : SUCCESS in 3h 38m 24s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/927344da18984e1ca768d8c66d7dc253 : SUCCESS in 54m 11s\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/27df27e9c0864aa1a28d66fd3ef8b9da : SUCCESS in 1h 41m 45s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/cb8d6f25eec94ba88f14297f48f3710d : SUCCESS in 1h 28m 30s\n- tripleo-ci-centos-9-scenario001-standalone https://zuul.opendev.org/t/openstack/build/4b062a3f2f034d32b8d3cc7601339cda : SUCCESS in 1h 38m 56s\n- tripleo-ci-centos-9-content-provider-zed https://zuul.opendev.org/t/openstack/build/7079493200114c9d8bd93dc632c53e77 : SUCCESS in 3h 16m 37s\n- tripleo-ci-centos-9-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/7ea4a98f2cba43939fc099a9be391724 : SUCCESS in 1h 17m 48s\n- tripleo-ci-centos-9-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/180d30bd9a4242c0b86e3bcc84e1fa1d : SUCCESS in 1h 33m 59s","accounts_in_message":[],"_revision_number":3},{"id":"29248be02c518ea538a1eec2f4d1718f0b753515","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-14 11:54:59.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"220da27ac82bad05b229cd1b38d4491625021fc9","revisions":{"05d8e22e5414fc0d04d9f0fa141089b634cab114":{"kind":"REWORK","_number":1,"created":"2022-12-07 02:12:55.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/17/866817/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tripleo","ref":"refs/changes/17/866817/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/17/866817/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/17/866817/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/17/866817/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tripleo refs/changes/17/866817/1"}}},"commit":{"parents":[{"commit":"1e80192b76e4b653727ac9726280fd8ae80d5a56","subject":"Merge \"Template option addition for --pids-limit on Galera cluster resource\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/1e80192b76e4b653727ac9726280fd8ae80d5a56"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-12-07 02:10:51.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-12-07 02:12:44.000000000","tz":540},"subject":"collectd: Allow duplicate plugins","message":"collectd: Allow duplicate plugins\n\nSometimes different services require the same collectd plugin. For\nexample, ceph-osd and ceph-mon require the common ceph.\n\nThis change makes sure that such duplicate requirements do not cause\nresource declaration error.\n\nChange-Id: I5592967b3a7bb928265e2ca909cf88a1f92964be\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/05d8e22e5414fc0d04d9f0fa141089b634cab114"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/05d8e22e5414fc0d04d9f0fa141089b634cab114"}]},"branch":"refs/heads/master"},"638e0420245d2f318d95314426b1e59d64243c9b":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-12-07 09:35:41.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/17/866817/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tripleo","ref":"refs/changes/17/866817/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/17/866817/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/17/866817/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/17/866817/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tripleo refs/changes/17/866817/2"}}},"commit":{"parents":[{"commit":"d79306a290c8930f7ed479d53805bfaba610c8b6","subject":"Merge \"Adapt to the heartbeat_key parameter migration\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/d79306a290c8930f7ed479d53805bfaba610c8b6"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-12-07 02:10:51.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-12-07 09:35:41.000000000","tz":0},"subject":"collectd: Allow duplicate plugins","message":"collectd: Allow duplicate plugins\n\nSometimes different services require the same collectd plugin. For\nexample, ceph-osd and ceph-mon require the common ceph.\n\nThis change makes sure that such duplicate requirements do not cause\nresource declaration error.\n\nChange-Id: I5592967b3a7bb928265e2ca909cf88a1f92964be\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/638e0420245d2f318d95314426b1e59d64243c9b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/638e0420245d2f318d95314426b1e59d64243c9b"}]},"branch":"refs/heads/master","description":"Rebase"},"220da27ac82bad05b229cd1b38d4491625021fc9":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2023-01-12 04:53:41.000000000","uploader":{"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"},"ref":"refs/changes/17/866817/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tripleo","ref":"refs/changes/17/866817/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/17/866817/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/17/866817/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/17/866817/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tripleo refs/changes/17/866817/3"}}},"commit":{"parents":[{"commit":"d79306a290c8930f7ed479d53805bfaba610c8b6","subject":"Merge \"Adapt to the heartbeat_key parameter migration\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/d79306a290c8930f7ed479d53805bfaba610c8b6"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-12-07 02:10:51.000000000","tz":540},"committer":{"name":"Yadnesh Kulkarni","email":"ykulkarn@redhat.com","date":"2023-01-12 04:53:41.000000000","tz":0},"subject":"collectd: Allow duplicate plugins","message":"collectd: Allow duplicate plugins\n\nSometimes different services require the same collectd plugin. For\nexample, ceph-osd and ceph-mon require the common ceph.\n\nThis change makes sure that such duplicate requirements do not cause\nresource declaration error.\n\nThis change is wallaby-backport-potential.\n\nChange-Id: I5592967b3a7bb928265e2ca909cf88a1f92964be\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/220da27ac82bad05b229cd1b38d4491625021fc9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/220da27ac82bad05b229cd1b38d4491625021fc9"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"}}]}],"submit_requirements":[]}
