)]}'
{"id":"openstack%2Fpuppet-trove~602854","triplet_id":"openstack%2Fpuppet-trove~master~Ia3d8c561c0c21ae3eb69b88272cb4e3c156013bf","project":"openstack/puppet-trove","branch":"master","topic":"bug/1792719","hashtags":[],"change_id":"Ia3d8c561c0c21ae3eb69b88272cb4e3c156013bf","subject":"Provide more useful fail message","status":"MERGED","created":"2018-09-15 15:19:55.000000000","updated":"2019-04-25 16:26:07.000000000","submitted":"2019-04-25 16:26:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"602854-1556209567025-fe66b0bc","meta_rev_id":"80a57d67605974015c247ce0f76ac08adc9bf233","_number":602854,"virtual_id_number":602854,"owner":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":0,"_account_id":21691,"name":"weizj","email":"wzj334965317@outlook.com","username":"ZhijunWei"},{"value":2,"date":"2019-04-25 16:26:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2019-01-02 10:09:55.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":1,"date":"2019-01-20 11:19:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":1,"date":"2018-09-15 16:06:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":2,"date":"2019-04-25 15:12:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2018-09-27 09:20:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":1,"date":"2018-10-01 06:48:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21691,"name":"weizj","email":"wzj334965317@outlook.com","username":"ZhijunWei"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":1,"date":"2019-04-25 15:12:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":0,"_account_id":21691,"name":"weizj","email":"wzj334965317@outlook.com","username":"ZhijunWei"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"_account_id":21691,"name":"weizj","email":"wzj334965317@outlook.com","username":"ZhijunWei"},{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-09-15 16:06:25.000000000","updated_by":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"reviewer":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"state":"REVIEWER"},{"updated":"2018-09-27 09:20:26.000000000","updated_by":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"reviewer":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"state":"REVIEWER"},{"updated":"2018-10-01 06:48:56.000000000","updated_by":{"_account_id":21691,"name":"weizj","email":"wzj334965317@outlook.com","username":"ZhijunWei"},"reviewer":{"_account_id":21691,"name":"weizj","email":"wzj334965317@outlook.com","username":"ZhijunWei"},"state":"REVIEWER"},{"updated":"2019-01-02 10:09:55.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2019-01-20 11:19:58.000000000","updated_by":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"reviewer":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"state":"REVIEWER"},{"updated":"2019-04-25 15:12:37.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-04-25 16:26:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"57bd237a4a590ef3d8df504ebff4c4225f7ae6af","author":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"date":"2018-09-15 15:19:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3ad4837df0ea8856fb9ac5bc49a074ee8fca3782","author":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"date":"2018-09-15 15:51:35.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"2c9b320d7633cd8d335e5f70e2f695749b90a5bb","author":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"date":"2018-09-15 16:06:25.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"e744d7b88e324b37e1e89ddcf14aae86a27583bf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-09-15 16:53:05.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint http://logs.openstack.org/54/602854/2/check/puppet-openstack-lint/57789b3/ : SUCCESS in 4m 05s\n- puppet-openstack-syntax-4 http://logs.openstack.org/54/602854/2/check/puppet-openstack-syntax-4/422e53b/ : SUCCESS in 3m 38s\n- puppet-openstack-syntax-5 http://logs.openstack.org/54/602854/2/check/puppet-openstack-syntax-5/84d0464/ : SUCCESS in 4m 07s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/54/602854/2/check/puppet-openstack-unit-4.8-centos-7/fd4339e/ : SUCCESS in 11m 55s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/54/602854/2/check/puppet-openstack-unit-5.5-centos-7/09f7e03/ : SUCCESS in 11m 22s\n- puppet-openstack-unit-latest-centos-7 http://logs.openstack.org/54/602854/2/check/puppet-openstack-unit-latest-centos-7/6037d85/ : SUCCESS in 11m 06s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic http://logs.openstack.org/54/602854/2/check/puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic/207fb2c/ : FAILURE in 49m 15s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 http://logs.openstack.org/54/602854/2/check/puppet-openstack-integration-5-scenario003-tempest-centos-7/0aead86/ : SUCCESS in 59m 32s\n- puppet-openstack-integration-4-scenario003-tempest-debian-stable http://logs.openstack.org/54/602854/2/check/puppet-openstack-integration-4-scenario003-tempest-debian-stable/95f18c1/ : SUCCESS in 52m 48s (non-voting)\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/54/602854/2/check/puppet-openstack-beaker-centos-7/dbb3871/ : SUCCESS in 23m 33s\n- puppet-openstack-beaker-ubuntu-bionic http://logs.openstack.org/54/602854/2/check/puppet-openstack-beaker-ubuntu-bionic/7c08695/ : SUCCESS in 12m 05s","accounts_in_message":[],"_revision_number":2},{"id":"dcb43b0ffde9f49c5f14ba312c03281f69e768d7","author":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"date":"2018-09-27 09:20:26.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"5f5e3ca942bfd5480dc225db1dd52cebf4081866","author":{"_account_id":21691,"name":"weizj","email":"wzj334965317@outlook.com","username":"ZhijunWei"},"date":"2018-10-01 06:48:56.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"d36a34c5614b15d5e7f22cbbe934a44a22e52cac","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-01-02 10:09:55.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"4ec542c7ee0d978f7878236ee484a2654db814cc","author":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"date":"2019-01-20 11:19:58.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"17b40dbdc31da3f1c8463b3c9901fd2c425fdfdf","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-04-25 15:12:37.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"ccb9f84920fe3e3e95924b0fbfbbb7f17cf241d9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-25 15:12:49.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"f1602704d5c1b5389db5ae0aaf1cb614731bd721","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-25 16:26:06.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/54/602854/2/gate/puppet-openstack-lint-ubuntu-bionic/3aff369/ : SUCCESS in 6m 35s\n- puppet-openstack-syntax-4-ubuntu-bionic http://logs.openstack.org/54/602854/2/gate/puppet-openstack-syntax-4-ubuntu-bionic/7b30d5c/ : SUCCESS in 3m 45s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/54/602854/2/gate/puppet-openstack-syntax-5-ubuntu-bionic/9a57f7c/ : SUCCESS in 3m 39s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/54/602854/2/gate/puppet-openstack-syntax-6-ubuntu-bionic/d51a28b/ : SUCCESS in 6m 20s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/54/602854/2/gate/puppet-openstack-unit-4.8-centos-7/7b5bdc4/ : SUCCESS in 17m 57s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/54/602854/2/gate/puppet-openstack-unit-5.5-centos-7/6905805/ : SUCCESS in 17m 49s\n- puppet-openstack-unit-6.0-ubuntu-bionic http://logs.openstack.org/54/602854/2/gate/puppet-openstack-unit-6.0-ubuntu-bionic/142430c/ : SUCCESS in 11m 46s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 http://logs.openstack.org/54/602854/2/gate/puppet-openstack-integration-5-scenario003-tempest-centos-7/f4fac85/ : SUCCESS in 1h 10m 41s\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/54/602854/2/gate/puppet-openstack-beaker-centos-7/ff45964/ : SUCCESS in 17m 04s\n- puppet-openstack-beaker-ubuntu-bionic http://logs.openstack.org/54/602854/2/gate/puppet-openstack-beaker-ubuntu-bionic/84b2785/ : FAILURE in 7m 01s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"55026aee5471f6e908a5c1f7c1db05aa562c9dde","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-25 16:26:07.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ca4c7ece371c9d82cf896363ed23910cfd0dbf3f","revisions":{"5c94f2945d4cb499b55d34f8e672daf18610289f":{"kind":"REWORK","_number":1,"created":"2018-09-15 15:19:55.000000000","uploader":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"ref":"refs/changes/54/602854/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-trove","ref":"refs/changes/54/602854/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/54/602854/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/54/602854/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/54/602854/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-trove refs/changes/54/602854/1"}}},"commit":{"parents":[{"commit":"e9da9c855ae3b4c7846669628e44447243496dbb","subject":"Merge \"Add the missing \u0027Table of Contents\u0027 in README\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/e9da9c855ae3b4c7846669628e44447243496dbb"}]}],"author":{"name":"zhulingjie","email":"easyzlj@gmail.com","date":"2018-09-14 00:10:24.000000000","tz":-240},"committer":{"name":"zhulingjie","email":"easyzlj@gmail.com","date":"2018-09-14 00:10:30.000000000","tz":-240},"subject":"Provide more useful fail message","message":"Provide more useful fail message\n\nChange-Id: Ia3d8c561c0c21ae3eb69b88272cb4e3c156013bf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/5c94f2945d4cb499b55d34f8e672daf18610289f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/5c94f2945d4cb499b55d34f8e672daf18610289f"}]},"branch":"refs/heads/master"},"ca4c7ece371c9d82cf896363ed23910cfd0dbf3f":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2018-09-15 15:51:35.000000000","uploader":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"ref":"refs/changes/54/602854/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-trove","ref":"refs/changes/54/602854/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/54/602854/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/54/602854/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/54/602854/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-trove refs/changes/54/602854/2"}}},"commit":{"parents":[{"commit":"e9da9c855ae3b4c7846669628e44447243496dbb","subject":"Merge \"Add the missing \u0027Table of Contents\u0027 in README\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/e9da9c855ae3b4c7846669628e44447243496dbb"}]}],"author":{"name":"zhulingjie","email":"easyzlj@gmail.com","date":"2018-09-14 00:10:24.000000000","tz":-240},"committer":{"name":"zhulingjie","email":"easyzlj@gmail.com","date":"2018-09-14 00:41:54.000000000","tz":-240},"subject":"Provide more useful fail message","message":"Provide more useful fail message\n\nChange-Id: Ia3d8c561c0c21ae3eb69b88272cb4e3c156013bf\nCloses-Bug: #1792719\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/ca4c7ece371c9d82cf896363ed23910cfd0dbf3f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/ca4c7ece371c9d82cf896363ed23910cfd0dbf3f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
