)]}'
{"id":"openstack%2Fpuppet-trove~654204","triplet_id":"openstack%2Fpuppet-trove~master~Ice7af4b556f71bd7bec4970e88a348639f0743c1","project":"openstack/puppet-trove","branch":"master","topic":"move-to-opendev","hashtags":[],"change_id":"Ice7af4b556f71bd7bec4970e88a348639f0743c1","subject":"Move to opendev","status":"MERGED","created":"2019-04-21 06:45:59.000000000","updated":"2019-04-22 18:22:52.000000000","submitted":"2019-04-22 18:22:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"654204-1555957372066-11ac1cc1","meta_rev_id":"b9f7d6da4ef3ada7bc79abceaef785146aaad1ec","_number":654204,"virtual_id_number":654204,"owner":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2019-04-22 18:22:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2019-04-22 13:24:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2019-04-22 17:06:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2019-04-22 17:06:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-22 13:24:22.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2019-04-22 17:06:46.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-04-22 18:22:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e467dc16444b55dd432e5893574dafb27b294f16","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-04-21 06:45:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a307935e9e93c2d0e6666c65fbed913d8f15b906","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-21 07:58:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/04/654204/1/check/puppet-openstack-lint-ubuntu-bionic/c03e7a8/ : SUCCESS in 4m 26s\n- puppet-openstack-syntax-4-ubuntu-bionic http://logs.openstack.org/04/654204/1/check/puppet-openstack-syntax-4-ubuntu-bionic/4223cb3/ : SUCCESS in 4m 06s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/04/654204/1/check/puppet-openstack-syntax-5-ubuntu-bionic/2508d74/ : SUCCESS in 4m 02s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/04/654204/1/check/puppet-openstack-syntax-6-ubuntu-bionic/4d24926/ : SUCCESS in 3m 54s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/04/654204/1/check/puppet-openstack-unit-4.8-centos-7/ac2cd29/ : SUCCESS in 16m 57s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/04/654204/1/check/puppet-openstack-unit-5.5-centos-7/4ebe407/ : SUCCESS in 17m 22s\n- puppet-openstack-unit-6.0-ubuntu-bionic http://logs.openstack.org/04/654204/1/check/puppet-openstack-unit-6.0-ubuntu-bionic/1cd2efc/ : SUCCESS in 7m 28s\n- puppet-openstack-unit-latest-ubuntu-bionic http://logs.openstack.org/04/654204/1/check/puppet-openstack-unit-latest-ubuntu-bionic/36f8c98/ : FAILURE in 10m 31s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic http://logs.openstack.org/04/654204/1/check/puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic/4ded5c9/ : FAILURE in 13m 31s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 http://logs.openstack.org/04/654204/1/check/puppet-openstack-integration-5-scenario003-tempest-centos-7/90c2d6d/ : SUCCESS in 57m 04s\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/04/654204/1/check/puppet-openstack-beaker-centos-7/c4d5eb3/ : SUCCESS in 15m 46s\n- puppet-openstack-beaker-ubuntu-bionic http://logs.openstack.org/04/654204/1/check/puppet-openstack-beaker-ubuntu-bionic/ccedf4b/ : FAILURE in 11m 02s (non-voting)\n- build-openstack-releasenotes http://logs.openstack.org/04/654204/1/check/build-openstack-releasenotes/ebb4139/html/ : SUCCESS in 2m 48s","accounts_in_message":[],"_revision_number":1},{"id":"e993a0da09c1c541695eab8b066e5ebff60b3859","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2019-04-22 13:24:22.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"2537536d55cb79bd117b8244ef90c39feac67cc3","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-04-22 17:06:46.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"12d3d1072a553d48642700c0b9e8d413c882add7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 17:07:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"daee0a4357761bb3ecfd670d91ff03d34c7c1865","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 18:22:51.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/04/654204/1/gate/puppet-openstack-lint-ubuntu-bionic/9629c87/ : SUCCESS in 6m 30s\n- puppet-openstack-syntax-4-ubuntu-bionic http://logs.openstack.org/04/654204/1/gate/puppet-openstack-syntax-4-ubuntu-bionic/30028e7/ : SUCCESS in 4m 01s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/04/654204/1/gate/puppet-openstack-syntax-5-ubuntu-bionic/1316e1b/ : SUCCESS in 4m 23s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/04/654204/1/gate/puppet-openstack-syntax-6-ubuntu-bionic/ac97ca0/ : SUCCESS in 5m 01s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/04/654204/1/gate/puppet-openstack-unit-4.8-centos-7/c614c99/ : SUCCESS in 12m 07s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/04/654204/1/gate/puppet-openstack-unit-5.5-centos-7/6613df5/ : SUCCESS in 11m 49s\n- puppet-openstack-unit-6.0-ubuntu-bionic http://logs.openstack.org/04/654204/1/gate/puppet-openstack-unit-6.0-ubuntu-bionic/6afc91b/ : SUCCESS in 7m 25s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 http://logs.openstack.org/04/654204/1/gate/puppet-openstack-integration-5-scenario003-tempest-centos-7/0a8df1b/ : SUCCESS in 1h 02m 26s\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/04/654204/1/gate/puppet-openstack-beaker-centos-7/fb79b8c/ : SUCCESS in 17m 19s\n- puppet-openstack-beaker-ubuntu-bionic http://logs.openstack.org/04/654204/1/gate/puppet-openstack-beaker-ubuntu-bionic/a1640ab/ : FAILURE in 10m 55s (non-voting)\n- build-openstack-releasenotes http://logs.openstack.org/04/654204/1/gate/build-openstack-releasenotes/6f0316a/html/ : SUCCESS in 2m 47s","accounts_in_message":[],"_revision_number":1},{"id":"e0d1394464b9325996ad745d23d9d08b23c2c3cd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 18:22:52.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b7ddd3b7eb31fc1318015b5245dd5cc3960c5328","revisions":{"b7ddd3b7eb31fc1318015b5245dd5cc3960c5328":{"kind":"REWORK","_number":1,"created":"2019-04-21 06:45:59.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/04/654204/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-trove","ref":"refs/changes/04/654204/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/04/654204/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/04/654204/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/04/654204/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-trove refs/changes/04/654204/1"}}},"commit":{"parents":[{"commit":"90e75ef1b2ede451b5160bedea231919debb992c","subject":"Fix Repository","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/90e75ef1b2ede451b5160bedea231919debb992c"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-04-21 06:44:49.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-04-21 06:44:49.000000000","tz":480},"subject":"Move to opendev","message":"Move to opendev\n\nChange-Id: Ice7af4b556f71bd7bec4970e88a348639f0743c1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/b7ddd3b7eb31fc1318015b5245dd5cc3960c5328"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/b7ddd3b7eb31fc1318015b5245dd5cc3960c5328"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
