)]}'
{"id":"openstack%2Fpuppet-trove~665819","triplet_id":"openstack%2Fpuppet-trove~master~Icc865b8369ec4151a3ff89871e52896b8b4f7ea9","project":"openstack/puppet-trove","branch":"master","topic":"665819","hashtags":[],"change_id":"Icc865b8369ec4151a3ff89871e52896b8b4f7ea9","subject":"assert_private() params and deps classes","status":"ABANDONED","created":"2019-06-17 23:39:01.000000000","updated":"2020-09-18 22:57:43.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"0285f3f46d2f28713cae3e15ad14955c9882dce4","_number":665819,"virtual_id_number":665819,"owner":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":-1,"date":"2019-06-18 01:04:51.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-18 01:04:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d061afb2a9f7f43e206682c48567b01ef1438305","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2019-06-17 23:39:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c0ec99eae35f43a3b48d39e2a7b72b2848abdfde","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2019-06-17 23:54:55.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f0eaff351ba448964c1796ff6a15f9a1868f13d5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 01:04:51.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/19/665819/2/check/puppet-openstack-lint-ubuntu-bionic/518c151/ : SUCCESS in 4m 14s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/19/665819/2/check/puppet-openstack-syntax-5-ubuntu-bionic/13c9834/ : SUCCESS in 5m 16s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/19/665819/2/check/puppet-openstack-syntax-6-ubuntu-bionic/266ba50/ : SUCCESS in 4m 14s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/19/665819/2/check/puppet-openstack-unit-5.5-centos-7/896db0f/ : FAILURE in 23m 18s\n- puppet-openstack-unit-6.4-ubuntu-bionic http://logs.openstack.org/19/665819/2/check/puppet-openstack-unit-6.4-ubuntu-bionic/431200f/ : FAILURE in 8m 28s\n- puppet-openstack-unit-latest-ubuntu-bionic http://logs.openstack.org/19/665819/2/check/puppet-openstack-unit-latest-ubuntu-bionic/e6588dd/ : FAILURE in 8m 17s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic http://logs.openstack.org/19/665819/2/check/puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic/ddb1fa5/ : FAILURE in 32m 52s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 http://logs.openstack.org/19/665819/2/check/puppet-openstack-integration-5-scenario003-tempest-centos-7/3f26530/ : SUCCESS in 1h 03m 11s\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/19/665819/2/check/puppet-openstack-beaker-centos-7/89cac1d/ : SUCCESS in 14m 37s\n- puppet-openstack-beaker-ubuntu-bionic http://logs.openstack.org/19/665819/2/check/puppet-openstack-beaker-ubuntu-bionic/461324f/ : FAILURE in 7m 18s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"7511deb55bf3d3197b21d59cf500d19569cf80a6","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-09-18 22:57:43.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"81c11cc374e863dab70c5e2ab7110bf51ebbf5dd","revisions":{"63220df40518f5f421e368b43eae14c3d5def909":{"kind":"REWORK","_number":1,"created":"2019-06-17 23:39:01.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/19/665819/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-trove","ref":"refs/changes/19/665819/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/19/665819/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/19/665819/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/19/665819/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-trove refs/changes/19/665819/1"}}},"commit":{"parents":[{"commit":"06e33d81bbf5c736b2019fe858bbc37d1a3e2d12","subject":"Prepare Train M2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/06e33d81bbf5c736b2019fe858bbc37d1a3e2d12"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2019-06-17 23:38:53.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2019-06-17 23:38:53.000000000","tz":120},"subject":"assert_private() params and deps classes","message":"assert_private() params and deps classes\n\nThese classes are private and should never be\nused outside of it\u0027s module.\n\nChange-Id: Icc865b8369ec4151a3ff89871e52896b8b4f7ea9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/63220df40518f5f421e368b43eae14c3d5def909"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/63220df40518f5f421e368b43eae14c3d5def909"}]},"branch":"refs/heads/master"},"81c11cc374e863dab70c5e2ab7110bf51ebbf5dd":{"kind":"REWORK","_number":2,"created":"2019-06-17 23:54:55.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/19/665819/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-trove","ref":"refs/changes/19/665819/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/19/665819/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/19/665819/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/19/665819/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-trove refs/changes/19/665819/2"}}},"commit":{"parents":[{"commit":"06e33d81bbf5c736b2019fe858bbc37d1a3e2d12","subject":"Prepare Train M2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/06e33d81bbf5c736b2019fe858bbc37d1a3e2d12"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2019-06-17 23:38:53.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2019-06-17 23:54:49.000000000","tz":120},"subject":"assert_private() params and deps classes","message":"assert_private() params and deps classes\n\nThese classes are private and should never be\nused outside of it\u0027s module.\n\nChange-Id: Icc865b8369ec4151a3ff89871e52896b8b4f7ea9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/81c11cc374e863dab70c5e2ab7110bf51ebbf5dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/81c11cc374e863dab70c5e2ab7110bf51ebbf5dd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
