)]}'
{"id":"openstack%2Fpuppet-trove~684898","triplet_id":"openstack%2Fpuppet-trove~master~Id1bb166d19c3b2d7ad988199a42b3c0e64b1db56","project":"openstack/puppet-trove","branch":"master","topic":"constraints","hashtags":[],"change_id":"Id1bb166d19c3b2d7ad988199a42b3c0e64b1db56","subject":"Update the constraints url","status":"MERGED","created":"2019-09-26 03:08:41.000000000","updated":"2019-09-27 03:15:07.000000000","submitted":"2019-09-27 03:13:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"684898-1569554027354-9cafdce8","meta_rev_id":"c984ce0fd463ce9680c51cdc2c67d68f3ece7e36","_number":684898,"virtual_id_number":684898,"owner":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-09-27 03:13:47.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2019-09-27 01:50:20.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2019-09-27 01:50:20.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-27 01:50:20.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2019-09-27 03:13:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ae956eb288742d26d38c660101cdaf7785a67667","author":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"date":"2019-09-26 03:08:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2202eeda0694a4b611c7f342373e95953b3a61e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-26 04:27:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/80ce55e8807b4ea89fe8ed38ac8747c6 : SUCCESS in 4m 59s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/69d864d6ea924f0580d8e679ae9aee5f : SUCCESS in 4m 52s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/579706c3b2a9489fba6a831cabc59e1f : SUCCESS in 4m 38s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/613fb617249a4d81b240e26eae14b754 : SUCCESS in 16m 39s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/75f1d5e942e74fb0a99b54d355011827 : SUCCESS in 9m 20s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7dec40689c6b4e0a9776b731e5118e24 : SUCCESS in 7m 26s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f9bd919563644be990fdb951f0dd862e : FAILURE in 17m 55s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/670816cc89684ce8bd86e1c597b5fec7 : SUCCESS in 1h 11m 16s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/9f73cb41268e41c9b2dbbd951968f84a : SUCCESS in 18m 04s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d4a11dc2b74f4b1f929d8611bf06aa71 : FAILURE in 7m 42s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/99b9d8afcb5e4f9bb3334bd86451f5e6 : SUCCESS in 3m 29s","accounts_in_message":[],"_revision_number":1},{"id":"87ba75e0d04d5ae3f87ce1af908f3da5475d958b","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-09-27 01:50:20.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"be25194ac9673d75031419808584ecc1802bff1f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-27 01:50:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"99f3ca541b0bb49119c669e87428b276280535b9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-27 03:13:47.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/62cd35f10f724fd6bb372c6c2ad197fc : SUCCESS in 7m 34s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/800c99ef6f2943c8a813d242d9261d4a : SUCCESS in 5m 24s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/273b088c895d499d807ab7ceaafcd93e : SUCCESS in 5m 44s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/72581dba697d41738448da4657d13c1f : SUCCESS in 12m 09s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2df0f469c67944ae89113eeefed14177 : SUCCESS in 10m 31s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/49b1ef80a82a4ef89ded57df770ddd1b : SUCCESS in 1h 08m 58s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/9d7114b1c8e04d72ba1d56bf226f2fdf : SUCCESS in 15m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/81a61ace99d64429993551c2354eae7b : SUCCESS in 4m 01s","accounts_in_message":[],"_revision_number":1},{"id":"68d5445eb5982a41e5c3bb1da1db4fca004ea877","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-27 03:13:47.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"c1d0b3de95a66d910a74a335bb13179c79db9fd9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-27 03:15:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/84a19b20109c43109ac6a4196449bd23 : SUCCESS in 1m 03s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0ac965ee3af73e014e8203f5ff4284af57a7e01f","revisions":{"0ac965ee3af73e014e8203f5ff4284af57a7e01f":{"kind":"REWORK","_number":1,"created":"2019-09-26 03:08:41.000000000","uploader":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"ref":"refs/changes/98/684898/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-trove","ref":"refs/changes/98/684898/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/98/684898/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/98/684898/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/98/684898/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-trove refs/changes/98/684898/1"}}},"commit":{"parents":[{"commit":"76a78add5f00bb1d85e97793ca29070c3972de29","subject":"Prepare Train M4","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/76a78add5f00bb1d85e97793ca29070c3972de29"}]}],"author":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-09-26 03:08:40.000000000","tz":480},"committer":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-09-26 03:09:24.000000000","tz":480},"subject":"Update the constraints url","message":"Update the constraints url\n\nFor more detail, see http://lists.openstack.org/pipermail/openstack-discuss/2019-May/006478.html\n\nChange-Id: Id1bb166d19c3b2d7ad988199a42b3c0e64b1db56\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/0ac965ee3af73e014e8203f5ff4284af57a7e01f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/0ac965ee3af73e014e8203f5ff4284af57a7e01f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
