)]}'
{"id":"openstack%2Fpuppet-trove~691787","triplet_id":"openstack%2Fpuppet-trove~master~Id1d41ef02346860f70cfd71361a0ca7aea2816cf","project":"openstack/puppet-trove","branch":"master","topic":"bug/1850236","hashtags":[],"change_id":"Id1d41ef02346860f70cfd71361a0ca7aea2816cf","subject":"Fix rabbitmq dependency","status":"MERGED","created":"2019-10-29 01:32:16.000000000","updated":"2020-04-03 08:33:36.000000000","submitted":"2019-10-29 01:57:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"691787-1572314258017-df6c426c","meta_rev_id":"73e4770e0883a54e8aea34aa390795dfe97e8ceb","_number":691787,"virtual_id_number":691787,"owner":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-29 01:57:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":2,"date":"2019-10-29 01:52:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":1,"date":"2019-10-29 01:52:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-29 01:57:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4be587753725b63bdd71d93ba46cb8fc1304f227","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-10-29 01:32:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4c308814d59365ace1471bdb4860b5bbec98e997","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-29 01:40:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/96ca742b489845e080b1b9c7ac4f03ef : SUCCESS in 2m 18s\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/42b75690587d4205bad752f31f4da172 : SUCCESS in 4m 08s","accounts_in_message":[],"_revision_number":1},{"id":"b7bb503461473e54a66782f0f3038104467ca3b0","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-10-29 01:52:21.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a03be83b7759d66e003c24e0f838cd7eefe43c24","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-29 01:52:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bdb27b0d59003e314cfb0a2e7c4d9e038e43ebc8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-29 01:57:37.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/74a849154dba43bc8705246b17be6fdb : SUCCESS in 2m 17s\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e34acdb21ec742a0bee390b329f005c6 : SUCCESS in 4m 18s","accounts_in_message":[],"_revision_number":1},{"id":"6fc5d44c2176eaa9edcc99ec48ea36b03f7c77e4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-29 01:57:38.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"c428acc8c784111879e271e1c7f0d71a370e2ebe","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-04-03 08:33:36.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit bddf0f9d3d201c30f9e4b70c64ed7e308037778d","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"298ab338012b58361f8b804da7b0644652d699fc","revisions":{"298ab338012b58361f8b804da7b0644652d699fc":{"kind":"REWORK","_number":1,"created":"2019-10-29 01:32:16.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/87/691787/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-trove","ref":"refs/changes/87/691787/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/87/691787/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/87/691787/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/87/691787/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-trove refs/changes/87/691787/1"}}},"commit":{"parents":[{"commit":"7013f5828a4b7f3e95a4dbdd4b27eb2c85e9e09e","subject":"Update master for stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/7013f5828a4b7f3e95a4dbdd4b27eb2c85e9e09e"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-10-29 01:31:05.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-10-29 01:31:05.000000000","tz":480},"subject":"Fix rabbitmq dependency","message":"Fix rabbitmq dependency\n\npuppetlabs/rabbitmq [1] is deprecated, and max version is 5.6.0 which can\nnot be fulfilled by this dependency.\n\nCorrect it to puppet-rabbitmq [2] instead which is being actively maintained\nand has the right version.\n\n[1]: https://forge.puppet.com/puppetlabs/rabbitmq\n[2]: https://forge.puppet.com/puppet/rabbitmq\n\nChange-Id: Id1d41ef02346860f70cfd71361a0ca7aea2816cf\nCloses-Bug: #1850236\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/298ab338012b58361f8b804da7b0644652d699fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/298ab338012b58361f8b804da7b0644652d699fc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
