)]}'
{"id":"openstack%2Fpuppet-trove~728208","triplet_id":"openstack%2Fpuppet-trove~stable%2Fussuri~I9f2967c1400fe6b95d5af361d59996655978e2ea","project":"openstack/puppet-trove","branch":"stable/ussuri","hashtags":[],"change_id":"I9f2967c1400fe6b95d5af361d59996655978e2ea","subject":"Switch to Ussuri","status":"MERGED","created":"2020-05-14 18:11:41.000000000","updated":"2020-05-18 07:25:56.000000000","submitted":"2020-05-18 07:25:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"728208-1589786756442-ed83af0c","meta_rev_id":"74526123d3ce2aa75c9ef41a4e9ee0e10d2864fe","_number":728208,"virtual_id_number":728208,"owner":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-18 07:25:56.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":2,"date":"2020-05-18 06:02:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":2,"date":"2020-05-15 22:42:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":1,"date":"2020-05-18 06:02:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-15 22:42:44.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2020-05-18 06:02:19.000000000","updated_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"reviewer":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"state":"REVIEWER"},{"updated":"2020-05-18 07:25:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c3260e32fb10344cc9312d8fda08640ad863a7ee","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-05-14 18:11:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"756d7d765fb686a734c9fa9b9b1fbbd1a259f010","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-14 19:30:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c3e5a1e0c7664cb0bf5f67b602277285 : SUCCESS in 8m 35s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/791029f6ced3448fbff900a3925792a2 : SUCCESS in 5m 54s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/587b0d225b234bc592c11164fa8ec952 : SUCCESS in 6m 44s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/2af70c26c9ed4b1eb90dc645325db8cf : SUCCESS in 17m 21s\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/cf8be45b2448491eaad86e8224786e10 : SUCCESS in 12m 55s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7e172cd14fc0456bb506df2fd1a07450 : SUCCESS in 11m 16s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7e5aeac484fe495da7ef4b3ce51e154a : SUCCESS in 14m 02s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4d7c8e19a1cd4bfdb4dd07093e89eca5 : SUCCESS in 44m 17s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/e640c1b1e85643aeb47040692fdc2f8d : SUCCESS in 1h 05m 56s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/e4f771ca89054c9ea4ba32d59950ed9a : SUCCESS in 18m 15s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/6f4c46a7987f45a69b7dd79faa911e7c : FAILURE in 14m 19s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7ae748704e924e85bb291bc58911881b : FAILURE in 9m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c0185db06e5bd51832f89ef284be7e8863be1fe4","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-05-15 22:42:44.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"4ed84f5fc754b98891bef5b18ff5d27bdd089c0f","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2020-05-18 06:02:19.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d1e2ffc13ce50f8c529fe80726744a61ff4a741a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-18 06:02:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ee4ce5e94a1dae9ded86065dd588b22960913167","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-18 07:25:56.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"74526123d3ce2aa75c9ef41a4e9ee0e10d2864fe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-18 07:25:56.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8e3c4fd0219d4ad49d519fb238403c52 : SUCCESS in 7m 40s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b8445245a08548f2bb3b331bafc04b22 : SUCCESS in 6m 18s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/497bc79667ee4e9bb5d5696c27aeb742 : SUCCESS in 6m 42s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/784a24f92e574739b16abe83d52d3191 : SUCCESS in 13m 53s\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/c4fa9530228448d0bc5122bb9345bfac : SUCCESS in 15m 41s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/08d730eb0b5b444eb7ce9495aa35483e : SUCCESS in 13m 19s\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/aac5b7214a674720a45cab5a5919c6cb : SUCCESS in 1h 09m 27s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/72a05a2a8a04408492b6227cedb82731 : SUCCESS in 21m 49s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"092cf73f8e644949eb26e3edb5b5802d36737bee","revisions":{"092cf73f8e644949eb26e3edb5b5802d36737bee":{"kind":"REWORK","_number":1,"created":"2020-05-14 18:11:41.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/08/728208/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-trove","ref":"refs/changes/08/728208/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/08/728208/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/08/728208/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/08/728208/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-trove refs/changes/08/728208/1"}}},"commit":{"parents":[{"commit":"9f79b2712317cc51f1f1a4628311cb2abfb2c521","subject":"Update TOX_CONSTRAINTS_FILE for stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/9f79b2712317cc51f1f1a4628311cb2abfb2c521"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2020-05-14 18:11:32.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2020-05-14 18:11:32.000000000","tz":120},"subject":"Switch to Ussuri","message":"Switch to Ussuri\n\nChange-Id: I9f2967c1400fe6b95d5af361d59996655978e2ea\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/092cf73f8e644949eb26e3edb5b5802d36737bee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/092cf73f8e644949eb26e3edb5b5802d36737bee"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
