)]}'
{"id":"openstack%2Fpuppet-trove~741400","triplet_id":"openstack%2Fpuppet-trove~master~If61e708f68f96ce6c485a681d39e17c3cf737d2a","project":"openstack/puppet-trove","branch":"master","topic":"at-interface","hashtags":[],"change_id":"If61e708f68f96ce6c485a681d39e17c3cf737d2a","subject":"Add support for the interface parameter in authtoken middleware","status":"MERGED","created":"2020-07-16 08:08:13.000000000","updated":"2020-07-16 16:19:21.000000000","submitted":"2020-07-16 16:17:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"741400-1594916268943-98bbf7d3","meta_rev_id":"d3c2beb7542cbf07f53cf74c991a094e0e55b612","_number":741400,"virtual_id_number":741400,"owner":{"_account_id":32114,"name":"Lewis Denny","email":"ldenny@redhat.com","username":"ldenny"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-07-16 11:31:59.000000000","_account_id":32114,"name":"Lewis Denny","email":"ldenny@redhat.com","username":"ldenny"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-16 16:17:48.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":0,"_account_id":32114,"name":"Lewis Denny","email":"ldenny@redhat.com","username":"ldenny"},{"value":2,"date":"2020-07-16 11:37:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2020-07-16 12:26:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":32114,"name":"Lewis Denny","email":"ldenny@redhat.com","username":"ldenny"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":1,"date":"2020-07-16 15:03:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32114,"name":"Lewis Denny","email":"ldenny@redhat.com","username":"ldenny"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-16 11:37:53.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2020-07-16 15:03:02.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2020-07-16 16:17:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7a92cf4ba29a9fe32323ac3fb68df3f62895a42c","author":{"_account_id":32114,"name":"Lewis Denny","email":"ldenny@redhat.com","username":"ldenny"},"date":"2020-07-16 08:08:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9158463646d51657f3fb7211c30f28fac35733ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-16 09:46:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/61b39acdfc5c4f2993f996675aa5b70b : SUCCESS in 4m 56s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/884572b5fa0d46f6a169ec0fe6557237 : SUCCESS in 5m 14s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6f07feb0624242d282d5b3bbe8feea1c : SUCCESS in 5m 18s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/3b7039c9f2c34dc0ba67cdca3d446691 : SUCCESS in 12m 04s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/9a4587eece9346deb994222fdcf84f49 : SUCCESS in 16m 20s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4f6d556cb86048088b13b0e9b15ddfd8 : SUCCESS in 11m 04s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/592dab54ca984b8291de443731357bc8 : FAILURE in 8m 30s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a5d82bcce41f4f688382374550cdae18 : SUCCESS in 35m 02s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/32cbf35f75c840f0a243730a14ac23d8 : SUCCESS in 1h 33m 56s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/86f123a1c3664a9daac4a092fb7ed23a : SUCCESS in 17m 10s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/d0ae4cd458f943a481656e0a54207170 : FAILURE in 11m 34s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4f5bd78246c243f5ba8021eb2e571ed6 : FAILURE in 9m 18s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ea6ef6e6bba244f1b7db85b2fb1bfa6c : SUCCESS in 3m 36s","accounts_in_message":[],"_revision_number":1},{"id":"f4b4b14a90b0287da1373db00e48aa1d5e174e6b","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-07-16 10:04:49.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b5a8465a6b13edd0814f15d2d936e639ba4421ca","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-07-16 10:35:12.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9a78316e57be1c4e2838597ed48d653574925d63","author":{"_account_id":32114,"name":"Lewis Denny","email":"ldenny@redhat.com","username":"ldenny"},"date":"2020-07-16 11:29:16.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"df24ab41896b11128f7dfa713663807e43e5a193","author":{"_account_id":32114,"name":"Lewis Denny","email":"ldenny@redhat.com","username":"ldenny"},"date":"2020-07-16 11:31:59.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a5a92898bc5e604ef48cb2c6a4000846a8b4eee2","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-07-16 11:37:53.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"1c02e74f0f7b10401d09e2109ad00f1a7d8564fc","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-07-16 12:26:40.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"bebcdba36110475bc4b7d2fcf4e30ac0a3fbdcd6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-16 12:44:29.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d0d75c95f16a4937b615d4898078396e : SUCCESS in 5m 10s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7c41982831524b36988c51fc517ef212 : SUCCESS in 3m 56s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f7560fcc7622400cb1dd9d652016e446 : SUCCESS in 4m 20s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/811019fea0a6483e94161b38fadd4644 : SUCCESS in 10m 26s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/c2d8c0cde0b74c78861a6893c04742fc : SUCCESS in 13m 19s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ae0dfb7007a0425a8d9c86f89a3a7321 : SUCCESS in 9m 17s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ca4b95d41c344bba83966b7b8f64cf55 : FAILURE in 7m 17s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bead8ec902d045a78b3c9af077861386 : SUCCESS in 47m 39s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/5f4c516d5cf24f17aa191a4c48dc8d21 : SUCCESS in 1h 11m 43s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/461c180ca92449c1991f9786b708881e : FAILURE in 8m 53s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/20de6f69cab344f5a8ae44b4a92e439e : FAILURE in 8m 29s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d9e3607063164cc2a6cc217c87e6ea70 : FAILURE in 8m 23s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c95520486c3547b7a26629a42d3905bc : SUCCESS in 4m 18s","accounts_in_message":[],"_revision_number":2},{"id":"d9a470811c8b2dc17bf9b701dd84ca6f30e7e371","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-07-16 13:50:49.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"765042cfd1d3b1fc70b1ccecb6db627b785f4c08","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-16 14:57:54.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/daeb7807a9e0425a8163196cf7e2ed84 : SUCCESS in 7m 07s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e57d7828e07e4e8eb529fa8ef0870a84 : SUCCESS in 4m 11s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/859083ec52a64288ae209e861ef3aa4f : SUCCESS in 5m 59s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/1218c11426ac40acbd74a0ce91af99f6 : SUCCESS in 12m 53s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/363d206ff924419b957a57bec689029b : SUCCESS in 13m 19s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/20c21e8771c04f339595d4b6dbf35abd : SUCCESS in 12m 18s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7af027887d6842cb92542da46870c5dc : FAILURE in 8m 14s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c205dbe27d6c4dcfbab91e50c9661680 : SUCCESS in 38m 53s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/5c5a6a91318e4833949ac3d1161ceb9f : SUCCESS in 1h 03m 49s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/d7a6918275b64649a6f1c01f6527373e : SUCCESS in 28m 27s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/3f63339d2aac4be9abf23dbf8759ee93 : FAILURE in 10m 40s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c5af633b45184b66b0c884b95db14111 : FAILURE in 8m 11s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2b0052d494bb461795859e22f921a496 : SUCCESS in 3m 39s","accounts_in_message":[],"_revision_number":2},{"id":"2c9595b7e2a3cafb0e26996e62a823a9a431b10c","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-07-16 15:03:02.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"261ad7bad411c0eddeb42a44582a73194ec5ab47","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-16 15:03:27.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"6fe5de5a5bd3d2a2e209db86ed5a5a7d98580924","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-16 16:17:48.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"2c3ce28db1191a7b8144a26083e757f350932ef5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-16 16:17:48.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/eb2bb5bfe6284a668163ba2caf13f918 : SUCCESS in 5m 09s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/26c455e1952c48bab2cca0d4d938fe36 : SUCCESS in 4m 58s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e18ee6742da944f596cbfabf3038f989 : SUCCESS in 5m 16s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/f0387d95658e499f963887de106132bc : SUCCESS in 12m 03s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/e69d60446cf74b0c91bfb78c622bfc4f : SUCCESS in 14m 35s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4ee142343fd2479fb05666f434612d15 : SUCCESS in 10m 48s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/4fc517acb5c7438cb446f755817582ef : SUCCESS in 1h 08m 21s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/2f96a47b35294b0498243a43c8bdcc39 : SUCCESS in 16m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c2ce5578fd784bae85a2f6e2df6f191c : SUCCESS in 5m 28s","accounts_in_message":[],"_revision_number":2},{"id":"1872657f06f44e45bf72328019c2b7458cd1ebf4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-16 16:19:21.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/39383290c2c34c9893987146dc615a46 : SUCCESS in 1m 14s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"7585243d3f60b4a7abd1037f7aa58aa3cd387bae","revisions":{"77b3cad03d2af4c56fce9f5bf9841df4996e8c67":{"kind":"REWORK","_number":1,"created":"2020-07-16 08:08:13.000000000","uploader":{"_account_id":32114,"name":"Lewis Denny","email":"ldenny@redhat.com","username":"ldenny"},"ref":"refs/changes/00/741400/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-trove","ref":"refs/changes/00/741400/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/00/741400/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/00/741400/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/00/741400/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-trove refs/changes/00/741400/1"}}},"commit":{"parents":[{"commit":"05a2bd8514c9ac5ff9aded5234968dcea0f9a276","subject":"Change to relative include","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/05a2bd8514c9ac5ff9aded5234968dcea0f9a276"}]}],"author":{"name":"Lewis Denny","email":"ldenny@redhat.com","date":"2020-07-16 08:06:54.000000000","tz":600},"committer":{"name":"Lewis Denny","email":"ldenny@redhat.com","date":"2020-07-16 08:08:10.000000000","tz":600},"subject":"Add support for the interface parameter in authtoken middleware","message":"Add support for the interface parameter in authtoken middleware\n\nThis patch adds support for [keystone_authtoken] interface parameter,\nso that operators can define which endpoint should be used by authtoken\nmiddleware.\n\nChange-Id: If61e708f68f96ce6c485a681d39e17c3cf737d2a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/77b3cad03d2af4c56fce9f5bf9841df4996e8c67"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/77b3cad03d2af4c56fce9f5bf9841df4996e8c67"}]},"branch":"refs/heads/master"},"7585243d3f60b4a7abd1037f7aa58aa3cd387bae":{"kind":"REWORK","_number":2,"created":"2020-07-16 11:29:16.000000000","uploader":{"_account_id":32114,"name":"Lewis Denny","email":"ldenny@redhat.com","username":"ldenny"},"ref":"refs/changes/00/741400/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-trove","ref":"refs/changes/00/741400/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/00/741400/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/00/741400/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/00/741400/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-trove refs/changes/00/741400/2"}}},"commit":{"parents":[{"commit":"05a2bd8514c9ac5ff9aded5234968dcea0f9a276","subject":"Change to relative include","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/05a2bd8514c9ac5ff9aded5234968dcea0f9a276"}]}],"author":{"name":"Lewis Denny","email":"ldenny@redhat.com","date":"2020-07-16 08:06:54.000000000","tz":600},"committer":{"name":"Lewis Denny","email":"ldenny@redhat.com","date":"2020-07-16 11:28:49.000000000","tz":600},"subject":"Add support for the interface parameter in authtoken middleware","message":"Add support for the interface parameter in authtoken middleware\n\nThis patch adds support for [keystone_authtoken] interface parameter,\nso that operators can define which endpoint should be used by authtoken\nmiddleware.\n\nChange-Id: If61e708f68f96ce6c485a681d39e17c3cf737d2a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/7585243d3f60b4a7abd1037f7aa58aa3cd387bae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/7585243d3f60b4a7abd1037f7aa58aa3cd387bae"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
