)]}'
{"id":"openstack%2Fpuppet-trove~749144","triplet_id":"openstack%2Fpuppet-trove~master~I24aa4a6e0e93712fb3240e25bba36558aac11887","project":"openstack/puppet-trove","branch":"master","hashtags":[],"change_id":"I24aa4a6e0e93712fb3240e25bba36558aac11887","subject":"Add Puppet Litmus","status":"MERGED","created":"2020-08-31 20:02:09.000000000","updated":"2020-09-11 21:01:51.000000000","submitted":"2020-09-11 21:01:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"749144-1599858111673-b51fe808","meta_rev_id":"b01f38ace51dd14c633483fa63f2d903e2742deb","_number":749144,"virtual_id_number":749144,"owner":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-11 21:01:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2020-09-11 12:37:12.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2020-09-11 19:40:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":1,"date":"2020-09-11 19:40:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-11 12:37:12.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-09-11 19:40:59.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2020-09-11 21:01:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f259011ae5c789ca3db45bd3f477ea924e85f84e","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-08-31 20:02:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e4dc4d849f2bbb16abe77eee120da6e33eb2e3dd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-31 21:08:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7b079a7c1156489c9b12f96bffc24e43 : SUCCESS in 3m 30s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/91b13bbbfa744c918ba1743408da245c : SUCCESS in 3m 25s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/4b67188e6c9b43179a35d29697bf9c3b : SUCCESS in 8m 51s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/196e64ea67b14f73b7d5667ff99a716f : SUCCESS in 7m 01s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/42f5007a79264b299dcc6a3db338532c : SUCCESS in 7m 30s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/280b0eb9bc814512a4143cc0b2b03d31 : SUCCESS in 32m 21s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/dc49cda4dd4e42d8af39d69f12ce7838 : SUCCESS in 56m 25s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/c4c9b0421c464e5f98c4161894d8e156 : SUCCESS in 14m 13s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5ed75b0d16dd4f7c85bcd5abf7949387 : SUCCESS in 11m 17s","accounts_in_message":[],"_revision_number":1},{"id":"9727cbb55c1f0ecc13f9ada7ce810322e110298f","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-09-08 18:27:31.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"fa54cad91c652aa41f57fd715ed86e35aa3eadc7","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-09-08 18:30:15.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2db300ed4288225c92e59bd8406b2f070981f2e8","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-09-09 21:26:38.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2efb37b43430291098feb237c8761e1e29acc830","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-09-09 22:56:58.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"7aee96f2539bcfe7da359eb3511a3ccdbb806cc6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-10 00:56:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/99ac015c1b244c1393610e117829796c : SUCCESS in 4m 37s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f170ced776314725a7b4e0ee713cccd0 : SUCCESS in 4m 27s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/b4275c9b80c248c08b55746944c1e2d5 : SUCCESS in 11m 54s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4bc2c8ec1c324b33aa7ab5872b3fe54a : SUCCESS in 10m 44s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/90539ab097cc4a6684c6e001d28dcbfe : SUCCESS in 11m 49s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/391682bd43304820b1cb016ab3e932fa : SUCCESS in 48m 06s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/5c560081159742eb8a9b741744a6b5d7 : SUCCESS in 1h 25m 00s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/012ebe5d44a44c5f908ace7cc1004244 : SUCCESS in 19m 32s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b5760ed301ff4bfc9ea9e2d11c778a59 : SUCCESS in 16m 29s","accounts_in_message":[],"_revision_number":1},{"id":"dbc40a5a71bdb36cda070aa494e206a269d4a278","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-10 00:56:26.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e83d37d69dc5716a2e7c4ddf127171a479d99417","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-10 01:50:45.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ea912eeeff5c4b33951393329091cd89 : SUCCESS in 3m 03s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d2533458de0d49ba98e964315c6bcb76 : SUCCESS in 3m 09s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/3a46f401156441b38ae762e443ed5a8c : SUCCESS in 9m 34s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d7d25750a44c44b7a1559ff998999f34 : SUCCESS in 9m 53s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/e5679a272e1e40f98483a2c3db471303 : SUCCESS in 52m 20s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/8a8da5cf342449c09afb6a1f414eb06d : SUCCESS in 21m 14s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/27e60f074e304dc88047d63654446c3e : FAILURE in 22m 49s","accounts_in_message":[],"_revision_number":1},{"id":"5b16a268e8da18264b597b31633b085edf7cd297","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-09-10 22:21:22.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"161ecc0dd537540947d6169043bf872a683b2ee2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-10 23:49:19.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/dce96ee4411e4b2c8b7a5ba2662e58df : SUCCESS in 3m 44s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/111ffc9bd3504a6c9438ec2c46861578 : SUCCESS in 3m 25s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/6805adcd9ace4cfb9cda27e1ecc976e2 : SUCCESS in 12m 38s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d3600548a84b4e5797680b4d88fd45f6 : SUCCESS in 8m 46s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b67b9853037c45cdb4d1dec089a1d826 : SUCCESS in 9m 24s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/948222b3506e4ee1b56cf63e460222d8 : SUCCESS in 49m 42s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/728c0517828242718e9e92b9be65077f : SUCCESS in 1h 24m 45s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/089bda5a9712454a9ec18799c2437e2b : SUCCESS in 17m 55s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6596fca36ac54531b5cf49b8687e1757 : FAILURE in 17m 45s","accounts_in_message":[],"_revision_number":1},{"id":"eb4e71c8b5862c6649d4a1a959277ad7eab63d74","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-09-11 09:43:24.000000000","message":"Patch Set 1:\n\ntaskamanger not idempotent issue","accounts_in_message":[],"_revision_number":1},{"id":"a126b2fcaf86098a3719dfbdf89f5c4b05105720","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-09-11 10:14:41.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"b58aa5acbf12a56d5025509d8cfb1236d5970889","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-11 11:44:16.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8954e59f678744f68e770cc38b3aa23b : SUCCESS in 4m 01s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c12d34aceb8a41d9848f2340569c22a1 : SUCCESS in 4m 02s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/a38c63b7b52a43c5b030dbe8093a7758 : SUCCESS in 12m 31s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/aeb02be821d14ed3b2d0fc12a3f45c0f : SUCCESS in 10m 15s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c11e9fa090f44bc990c960f6049b1e14 : SUCCESS in 11m 04s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/86278517c90547c49d30ee9f1cf2c395 : SUCCESS in 50m 30s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/3d67a9c6c76a49aeba13648469de75ec : SUCCESS in 1h 21m 49s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/e6c0fa4b28a74ddfbc7a9ab7cceeac89 : SUCCESS in 20m 39s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8e0728be3eb24f7e89dec4d9eaca3990 : SUCCESS in 15m 11s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"1107aef8370c764422fa6eccef1fb0e5cfd6f96e","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-09-11 12:37:12.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"9d0159d3bcc6de4ef919010b2ac007e9565c9c75","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-09-11 19:40:59.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"cbf9cba702d5db1861249b0101de53cd376e501a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-11 19:41:25.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d7fad0b9c808f06a3c6c83f0e8bc24f8381f763d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-11 21:01:51.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"b01f38ace51dd14c633483fa63f2d903e2742deb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-11 21:01:51.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/70d2f8d4e960491cbc5eece9b220757b : SUCCESS in 4m 54s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1b58fd7187b54e72a117dc64df8a1d0c : SUCCESS in 4m 16s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/26452e6d49054966a4277359f4636352 : SUCCESS in 12m 20s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c8972b13be3e4ef092ee1fdc212a2d7a : SUCCESS in 12m 57s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/6f2d9f71644549e9b34601a1e101a6f7 : SUCCESS in 1h 13m 23s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/ffdda2bf97594cf593af73c3c37e0a6d : SUCCESS in 21m 01s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/14d2550617bc47a384f8c813a328d33f : SUCCESS in 16m 54s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"1de53afc37da2d86732cf82465f2d52088d1e866","revisions":{"859cf6c00b8cfefbdd53807cdfcb07ddbe596769":{"kind":"REWORK","_number":1,"created":"2020-08-31 20:02:09.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/44/749144/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-trove","ref":"refs/changes/44/749144/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/44/749144/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/44/749144/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/44/749144/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-trove refs/changes/44/749144/1"}}},"commit":{"parents":[{"commit":"3fc1d517547e0a55ccc7be7b866ad44ec3d688ae","subject":"Prepare Victoria M1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/3fc1d517547e0a55ccc7be7b866ad44ec3d688ae"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2020-08-31 20:02:01.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2020-08-31 20:02:01.000000000","tz":120},"subject":"Add Puppet Litmus","message":"Add Puppet Litmus\n\nDepends-On: https://review.opendev.org/#/c/740601/\nChange-Id: I24aa4a6e0e93712fb3240e25bba36558aac11887\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/859cf6c00b8cfefbdd53807cdfcb07ddbe596769"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/859cf6c00b8cfefbdd53807cdfcb07ddbe596769"}]},"branch":"refs/heads/master"},"1de53afc37da2d86732cf82465f2d52088d1e866":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-09-11 10:14:41.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/44/749144/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-trove","ref":"refs/changes/44/749144/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/44/749144/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/44/749144/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/44/749144/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-trove refs/changes/44/749144/2"}}},"commit":{"parents":[{"commit":"3fc1d517547e0a55ccc7be7b866ad44ec3d688ae","subject":"Prepare Victoria M1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/3fc1d517547e0a55ccc7be7b866ad44ec3d688ae"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2020-08-31 20:02:01.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2020-09-11 10:14:39.000000000","tz":0},"subject":"Add Puppet Litmus","message":"Add Puppet Litmus\n\nDepends-On: https://review.opendev.org/#/c/751233\nChange-Id: I24aa4a6e0e93712fb3240e25bba36558aac11887\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/1de53afc37da2d86732cf82465f2d52088d1e866"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/1de53afc37da2d86732cf82465f2d52088d1e866"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
