)]}'
{"id":"openstack%2Fpuppet-trove~785810","triplet_id":"openstack%2Fpuppet-trove~master~If5a1becaa44bf082ef4b8430e87348f4df980abe","project":"openstack/puppet-trove","branch":"master","topic":"policy-dirs","hashtags":[],"change_id":"If5a1becaa44bf082ef4b8430e87348f4df980abe","subject":"Allow to configure policy_dirs","status":"MERGED","created":"2021-04-11 21:49:43.000000000","updated":"2021-04-12 13:42:30.000000000","submitted":"2021-04-12 13:40:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"785810-policy-dirs","meta_rev_id":"bea48971eeddf19b8e2995be36288f5bf7793c7b","_number":785810,"virtual_id_number":785810,"owner":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-12 13:40:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2021-04-11 22:46:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2021-04-12 11:58:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":1,"date":"2021-04-12 11:58:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-11 22:46:53.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2021-04-11 23:24:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-12 11:58:25.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"4c29a598f8ffb523bdfea98fac61b4eb963c15e6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2021-04-11 21:49:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"73a172fa52216cbce3bedc078d5b9675e40c4bea","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-04-11 22:46:53.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"fd85d5f6869a7f494ade4a7bc1c66c013fd51144","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-04-11 22:47:00.000000000","message":"Topic set to policy-dirs","accounts_in_message":[],"_revision_number":1},{"id":"e6fef01fd8cf0c7badfc9067841fbb663de775a5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-11 23:24:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d529bb74a8db4530baa3c8f5d7e9b413 : SUCCESS in 4m 07s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/418b768630e84a50a751dcd0ef44cc4a : SUCCESS in 3m 41s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/345a5232db9c4a8bb49622730996d49f : SUCCESS in 11m 39s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/df4629d6135947b09c4319df94cf5ba2 : SUCCESS in 8m 04s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3a2bf5824b7d4fcab0451d4f36a7b82e : SUCCESS in 6m 22s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/badab976ff0044dd832c6f1753988a98 : FAILURE in 34m 56s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/6750fde765fd49d1ab6276604082b594 : SUCCESS in 1h 33m 12s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/9ae5d42a9c304d5baaf724f2cb2037ea : SUCCESS in 18m 32s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4337dc42dfbc4545b5eb0f56888ef341 : SUCCESS in 12m 26s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/307f7ea44e0149a18f6d2cb78797d55b : SUCCESS in 3m 38s","accounts_in_message":[],"_revision_number":1},{"id":"6a20030a52a07d24c1e3f929ca9b8bad146531ef","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2021-04-12 11:58:25.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"dfed8daf56ecec028f833b8c519bbb7c04c92011","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-12 11:58:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0cdfd053e3ddd91c786712888591f118ed0ffe48","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-12 13:40:12.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e504e652f0224b82a1af131e55924250 : SUCCESS in 3m 56s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3004509252cb47dc96d75123ebc13956 : SUCCESS in 4m 26s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/21b9801060014fb884a7fc56477d9740 : SUCCESS in 10m 14s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/30e0d30fd69441a2b7b0647786a0bddf : SUCCESS in 8m 29s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/bbc776dc00a14d9c96cd7b3f7ebdd413 : SUCCESS in 1h 35m 48s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/22083eabcd8d43e49cf1c70ce6dc4979 : SUCCESS in 18m 59s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/042785e6411e485882b49ae2d2121b52 : SUCCESS in 13m 46s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/146e358d4f254ffa9fd62340698a0da6 : SUCCESS in 4m 05s","accounts_in_message":[],"_revision_number":1},{"id":"7b8b3d0613745c50bf46484d75206ec514da3aba","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-12 13:40:13.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"bea48971eeddf19b8e2995be36288f5bf7793c7b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-12 13:42:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/70d21a008d6b42fc845931356d6351e9 : SUCCESS in 1m 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a4f4412946112a0c0dfda749c2a9b1cc38f228fb","revisions":{"a4f4412946112a0c0dfda749c2a9b1cc38f228fb":{"kind":"REWORK","_number":1,"created":"2021-04-11 21:49:43.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/10/785810/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-trove","ref":"refs/changes/10/785810/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/10/785810/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/10/785810/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/10/785810/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-trove refs/changes/10/785810/1"}}},"commit":{"parents":[{"commit":"d3bdda4aff03d4e98699691a9b36c8c1c378eeaf","subject":"Update master for stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/d3bdda4aff03d4e98699691a9b36c8c1c378eeaf"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2021-04-11 21:49:06.000000000","tz":120},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2021-04-11 21:49:39.000000000","tz":120},"subject":"Allow to configure policy_dirs","message":"Allow to configure policy_dirs\n\nThis patch makes it possible to override the current service default,\nwhich is /etc/\u003cservice\u003e/policy.d .\n\nChange-Id: If5a1becaa44bf082ef4b8430e87348f4df980abe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/a4f4412946112a0c0dfda749c2a9b1cc38f228fb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/a4f4412946112a0c0dfda749c2a9b1cc38f228fb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
