)]}'
{"id":"openstack%2Fpuppet-trove~787886","triplet_id":"openstack%2Fpuppet-trove~master~Ie95e02f65bf295138d08bac7f8caaa79617c7b0d","project":"openstack/puppet-trove","branch":"master","topic":"trove_config","hashtags":[],"change_id":"Ie95e02f65bf295138d08bac7f8caaa79617c7b0d","subject":"WIP: Register common options to trove.conf","status":"ABANDONED","created":"2021-04-25 14:28:02.000000000","updated":"2022-02-22 17:24:49.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"b2f4e15f50f07146a8726c476cd906e86bac928f","_number":787886,"virtual_id_number":787886,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-04-26 02:48:37.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-25 15:00:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"72ac57691bbcfebcbeb618b933133ede9183a6c4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-04-25 14:28:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9811766ed8916423584762b23778ce7063c84225","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-25 15:00:49.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2abd2ab51d7f4259b854f50fc7bb6628 : SUCCESS in 3m 03s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aac08801baac44f0ab1008bc4fa21351 : SUCCESS in 4m 06s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/695dba1104dc446b9c248ca3949ed710 : FAILURE in 10m 18s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/54bc57514fe646e488ce217c4efeb714 : FAILURE in 8m 52s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6210cbcc78a04f67af9006135f90ae04 : FAILURE in 7m 58s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c02ef4edb2ad4bd38145fd1ce8b1cc1d : FAILURE in 27m 44s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/a76903c464e54470a4b62a5025382919 : FAILURE in 10m 54s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/65566550a5b84139a2c58609adeaf47e : FAILURE in 9m 20s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/12d65dad2dee43739207462de5e3108e : FAILURE in 6m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fc9a38f526b3f2a6a9347bb813a44b44bae7f93d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-04-25 15:27:53.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ddf44e430977f8ce74b7824ac355a7d7538b8611","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-25 16:05:42.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1aaac643e9604b1ebceade65629cce47 : SUCCESS in 3m 36s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7ecaa1df78dd466795f68757d15b03df : SUCCESS in 3m 37s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/9f57c525e6194c00945af3230bfee818 : FAILURE in 13m 54s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2951eaec3ac74136ac3e45e1b0aecf3d : FAILURE in 9m 46s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/910c20043b2c4ea79e4dd9048ef5f0cc : FAILURE in 14m 27s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/157114036f5443c2b4d6582b64120a5a : FAILURE in 28m 41s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/39b847703bff4f3d8789a24084865d8b : FAILURE in 34m 24s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/099e6816cbab4eab933e8057bee5b8fe : FAILURE in 19m 23s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/930661a0c6a246f1b800fc3ae9b12d42 : FAILURE in 20m 35s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"5398f49aa046cedae6f6c57d996dfa01474a1211","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-04-26 00:51:56.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"1e264bda94836bd8d2f9043ee57a8f09ceeb0522","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-04-26 00:55:38.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"02f94b154c86775b2b2a4262b47f415dc85cf670","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-26 02:48:37.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2687cec7abea4a7eaffec8ff4c3f516f : FAILURE in 3m 35s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/133db6ffa98c4375b0df6bcaadea527a : FAILURE in 3m 06s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/c247ea62f25e48858202edd5ec544d1b : FAILURE in 9m 40s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a824f69760de4a83865ab2e0c25c5838 : FAILURE in 6m 33s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/40bb03927f014c3589144e62127a846b : FAILURE in 8m 00s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aa236f80fe7c4091b03e8529c6c14124 : FAILURE in 22m 05s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/76f4e07d6f7c4173892f436358237952 : SUCCESS in 1h 36m 57s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/b609215ac4cd4e4cb58c8a24d950a463 : SUCCESS in 14m 02s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cc92a63be0764ae2a7eda39eb149f17d : SUCCESS in 12m 31s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"b2f4e15f50f07146a8726c476cd906e86bac928f","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-02-22 17:24:49.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"7c03f0ff76e57369e02f55192e033510fb606fc9","revisions":{"7fc896a1bec8cda91630662643c612807ae942d1":{"kind":"REWORK","_number":1,"created":"2021-04-25 14:28:02.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/86/787886/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-trove","ref":"refs/changes/86/787886/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/86/787886/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/86/787886/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/86/787886/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-trove refs/changes/86/787886/1"}}},"commit":{"parents":[{"commit":"8781fa19ea9a439a192612aea6109b405a787ef4","subject":"Merge \"Bump dependency in metadata.json\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/8781fa19ea9a439a192612aea6109b405a787ef4"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-04-25 14:27:35.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-04-25 14:27:35.000000000","tz":540},"subject":"WIP: Register common options to trove.conf","message":"WIP: Register common options to trove.conf\n\nChange-Id: Ie95e02f65bf295138d08bac7f8caaa79617c7b0d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/7fc896a1bec8cda91630662643c612807ae942d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/7fc896a1bec8cda91630662643c612807ae942d1"}]},"branch":"refs/heads/master"},"911cd81e056af8d4c5837df794726ef8ddfd1d0f":{"kind":"REWORK","_number":2,"created":"2021-04-25 15:27:53.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/86/787886/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-trove","ref":"refs/changes/86/787886/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/86/787886/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/86/787886/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/86/787886/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-trove refs/changes/86/787886/2"}}},"commit":{"parents":[{"commit":"8781fa19ea9a439a192612aea6109b405a787ef4","subject":"Merge \"Bump dependency in metadata.json\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/8781fa19ea9a439a192612aea6109b405a787ef4"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-04-25 14:27:35.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-04-25 15:27:44.000000000","tz":540},"subject":"WIP: Register common options to trove.conf","message":"WIP: Register common options to trove.conf\n\nChange-Id: Ie95e02f65bf295138d08bac7f8caaa79617c7b0d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/911cd81e056af8d4c5837df794726ef8ddfd1d0f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/911cd81e056af8d4c5837df794726ef8ddfd1d0f"}]},"branch":"refs/heads/master"},"568a65b4fb5164896a6adf7e3fce6ebd430a26cb":{"kind":"REWORK","_number":3,"created":"2021-04-26 00:51:56.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/86/787886/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-trove","ref":"refs/changes/86/787886/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/86/787886/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/86/787886/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/86/787886/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-trove refs/changes/86/787886/3"}}},"commit":{"parents":[{"commit":"8781fa19ea9a439a192612aea6109b405a787ef4","subject":"Merge \"Bump dependency in metadata.json\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/8781fa19ea9a439a192612aea6109b405a787ef4"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-04-25 14:27:35.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-04-26 00:51:46.000000000","tz":540},"subject":"WIP: Register common options to trove.conf","message":"WIP: Register common options to trove.conf\n\nChange-Id: Ie95e02f65bf295138d08bac7f8caaa79617c7b0d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/568a65b4fb5164896a6adf7e3fce6ebd430a26cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/568a65b4fb5164896a6adf7e3fce6ebd430a26cb"}]},"branch":"refs/heads/master"},"7c03f0ff76e57369e02f55192e033510fb606fc9":{"kind":"REWORK","_number":4,"created":"2021-04-26 00:55:38.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/86/787886/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-trove","ref":"refs/changes/86/787886/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/86/787886/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/86/787886/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/86/787886/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-trove refs/changes/86/787886/4"}}},"commit":{"parents":[{"commit":"8781fa19ea9a439a192612aea6109b405a787ef4","subject":"Merge \"Bump dependency in metadata.json\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/8781fa19ea9a439a192612aea6109b405a787ef4"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-04-25 14:27:35.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-04-26 00:55:22.000000000","tz":540},"subject":"WIP: Register common options to trove.conf","message":"WIP: Register common options to trove.conf\n\nChange-Id: Ie95e02f65bf295138d08bac7f8caaa79617c7b0d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/7c03f0ff76e57369e02f55192e033510fb606fc9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/7c03f0ff76e57369e02f55192e033510fb606fc9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
