)]}'
{"id":"openstack%2Fpuppet-trove~794128","triplet_id":"openstack%2Fpuppet-trove~master~I93c83dfe9b92ef772e1ae3c15f70aeac39ccf449","project":"openstack/puppet-trove","branch":"master","topic":"bug/1930403","hashtags":[],"change_id":"I93c83dfe9b92ef772e1ae3c15f70aeac39ccf449","subject":"Set fixture paths for unit tests","status":"MERGED","created":"2021-06-02 05:26:29.000000000","updated":"2021-06-03 00:14:53.000000000","submitted":"2021-06-03 00:14:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"794128-bug/1930403","meta_rev_id":"814fb07f73d80cd2c727e12cb49965c6401d96b8","_number":794128,"virtual_id_number":794128,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-03 00:14:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2021-06-02 23:53:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2021-06-02 15:53:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2021-06-02 23:53:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-02 07:07:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-02 15:53:47.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"d9de6f7918099cda4ae883782250269cbe9167ff","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-06-02 05:26:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"985cabac17100ae05573cdec7bfbd0683942f482","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-02 07:07:36.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d5bf4dc99a1c43d486eae3de59916c25 : SUCCESS in 3m 39s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cf5786965991439c8be4f33866869393 : SUCCESS in 4m 36s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5b28acff529f411bb1789aadd64b59e5 : SUCCESS in 3m 27s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/315c15b109034faba0164cfc693f7786 : SUCCESS in 11m 33s\n- puppet-openstack-unit-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/4ad065ab177b4a3b8b01beecd6549fb8 : SUCCESS in 12m 00s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a15aa1cd18f14c6999984bdcbd037e54 : SUCCESS in 8m 54s\n- puppet-openstack-unit-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dc9b5aea60154ac68a403eeaa3418157 : SUCCESS in 8m 52s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dd3c1204644b49d79f85ac752a4f5c0d : SUCCESS in 9m 55s (non-voting)\n- puppet-openstack-litmus-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/a7aa8e1c15b74d6b8b330947258b140f : SUCCESS in 20m 25s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f7ceffaab2ad42949bee2977795a2c95 : FAILURE in 15m 05s (non-voting)\n- puppet-openstack-litmus-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/9ac1008d45bd44fe961647957f517ad9 : TIMED_OUT in 1h 39m 11s\n- puppet-openstack-litmus-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/67332e922cef42c29cbbb5d0d93e7ef2 : FAILURE in 15m 00s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ea2ca92a7d4148a1be83f0993d95acb1 : FAILURE in 15m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8019f31b7596d3ce51257bd1e02630b509e6c34f","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-06-02 08:03:43.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"a6aba28b1904ea38d26d0bf1937fc3cc9df56d01","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-02 08:38:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f6a32c6f274e48919deedd78508a3a99 : SUCCESS in 4m 04s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a623cb68e9f244c79b9ba96c453add96 : SUCCESS in 4m 09s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/853c39cef6574943b8030aa6be641abd : SUCCESS in 3m 53s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/7a804909c95645708fbb8edc56465ec9 : SUCCESS in 12m 43s\n- puppet-openstack-unit-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/eafedfd614924e80b3aae6d078884768 : SUCCESS in 12m 46s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0b883b222c1645ef89998afb51dc7850 : SUCCESS in 11m 51s\n- puppet-openstack-unit-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/37922b94cf9d4910a342b66c5e2b2d6b : SUCCESS in 11m 18s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9619e3e07f18421c855cc48defe63430 : SUCCESS in 10m 28s (non-voting)\n- puppet-openstack-litmus-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/26d00526f5864c57aaa84b8843808187 : SUCCESS in 28m 25s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9086b6a15b8248de8c89b46265c45792 : FAILURE in 14m 40s (non-voting)\n- puppet-openstack-litmus-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/614a96286d944715a8d039a8ddc4faf5 : SUCCESS in 18m 46s\n- puppet-openstack-litmus-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0c6a1f233d2b4d6db354a4bdee414f18 : FAILURE in 17m 02s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b81fdc4e1ea448e7b93cdad159827b99 : FAILURE in 20m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"920cd3c62ea280b5880f15037d7d8f891e0c829c","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-06-02 15:53:47.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c729e7df0d322d58dbb35c64316f41d7735e5ca6","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-06-02 23:53:17.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nFast approving because this is a gate blocker now","accounts_in_message":[],"_revision_number":1},{"id":"f1de69c35f2e79404f6ee1e10a32305a7053535d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-02 23:54:15.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bb0999d01490348ae0873242851508e4b1b55d0c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-03 00:14:51.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0c2e34fdebf54611aa132ad1bb131588 : SUCCESS in 3m 58s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ae5a90a981af46b7a34c41b2134abb0f : SUCCESS in 3m 42s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/60772ebeb1d443ab83b346863ba8e521 : SUCCESS in 4m 28s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/26b2d62f93054d6391b1da2886cab9c3 : SUCCESS in 13m 23s\n- puppet-openstack-unit-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/0325383eabb14496b6e6c535c7c43833 : SUCCESS in 13m 31s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/666cd67fd75d499cb90c3d4bbf3b1865 : SUCCESS in 9m 41s\n- puppet-openstack-unit-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6312235ff66f4caa8da99ba5311f8b9c : SUCCESS in 10m 22s\n- puppet-openstack-litmus-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/c835d9036beb4cc4ba65de3695ea9e10 : SUCCESS in 19m 35s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0b0297cc517542a1a1f9b1517b59c202 : FAILURE in 14m 38s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"814fb07f73d80cd2c727e12cb49965c6401d96b8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-03 00:14:53.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fe0f6508c45e0a0f1b6061f3d7b8ed6299691958","revisions":{"fe0f6508c45e0a0f1b6061f3d7b8ed6299691958":{"kind":"REWORK","_number":1,"created":"2021-06-02 05:26:29.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/28/794128/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-trove","ref":"refs/changes/28/794128/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/28/794128/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/28/794128/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/28/794128/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-trove refs/changes/28/794128/1"}}},"commit":{"parents":[{"commit":"7344f7c463d7f882684e2830f464211efdbfe9a5","subject":"setup.cfg: Replace dashes with underscores","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/7344f7c463d7f882684e2830f464211efdbfe9a5"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-06-02 05:26:17.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-06-02 05:26:20.000000000","tz":540},"subject":"Set fixture paths for unit tests","message":"Set fixture paths for unit tests\n\nThis change defines manifest_dir and module_path expicitly in unit\ntests so that modules installed under fixtures directory is properly\nloaded.\n\nCloses-Bug: #1930403\nChange-Id: I93c83dfe9b92ef772e1ae3c15f70aeac39ccf449\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/fe0f6508c45e0a0f1b6061f3d7b8ed6299691958"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/fe0f6508c45e0a0f1b6061f3d7b8ed6299691958"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
