)]}'
{"id":"openstack%2Fpuppet-trove~796132","triplet_id":"openstack%2Fpuppet-trove~master~I118d45d409b22b34665383ea58fb6f31c0484efc","project":"openstack/puppet-trove","branch":"master","topic":"pyvers-removal","hashtags":[],"change_id":"I118d45d409b22b34665383ea58fb6f31c0484efc","subject":"Get rid of the $pyvers variable","status":"ABANDONED","created":"2021-06-12 14:51:13.000000000","updated":"2021-06-14 10:34:52.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"2597a497a9e6fa3d73fa64a823ff921b8b481225","_number":796132,"virtual_id_number":796132,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2021-06-13 03:54:04.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-12 16:17:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"63a40ca51e0b037b7fcdff4689ff878f7121f8fa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-06-12 14:51:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3490a274359a74c27874b9c51657510378e1bc64","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-06-12 14:54:44.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"182ff483e1158dee908a7c808d728d678f2e0f98","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-06-12 14:57:47.000000000","message":"Uploaded patch set 3: New patch set was added with same tree, parent, and commit message as Patch Set 2.","accounts_in_message":[],"_revision_number":3},{"id":"28f6f72e5c5fa0419b02065b6245ef06657efe93","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-12 16:17:21.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e71486f819324af9ade7944d48208de7 : SUCCESS in 3m 45s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3e13a840c7094f8296cf160653d61dfd : SUCCESS in 3m 49s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3234014b18114562909674e1290cf60b : SUCCESS in 11m 01s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/78ed88f93a4a4d159ec309ffdbc4c72a : SUCCESS in 11m 40s\n- puppet-openstack-unit-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/9eb0e96ff5f14a40a7a12fb1eb118773 : SUCCESS in 11m 32s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a03704ce88dd4224babf882e72ea4333 : SUCCESS in 8m 16s\n- puppet-openstack-unit-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c92e6c631acf42f09654210bfb63c35f : SUCCESS in 8m 11s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6d5b4d94d33e4a73adfb4dff29189b94 : SUCCESS in 7m 34s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9668ad428d5545faad7ca6e2232c8144 : FAILURE in 27m 41s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/d0dabcc5259e437caecda6888ddbd7a6 : FAILURE in 54m 45s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/9bceeba0d38c412eb8e76a024b3cf232 : SUCCESS in 1h 18m 59s\n- puppet-openstack-litmus-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/b2cf622d5b36422eb1c6fe229d115c36 : SUCCESS in 20m 07s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/149afc9348ad4c599e37d3d4fe31c018 : FAILURE in 15m 05s (non-voting)\n- puppet-openstack-litmus-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/4655dac62a174639bd402b75e7bd314b : SUCCESS in 26m 35s\n- puppet-openstack-litmus-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b0fa8ad7c83f4312b5b7652106f5c3b7 : FAILURE in 14m 09s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/714d1a93fec24b118c964d782116c68d : FAILURE in 14m 05s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"df36a10652e0ad52f30790f99c59a1d4687b5603","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-06-13 02:09:40.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"c00ee52d53fa89f00624d534de1c5562694ca895","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-13 03:54:04.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ba38c445418a443ba255daab4245fedd : SUCCESS in 4m 56s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0d8a05ac72b441f2872b0015e65ae38f : SUCCESS in 3m 22s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b3bdf62417dc40c6bfa8bb0c91a641a4 : SUCCESS in 3m 38s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/20dc6001f0d84facaca3bd5f20c2b204 : SUCCESS in 10m 45s\n- puppet-openstack-unit-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/8fb133d864b7494eae736481acfec745 : SUCCESS in 11m 40s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b7e6126edede45e89a04d2578c225146 : SUCCESS in 8m 12s\n- puppet-openstack-unit-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3256899052af45ec9360cc96b48d4b6d : SUCCESS in 8m 30s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d764629a863e4f0f827bd3355435f88b : SUCCESS in 8m 39s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/24e161fe130a495ebc32f04fa9d55f0d : FAILURE in 28m 35s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/3f3712d2b4ba42c2a4a95c8a5785460a : SUCCESS in 1h 27m 51s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/7e757ecac7e845a7a389b81183cb04e4 : SUCCESS in 1h 38m 22s\n- puppet-openstack-litmus-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/c0512a0eb82a434b99fb67cc407e01bc : SUCCESS in 20m 06s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/812865537a054c118ccafaf89e880c9d : FAILURE in 14m 45s (non-voting)\n- puppet-openstack-litmus-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/4aeab41efe55460d956ac69bc18d69e2 : SUCCESS in 19m 24s\n- puppet-openstack-litmus-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aae8a8e64edf44e69f91ad2057282eec : FAILURE in 14m 00s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f997f685df7141e9b775050540ddb774 : FAILURE in 13m 46s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"2597a497a9e6fa3d73fa64a823ff921b8b481225","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-06-14 10:34:52.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"659fd18cdcf59e6381113703515a94ee7b02b422","revisions":{"35191012a0aa1eddcc01e482a13fde9ca85e255b":{"kind":"REWORK","_number":1,"created":"2021-06-12 14:51:13.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/32/796132/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-trove","ref":"refs/changes/32/796132/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/32/796132/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/32/796132/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/32/796132/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-trove refs/changes/32/796132/1"}}},"commit":{"parents":[{"commit":"fe0f6508c45e0a0f1b6061f3d7b8ed6299691958","subject":"Set fixture paths for unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/fe0f6508c45e0a0f1b6061f3d7b8ed6299691958"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-06-12 14:50:25.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-06-12 14:51:04.000000000","tz":540},"subject":"Get rid of the $pyvers variable","message":"Get rid of the $pyvers variable\n\nSince everyone has switched to Python3, it\u0027s time for the removal of the\n$pyvers variable.\n\nCo-authored-by: Thomas Goirand \u003czigo@debian.org\u003e\nChange-Id: I118d45d409b22b34665383ea58fb6f31c0484efc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/35191012a0aa1eddcc01e482a13fde9ca85e255b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/35191012a0aa1eddcc01e482a13fde9ca85e255b"}]},"branch":"refs/heads/master"},"9366a409f4532098b67fc1f7e8f53d5965cdd6ac":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-06-12 14:54:44.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/32/796132/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-trove","ref":"refs/changes/32/796132/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/32/796132/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/32/796132/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/32/796132/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-trove refs/changes/32/796132/2"}}},"commit":{"parents":[{"commit":"b07c761623d00d6ab5572853862bd74575f61e54","subject":"Merge \"Set fixture paths for unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/b07c761623d00d6ab5572853862bd74575f61e54"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-06-12 14:50:25.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-06-12 14:54:44.000000000","tz":0},"subject":"Get rid of the $pyvers variable","message":"Get rid of the $pyvers variable\n\nSince everyone has switched to Python3, it\u0027s time for the removal of the\n$pyvers variable.\n\nCo-authored-by: Thomas Goirand \u003czigo@debian.org\u003e\nChange-Id: I118d45d409b22b34665383ea58fb6f31c0484efc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/9366a409f4532098b67fc1f7e8f53d5965cdd6ac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/9366a409f4532098b67fc1f7e8f53d5965cdd6ac"}]},"branch":"refs/heads/master","description":"Rebase"},"659fd18cdcf59e6381113703515a94ee7b02b422":{"kind":"NO_CHANGE","_number":3,"created":"2021-06-12 14:57:47.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/32/796132/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-trove","ref":"refs/changes/32/796132/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/32/796132/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/32/796132/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/32/796132/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-trove refs/changes/32/796132/3"}}},"commit":{"parents":[{"commit":"b07c761623d00d6ab5572853862bd74575f61e54","subject":"Merge \"Set fixture paths for unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/b07c761623d00d6ab5572853862bd74575f61e54"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-06-12 14:50:25.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-06-12 14:57:22.000000000","tz":540},"subject":"Get rid of the $pyvers variable","message":"Get rid of the $pyvers variable\n\nSince everyone has switched to Python3, it\u0027s time for the removal of the\n$pyvers variable.\n\nCo-authored-by: Thomas Goirand \u003czigo@debian.org\u003e\nChange-Id: I118d45d409b22b34665383ea58fb6f31c0484efc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/659fd18cdcf59e6381113703515a94ee7b02b422"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/659fd18cdcf59e6381113703515a94ee7b02b422"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
