)]}'
{"id":"openstack%2Fpuppet-trove~810629","triplet_id":"openstack%2Fpuppet-trove~master~I650cd44a18d8d5f4c8da18febc2c35705e01262b","project":"openstack/puppet-trove","branch":"master","topic":"update-stdlib-version-8.0","hashtags":[],"change_id":"I650cd44a18d8d5f4c8da18febc2c35705e01262b","subject":"Fix stdlib version in metadata","status":"MERGED","created":"2021-09-23 12:01:17.000000000","updated":"2021-09-23 12:28:07.000000000","submitted":"2021-09-23 12:28:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"810629-update-stdlib-version-8.0","meta_rev_id":"4f85c201bcfc4fb80cbf21b3f15e02b12af99006","_number":810629,"virtual_id_number":810629,"owner":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-23 12:28:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":2,"date":"2021-09-23 12:19:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":1,"date":"2021-09-23 12:19:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-23 12:09:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ec22406da015be55f97841bb23932939b46d0931","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2021-09-23 12:01:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"eec673aed4dc023a058b6b64ab41941de3cc17dd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-23 12:09:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/48038123756541aa81eb03c2fb69bbfa : SUCCESS in 2m 47s\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4ec4170c54e148db8a02312d9f9c2887 : SUCCESS in 4m 07s","accounts_in_message":[],"_revision_number":1},{"id":"58e95bbe06e5a3df9a83ce3dd495512cc4a6557b","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2021-09-23 12:19:03.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7c4a21f9bb41af935af9b655008a8f5750ba4646","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-23 12:21:58.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8fa54203fb76ceb7de53728f460ce6c6545b1ead","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-23 12:28:07.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/7c7829ca1fdd48448177924ce6159137 : SUCCESS in 3m 21s\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/22dfdc05d1d64540806d94265c6e5868 : SUCCESS in 4m 34s","accounts_in_message":[],"_revision_number":1},{"id":"4f85c201bcfc4fb80cbf21b3f15e02b12af99006","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-23 12:28:07.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5516ff56bedec7be68597d7666afa9c1ced848b2","revisions":{"5516ff56bedec7be68597d7666afa9c1ced848b2":{"kind":"REWORK","_number":1,"created":"2021-09-23 12:01:17.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/29/810629/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-trove","ref":"refs/changes/29/810629/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/29/810629/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/29/810629/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/29/810629/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-trove refs/changes/29/810629/1"}}},"commit":{"parents":[{"commit":"f5ba799ce0b5f94019ba67c35a35a8b105c2fc7f","subject":"Prepare Xena RC1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/f5ba799ce0b5f94019ba67c35a35a8b105c2fc7f"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2021-09-23 11:44:23.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2021-09-23 11:44:23.000000000","tz":480},"subject":"Fix stdlib version in metadata","message":"Fix stdlib version in metadata\n\npuppetlabs-stdlib is now using the 8.0.0 version, so we should fix the\nstdlib version in metadata.\n\nChange-Id: I650cd44a18d8d5f4c8da18febc2c35705e01262b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/5516ff56bedec7be68597d7666afa9c1ced848b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/5516ff56bedec7be68597d7666afa9c1ced848b2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
