)]}'
{"id":"openstack%2Fpuppet-trove~826015","triplet_id":"openstack%2Fpuppet-trove~master~I3ace0192837442a4d3952cc7b75ed2e93426effa","project":"openstack/puppet-trove","branch":"master","topic":"auth-ut","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-01-24 17:37:11.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I3ace0192837442a4d3952cc7b75ed2e93426effa","subject":"Do not test detail of keystone::auth::service_identity","status":"MERGED","created":"2022-01-24 06:24:41.000000000","updated":"2022-01-24 17:37:11.000000000","submitted":"2022-01-24 17:37:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"826015-auth-ut","meta_rev_id":"d06ba754a404d4f546a456ec89af7afbe76ae802","_number":826015,"virtual_id_number":826015,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-24 17:37:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2022-01-24 07:59:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2022-01-24 17:16:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2022-01-24 17:16:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-24 06:43:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-24 07:59:39.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2022-01-24 17:16:58.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"8cb355b4706ba9cad832ee7367195d35b26acace","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-01-24 06:24:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d8feeebd336cb1276ddb19d1226fa8b4b5b217d8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-01-24 06:25:09.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"310bde81391a6075e186d286d67ae5749c41ab25","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-24 06:43:29.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b753060cc003443ba36bf42470e91637 : SUCCESS in 3m 22s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bec9463d206a461abfb8f5cd25627819 : SUCCESS in 2m 54s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3fe8ae4b4f004ec6b880beb39c51464c : SUCCESS in 2m 41s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/3e007b2423604c5f94f72c83c7f876bd : SUCCESS in 9m 22s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/069bf8d14c874567aac3d5592dbfa14f : SUCCESS in 9m 13s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/53c917069ed048bdb150e9651636fa9e : SUCCESS in 6m 59s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0843ecebcab246b4b88f7ea407be0f19 : SUCCESS in 12m 32s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/467e7ebca28d467ab2bff824adb2862e : SUCCESS in 7m 06s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"68fbc3b9e1b22825788b7212f9b8c021e014c064","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-01-24 07:59:39.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"f4ff19f4accce092ae88dd3f6055934080787bcf","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2022-01-24 17:16:58.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"83affbc23f411ea477b89156e2a3821408083f94","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-24 17:18:35.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"de4c6292e3fc59de0159a36ffb99bd38baa31a00","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-24 17:37:10.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f0777c6d2d53496abde613b806ffbfa4 : SUCCESS in 4m 05s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/658aed9de90d4f7d882742610f6557f0 : SUCCESS in 4m 25s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/317b6ec668f1437fa4998c13c8b0eec9 : SUCCESS in 4m 27s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/56c97a509fa541058ed67c79506163da : SUCCESS in 12m 28s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/aa96b568b14d4178a63446885182f84b : SUCCESS in 13m 46s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8e7579630fe841c9b6fb02b1c741b576 : SUCCESS in 11m 46s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/58a14eb1a9b94be490d1ad881a7b1b4d : SUCCESS in 12m 10s","accounts_in_message":[],"_revision_number":2},{"id":"d06ba754a404d4f546a456ec89af7afbe76ae802","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-24 17:37:11.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"2a28f521ecfe54113d20015a1d22ba33648309b0","revisions":{"abccfa4c7378c5bf1e95bbb13f38255acef0fd75":{"kind":"REWORK","_number":1,"created":"2022-01-24 06:24:41.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/15/826015/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-trove","ref":"refs/changes/15/826015/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/15/826015/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/15/826015/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/15/826015/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-trove refs/changes/15/826015/1"}}},"commit":{"parents":[{"commit":"caab4a92864bb5538028049db99415a1786cd835","subject":"Merge \"Fix ignored auth_url parameters\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/caab4a92864bb5538028049db99415a1786cd835"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-24 06:24:12.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-24 06:24:25.000000000","tz":540},"subject":"Do not test detail of keystone::auth::service_identity","message":"Do not test detail of keystone::auth::service_identity\n\n... but test interfaces of that resource type. This helps us avoid\ndirect breakage caused by any change in puppet-keystone.\n\nChange-Id: I3ace0192837442a4d3952cc7b75ed2e93426effa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/abccfa4c7378c5bf1e95bbb13f38255acef0fd75"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/abccfa4c7378c5bf1e95bbb13f38255acef0fd75"}]},"branch":"refs/heads/master"},"2a28f521ecfe54113d20015a1d22ba33648309b0":{"kind":"REWORK","_number":2,"created":"2022-01-24 06:25:09.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/15/826015/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-trove","ref":"refs/changes/15/826015/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/15/826015/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/15/826015/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/15/826015/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-trove refs/changes/15/826015/2"}}},"commit":{"parents":[{"commit":"caab4a92864bb5538028049db99415a1786cd835","subject":"Merge \"Fix ignored auth_url parameters\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/caab4a92864bb5538028049db99415a1786cd835"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-24 06:24:12.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-24 06:25:00.000000000","tz":540},"subject":"Do not test detail of keystone::auth::service_identity","message":"Do not test detail of keystone::auth::service_identity\n\n... but test interfaces of that resource type. This helps us avoid\ndirect breakage caused by any change in puppet-keystone.\n\nChange-Id: I3ace0192837442a4d3952cc7b75ed2e93426effa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/2a28f521ecfe54113d20015a1d22ba33648309b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/2a28f521ecfe54113d20015a1d22ba33648309b0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
