)]}'
{"id":"openstack%2Fpuppet-trove~826231","triplet_id":"openstack%2Fpuppet-trove~master~I6907dd4b41dfe009a69fecd3ee5d8332c4c6a424","project":"openstack/puppet-trove","branch":"master","topic":"system-scope-params","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-01-26 14:49:28.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I6907dd4b41dfe009a69fecd3ee5d8332c4c6a424","subject":"Accept system scope credentials for Keystone API request","status":"MERGED","created":"2022-01-25 01:44:22.000000000","updated":"2022-01-26 14:51:56.000000000","submitted":"2022-01-26 14:49:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"826231-system-scope-params","meta_rev_id":"21a8f38d9f904d6bcd1d9b32145e9bc06abdbfa6","_number":826231,"virtual_id_number":826231,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-26 14:49:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2022-01-25 16:05:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2022-01-26 13:33:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2022-01-26 13:33:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-25 03:14:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-25 16:05:04.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2022-01-26 13:33:33.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"c749a9be092f0f3f00231e3b3b62fc1dc0f78c09","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-01-25 01:44:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"32d0a98bcc0b8aae74479de8d6eb7a481c31bcb9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-01-25 01:54:23.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e8b6775b5500a1b17891baad3a1a30f0c222c681","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-25 03:14:28.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/60800a76a51a483eb2700408ed6d8353 : SUCCESS in 3m 42s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/32beb6f0c1024d08b72dce9191e59139 : SUCCESS in 4m 17s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dd3c036c260f456b8a4612db867a7c54 : SUCCESS in 3m 05s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/2e44cd32316948b18bc74b563e9feb53 : SUCCESS in 13m 45s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/769f34470f8a432ea9ccdbe59ec25339 : SUCCESS in 11m 27s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d21e62616df746ea942b0e26e93da073 : SUCCESS in 8m 35s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/41ef35c0cac04e4ca803da8822dafd5f : SUCCESS in 11m 39s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c979934e0edc4d5eab4598596b67b498 : SUCCESS in 9m 10s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6e5b2a1f60e84d0fadb9f0c71df10871 : FAILURE in 7m 31s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/4017e706b5fa46fabf1b42f44e63f7b9 : SUCCESS in 1h 16m 18s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/e824451031e0409d91e2a91246633ca8 : SUCCESS in 48m 43s (non-voting)\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3c903a48de874966b3a4db8e40bdacbf : SUCCESS in 13m 19s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/bf8b707881144705a6a61e8fa93ba6d8 : SUCCESS in 23m 09s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ee5f0418bfa440f08fdaef526dfb1d0d : SUCCESS in 15m 03s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f1e79f6d4cb44684819c5853c4fc8630 : FAILURE in 13m 29s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/518916e6ee414342aad188a559409a4e : SUCCESS in 4m 43s","accounts_in_message":[],"_revision_number":2},{"id":"9b7c1af349a09b7e107e20c9a15d7ee3ae033be5","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2022-01-25 16:05:04.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"aa8a1b7e7f2bfd5afd33400cf45a16f757df7871","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-01-26 13:33:33.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"35ca838eb2a253f78a0b01081facbb92192c82fe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-26 13:34:47.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"4d2f153b4fa52fbcd48d5c6869303c980006b682","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-26 14:49:28.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/576ab59a7c5b4773800517e9ccb562b1 : SUCCESS in 8m 57s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2ae86e3e066b4f6386c33708903bbc1a : SUCCESS in 3m 48s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/92b8e6f7140844d5a1b0c647c882e542 : SUCCESS in 3m 37s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/11e9cb3eac8343bea1ec4c7b6b8c5706 : SUCCESS in 11m 57s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/8afb1f9d6a8f48eabe7c6ae71c31f1e9 : SUCCESS in 11m 35s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3e7789ef6dd64fef9b5f5e6f4b327fea : SUCCESS in 9m 22s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/11fbc04e50a347e8b7b97e3d4f82fbbb : SUCCESS in 12m 08s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/28296f818ca643c9b6e0c1d67849dfa3 : SUCCESS in 1h 13m 56s\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/4c22fbf1a1e745cb996c127f28f7fadb : SUCCESS in 20m 20s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f285653933ba480a89cf939ead5f5bae : SUCCESS in 15m 23s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/53158fa774e842d3a51c9befce09820f : SUCCESS in 3m 50s","accounts_in_message":[],"_revision_number":2},{"id":"ca4c617e4639452400a464df5b5f791c861ebfcc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-26 14:49:28.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"21a8f38d9f904d6bcd1d9b32145e9bc06abdbfa6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-26 14:51:56.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b621adcbb286400e98b3ff08414ec221 : SUCCESS in 1m 29s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"95f516939391a0e0747348cf391c496103c39497","revisions":{"cdc2ef42b47efc6f94e81ff6ca1ba8807b35cabc":{"kind":"REWORK","_number":1,"created":"2022-01-25 01:44:22.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/31/826231/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-trove","ref":"refs/changes/31/826231/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/31/826231/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/31/826231/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/31/826231/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-trove refs/changes/31/826231/1"}}},"commit":{"parents":[{"commit":"2eb937c16cf63fc7faed236f6ed131c2aaf392ea","subject":"Merge \"Add support for [keystone_authtoken] service_type\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/2eb937c16cf63fc7faed236f6ed131c2aaf392ea"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-25 01:43:40.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-25 01:43:40.000000000","tz":540},"subject":"Accept system scope credentials for Keystone API request","message":"Accept system scope credentials for Keystone API request\n\nThis change is the first step to support secure RBAC and allows usage\nof system scope credentials for Keystone API request.\n\nThis change covers the following two items.\n - assignment of system scope roles to system user\n - credential parameters for authtoken middleware\n\nDepends-on: https://review.opendev.org/804325\nChange-Id: I6907dd4b41dfe009a69fecd3ee5d8332c4c6a424\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/cdc2ef42b47efc6f94e81ff6ca1ba8807b35cabc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/cdc2ef42b47efc6f94e81ff6ca1ba8807b35cabc"}]},"branch":"refs/heads/master"},"95f516939391a0e0747348cf391c496103c39497":{"kind":"REWORK","_number":2,"created":"2022-01-25 01:54:23.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/31/826231/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-trove","ref":"refs/changes/31/826231/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/31/826231/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/31/826231/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/31/826231/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-trove refs/changes/31/826231/2"}}},"commit":{"parents":[{"commit":"2eb937c16cf63fc7faed236f6ed131c2aaf392ea","subject":"Merge \"Add support for [keystone_authtoken] service_type\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/2eb937c16cf63fc7faed236f6ed131c2aaf392ea"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-25 01:43:40.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-25 01:54:14.000000000","tz":540},"subject":"Accept system scope credentials for Keystone API request","message":"Accept system scope credentials for Keystone API request\n\nThis change is the first step to support secure RBAC and allows usage\nof system scope credentials for Keystone API request.\n\nThis change covers the following two items.\n - assignment of system scope roles to system user\n - credential parameters for authtoken middleware\n\nDepends-on: https://review.opendev.org/804325\nChange-Id: I6907dd4b41dfe009a69fecd3ee5d8332c4c6a424\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/95f516939391a0e0747348cf391c496103c39497"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/95f516939391a0e0747348cf391c496103c39497"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
