)]}'
{"id":"openstack%2Fpuppet-trove~828959","triplet_id":"openstack%2Fpuppet-trove~master~I372a41b5fc6c6632a0a1a615f23d69ab6203dfac","project":"openstack/puppet-trove","branch":"master","topic":"paste-ini-sep","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-02-14 19:22:10.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I372a41b5fc6c6632a0a1a615f23d69ab6203dfac","subject":"Allow customizing separator for api-paste.ini","status":"MERGED","created":"2022-02-12 16:38:05.000000000","updated":"2022-02-14 19:24:03.000000000","submitted":"2022-02-14 19:22:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"828959-paste-ini-sep","meta_rev_id":"3f3772451ee7ba6d3d5682d10f6ecd3aa1405e75","_number":828959,"virtual_id_number":828959,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-14 19:22:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2022-02-14 09:03:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2022-02-14 18:00:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2022-02-14 18:00:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-12 18:21:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-14 09:03:00.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2022-02-14 18:00:12.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"eac3e126cabc31aa51da31191852b884606b89b5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-02-12 16:38:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"22f1d5580602aa9a385e2d89fefd602174464b0b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-12 18:21:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/878fa98efd3642a788a9ba8b103b9350 : SUCCESS in 3m 30s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/784f9cc8f6ab4f82895a8d86ea36efa1 : SUCCESS in 3m 02s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fb10c806e9a7445aba3f17e6160aecb3 : SUCCESS in 3m 16s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/d9593c68e6e24b8b95318a721180465f : SUCCESS in 10m 27s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/3f74df05c4564fb487d00e94715da855 : SUCCESS in 11m 02s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/365c2e672ec340a19561d88806b26594 : SUCCESS in 7m 48s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4a3e7b8e6bd74c8e9d89f68b896bf803 : SUCCESS in 7m 38s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/491c150ac9034a6db571fcd33285c1f0 : SUCCESS in 8m 36s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1f23628907864ce88b7e0b6c46fdeba8 : FAILURE in 30m 59s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/1d392fea6d3542788907f4cb4ab31786 : SUCCESS in 1h 36m 27s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/7f195c9e04c24b4aba246debc952355a : SUCCESS in 1h 08m 54s (non-voting)\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9e44ba971dd24a799d377d665854781f : SUCCESS in 15m 06s (non-voting)\n- puppet-openstack-litmus-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/e3d16e28d23c402183dc3ad2ad8cb57c : SUCCESS in 19m 16s\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/aaf93ab170574da0aece1366853045da : FAILURE in 12m 16s (non-voting)\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ebcb8ddcc2dc4554a7b0e8355111db08 : SUCCESS in 16m 18s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4a6f9c2e2ec644d9a29ba68740c55284 : SUCCESS in 15m 26s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e9cdb353c31549d08e537c379710024f : SUCCESS in 4m 07s","accounts_in_message":[],"_revision_number":1},{"id":"d4192178d6ed71a4722b279c9ba7cf19a2be3bb4","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-02-14 09:03:00.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"dc67aa9872a9bed1865a8661031ddfc096ebdc9f","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2022-02-14 18:00:12.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2443d01b2a45bf42a23d42dc00b100507c6c89e1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-14 18:01:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4d7382416f95b0351a0d294ce6f16dc9abc3f85f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-14 19:22:10.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2272fea271a246a7a598ca672fea046d : SUCCESS in 4m 16s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4fa6fd5c02994b1b9a83ca7d30b2dbca : SUCCESS in 4m 22s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b46c9161131444a8914de8e81bbf18f4 : SUCCESS in 4m 07s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/cb6750c2a52143ada32269db4ae11d85 : SUCCESS in 13m 38s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/fba679cd17a746a9bf0a6a18aa916aea : SUCCESS in 14m 19s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0d7a128a24f34b2dbb4c3b7e82cde1fc : SUCCESS in 11m 15s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f6f9b89c213043a8bc60bfbfe7021299 : SUCCESS in 12m 41s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/c689c1789b924594b1f198a3cd9bc7cb : SUCCESS in 1h 16m 26s\n- puppet-openstack-litmus-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/2b7360cf05b14850b56dcafa7c492d17 : SUCCESS in 22m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ce6cc88e8f2a4814ab60afd9c2b63fd8 : SUCCESS in 4m 30s","accounts_in_message":[],"_revision_number":1},{"id":"1c45be3fdf32db3ffb56973e17533b51cef7763d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-14 19:22:10.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"3f3772451ee7ba6d3d5682d10f6ecd3aa1405e75","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-14 19:24:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c6af185aee674ebbb72ed96fce4bd3fd : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"538f6d5237c5a47b27a3f969891f3f70206803f8","revisions":{"538f6d5237c5a47b27a3f969891f3f70206803f8":{"kind":"REWORK","_number":1,"created":"2022-02-12 16:38:05.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/59/828959/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-trove","ref":"refs/changes/59/828959/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/59/828959/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/59/828959/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/59/828959/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-trove refs/changes/59/828959/1"}}},"commit":{"parents":[{"commit":"69eb2582c16733f1d796af284cc9c326b6d2c6af","subject":"Simplify definition to ensure keystone resource creation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/69eb2582c16733f1d796af284cc9c326b6d2c6af"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-12 16:37:56.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-12 16:37:56.000000000","tz":540},"subject":"Allow customizing separator for api-paste.ini","message":"Allow customizing separator for api-paste.ini\n\nThe api-paste.ini accepts not only \"\u003d\" but also \":\" and some services\nlike Barbican have been using \":\" for their default api-paste.ini\nfiles.\n\n[composite:main]\nuse \u003d egg:Paste#urlmap\n/: barbican_version\n/v1: barbican-api-keystone\n\nThis change allows users to use \":\" so that they can update the ini\nfiles with keeping it consistent with the default fules\n\nDepends-on: https://review.opendev.org/813614\nChange-Id: I372a41b5fc6c6632a0a1a615f23d69ab6203dfac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/538f6d5237c5a47b27a3f969891f3f70206803f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/538f6d5237c5a47b27a3f969891f3f70206803f8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
