)]}'
{"id":"openstack%2Fpuppet-trove~829325","triplet_id":"openstack%2Fpuppet-trove~master~Ia8327fbc6a678b2c3b6eddcbf55055c4fb8e2709","project":"openstack/puppet-trove","branch":"master","topic":"c9-ut","hashtags":[],"change_id":"Ia8327fbc6a678b2c3b6eddcbf55055c4fb8e2709","subject":"DNM: Test unit tests for CentOS 9","status":"ABANDONED","created":"2022-02-15 15:34:14.000000000","updated":"2022-02-15 16:46:04.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"777076f6376d93bed7ee9f6756d524e83d050c70","_number":829325,"virtual_id_number":829325,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2022-02-15 16:43:45.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-15 16:43:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"df0884680749b4ee012644bcfc1899855d6d4ed5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-02-15 15:34:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e5146831fbddee3b18c198e9d77660fa8192196b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-15 16:43:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2923fc7d31c94da3a71b3d33a54ab22e : SUCCESS in 4m 34s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a53a08c5f3cc4b06a927e1ac5502971d : SUCCESS in 4m 32s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8028e9ec44d44abfb25bb1b0872aa153 : SUCCESS in 4m 30s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/9bcfafefd273493380eb4dcf9d6e83fd : SUCCESS in 13m 20s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/5298c6efe9cf45109f80bb4d5063485f : SUCCESS in 14m 08s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/22214c2b00444741b70eb0aa72fea5cb : SUCCESS in 20m 34s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0e07e7c7fc914429ac6110a58a9c7bff : SUCCESS in 24m 48s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6d1df501199143ee8aa8469b9a1be5e8 : SUCCESS in 18m 34s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"777076f6376d93bed7ee9f6756d524e83d050c70","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-02-15 16:46:04.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6cf26a505627d2161b31791bdf2828910a07c6f8","revisions":{"6cf26a505627d2161b31791bdf2828910a07c6f8":{"kind":"REWORK","_number":1,"created":"2022-02-15 15:34:14.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/25/829325/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-trove","ref":"refs/changes/25/829325/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/25/829325/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/25/829325/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/25/829325/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-trove refs/changes/25/829325/1"}}},"commit":{"parents":[{"commit":"d07a47d2cbc51456a89936c3751013297fac96e9","subject":"CentOS 9: Disable unit tests dependent on puppet-postgresql","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/d07a47d2cbc51456a89936c3751013297fac96e9"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-15 15:33:58.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-15 15:33:58.000000000","tz":540},"subject":"DNM: Test unit tests for CentOS 9","message":"DNM: Test unit tests for CentOS 9\n\nDepends-on: https://review.opendev.org/c/openstack/puppet-openstack_spec_helper/+/822769\nChange-Id: Ia8327fbc6a678b2c3b6eddcbf55055c4fb8e2709\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/6cf26a505627d2161b31791bdf2828910a07c6f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/6cf26a505627d2161b31791bdf2828910a07c6f8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
