)]}'
{"id":"openstack%2Fpuppet-trove~830226","triplet_id":"openstack%2Fpuppet-trove~master~Ib848ce76f842255242ef099c0fc04b06ec5e1706","project":"openstack/puppet-trove","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-03-03 11:50:55.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib848ce76f842255242ef099c0fc04b06ec5e1706","subject":"Use openstack CLI instad of trove CLI","status":"MERGED","created":"2022-02-21 16:05:36.000000000","updated":"2022-03-03 11:50:55.000000000","submitted":"2022-03-03 11:50:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"830226","meta_rev_id":"a537d7b0ee17e507213ea9651bb9eed2510f29ff","_number":830226,"virtual_id_number":830226,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-03 11:50:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2022-02-22 20:47:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2022-03-03 10:36:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2022-03-03 10:36:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-21 17:30:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-22 20:47:34.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2022-03-03 10:36:53.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"ccfe3f22c03db313bfab9fb3beb86dac450e5ef2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-02-21 16:05:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2aba4dff550b67141aeb44f8a6abbf626f07ec5d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-21 17:30:32.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d4cc353a487f408cb7d2d80a1cabbbd8 : SUCCESS in 4m 37s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/62a1975ab99e4ec4b0b193f609f7aa64 : SUCCESS in 3m 48s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/61c89940742541428b64ebaee2cfe485 : SUCCESS in 3m 35s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/9dda0a8463e941d295d67079e73741e9 : FAILURE in 14m 52s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/d4326d2ea21c438cac53ec443ce775cf : FAILURE in 13m 23s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/36039b568b2d41ea97c1012aac2294dd : FAILURE in 11m 55s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/be180b2cdf85499482715439dc571ad2 : FAILURE in 9m 50s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/467cacee0fa847bc9a40f16d8197c2e9 : FAILURE in 10m 11s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/abe48de041af4c878836cf2dd824b80b : FAILURE in 36m 30s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/2269cb5924e3411a94a3b49ef70b0a0c : SUCCESS in 1h 16m 55s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/ef99ef477b894e49a3c8c7236c678669 : SUCCESS in 59m 37s (non-voting)\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/906f3018397b4199a37987e6790679e8 : SUCCESS in 14m 18s (non-voting)\n- puppet-openstack-litmus-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/40c29a4182674d3bb45e64e1135949e8 : SUCCESS in 22m 35s\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/d8be2bdf129a4f3ab53b49e5380959f1 : FAILURE in 8m 53s (non-voting)\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2f09e6ff5f7e4b43850e20b809703ae4 : SUCCESS in 15m 29s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ce766715b0ef4bf6af997293ad197e7c : SUCCESS in 16m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f0b7aa1efdc9a82c5aaa7fc9d68319a08ee80941","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-02-21 17:50:03.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ff8d44b8f607dd0257be72a753fad13c7997bc27","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-21 19:18:45.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6c1c3143da934c10857b7e467540f158 : SUCCESS in 4m 07s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e2e54373a6ba4e6f8283f83fcaa62287 : SUCCESS in 3m 28s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2d05a886cda347f3ba4dd60d4dcc13f1 : SUCCESS in 3m 33s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/476f21ad60a04baca7ac8a2cb816878a : FAILURE in 12m 44s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/0eb7ce881adc4234aced0e8d2b91c9b0 : FAILURE in 12m 38s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d65842821f0b455c87922429cc61f307 : FAILURE in 10m 35s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a13be9fedfc94795b2541c9e6aa053a4 : FAILURE in 10m 14s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4eb4b3234c164e69915ca12ec8356ebb : FAILURE in 10m 06s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/39c5f038dbd5429cb174d062c040baf2 : FAILURE in 32m 20s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/04204c1e21e649b3bc68582a43b61e40 : SUCCESS in 1h 22m 11s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/b9f74de0aad04a4c924f1df4c32f129b : SUCCESS in 1h 04m 42s (non-voting)\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/193200f055ba43bda638a9b4acf449af : SUCCESS in 17m 40s (non-voting)\n- puppet-openstack-litmus-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/886e0176c5034c3f8452ee977a2f4a83 : SUCCESS in 20m 15s\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/d17a7926d3e74352830700456ed22610 : FAILURE in 18m 43s (non-voting)\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/108b16b5215b4627856bd080609921fc : SUCCESS in 15m 29s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ab7f12c0d8994a59a37d723744bfe972 : SUCCESS in 14m 52s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d283d66f878f15fb44c9ca1b8fa1b81aaf1580b4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-02-22 10:43:53.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"f34139a575d283abcf29c9ecab6e1f08078809fe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 12:02:45.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4a2dd57d8a554d8eba2e1f2e05a7e9af : SUCCESS in 3m 23s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/514ea828c6f64b71ac9f4cf32ef3b3d7 : SUCCESS in 3m 27s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/29a24dbd526540989f7085c2293127c3 : SUCCESS in 3m 25s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/a980cd82a7d34a6d9e3f37a643789676 : FAILURE in 13m 29s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/bd50a9d66fe64de381ffafefc7b2344c : FAILURE in 11m 42s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c8e4bde5ea804682929996a28bde23fa : FAILURE in 9m 04s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3d4d0f563dcc4d9ba94e9e92513bf663 : FAILURE in 9m 39s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1d0167c9d64d498e99020f7ba53de8d9 : FAILURE in 8m 53s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/acd5ada7fbd948f79567800e1b248ca7 : FAILURE in 34m 17s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/9c58e7391ac94bd688fddcbb994b5dd0 : SUCCESS in 1h 16m 27s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/5a7557f94498463092fb9dbc890db22e : SUCCESS in 1h 10m 07s (non-voting)\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4805c7ed7bc24da49a07df516ba525d5 : SUCCESS in 15m 52s (non-voting)\n- puppet-openstack-litmus-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/58ec2097e19f497984a03b2ee979c9b6 : SUCCESS in 25m 38s\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/345e3145831e476eb124b9996e1dcf2e : FAILURE in 10m 01s (non-voting)\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a750ba2fabf542adaed0391032855327 : SUCCESS in 16m 27s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1311d1718c934b619a0051ccf4d8963f : SUCCESS in 18m 25s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"3f6f9379124a91d00583234435d54f6006aef46a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-02-22 12:19:39.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"d85c5d30ebf43d51f170325aa6bc6debaee5d85b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 13:48:43.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bf3430d07b7e4a4eb9ee1b3bf6969644 : SUCCESS in 4m 32s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/89472fa1b32f45458009d81434faca36 : SUCCESS in 3m 51s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/acec0a109f524336bbfd4d1afe838a7a : SUCCESS in 4m 07s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/1759857b1cd045a187e1b0e6907e1841 : FAILURE in 14m 00s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/3a074317856c4f19b8a28d007fc9cd20 : FAILURE in 12m 53s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/545976e15270462a9be35bd1fb9af7a7 : FAILURE in 10m 54s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4e82c72804994bbea035036f5391105c : FAILURE in 11m 31s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ce53914e32644b12b958cba0fad04ef3 : FAILURE in 11m 12s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/137f55f3771b438a8aea07632e769e35 : FAILURE in 37m 55s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/d1ef058c632e4e00bf0d6c0163d11713 : SUCCESS in 1h 24m 53s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/86f4c97520a54273b8739b0ef062717e : SUCCESS in 1h 11m 52s (non-voting)\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/be8e60d6cabd430998ecf9686cd7d9e5 : SUCCESS in 17m 43s (non-voting)\n- puppet-openstack-litmus-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/b87ba360ea584ac2af7df3692c52383b : SUCCESS in 23m 00s\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/102fbde138b84c62825162b82260d87a : FAILURE in 8m 27s (non-voting)\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0d0dbd2d5f65460788e87c46dad8cb91 : SUCCESS in 18m 08s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a53974cbbc2c46799f448810bb9c5532 : SUCCESS in 18m 04s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"e8e4ccb1c5b7f0cf85a418af78f34c0a5a3b232b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-02-22 14:06:34.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"d993d583d6308ec457e791607f3ba47f2d0f3a11","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-02-22 14:41:13.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"f731db3eaa78e8079bf62980d28a245ef508e7fa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-02-22 14:43:08.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"e4cb91fac50ddd177ecc98b42238c913583ca841","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-02-22 15:32:09.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"49e592a7e0df4d45e67001450c9126aa3236b018","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 17:00:44.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/95a04c26ec1c4d2a9d3a33ddeb035d58 : SUCCESS in 4m 19s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/480e88422d7c481e80455d972d05fd30 : SUCCESS in 5m 13s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4b90097d697947a28c4d719b8f2a8eb2 : SUCCESS in 4m 23s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/65f7cdfeb4c94e799e0a7e497e89139b : SUCCESS in 13m 50s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/1375f318e5364e5e85d9c410f0781949 : SUCCESS in 16m 31s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ea1cb9138ef648d488769942d21933ca : SUCCESS in 13m 10s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/68415e38f49948e8b57d42c5babf8e3c : SUCCESS in 13m 48s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f4a9db010b0e4727b470938986471b10 : SUCCESS in 13m 27s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c61f82cb43d44b1e88ee51efc33d8184 : FAILURE in 36m 55s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/9f4ec285da0e4810b62b9eb8c86b326e : SUCCESS in 1h 19m 22s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/9f2053ece4e4446d85f7d956568630ab : SUCCESS in 59m 47s (non-voting)\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/71a89df022ee4350b2cb7e0726cfd1e7 : SUCCESS in 17m 09s (non-voting)\n- puppet-openstack-litmus-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/ffa8cc424f13446ba7a2c80450aa0813 : SUCCESS in 25m 15s\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/dadd69a9b44b4a4ea35be1607a29d580 : FAILURE in 9m 27s (non-voting)\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/afbc09d3c7c54caba92190bd1aad48b5 : SUCCESS in 14m 38s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1a6406444de744bcb12481eff860933d : SUCCESS in 15m 35s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"28193e2a28efe94dc324dadcef9922df569b398a","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2022-02-22 20:47:34.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"94e8ec8b2183c8bd7a12d937567ca1d8fd2a171c","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-03-03 10:36:53.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"1b64fadc8c2427f8da1a2a4f937c92f2fb6b5a5c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-03 10:42:47.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"6bf6e4baab91871c094de7022d459b238e953430","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-03 11:50:54.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a36e59f3f15448038c9758419ecebce7 : SUCCESS in 5m 01s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b3dff823e52c47b6b2fea72bdc735f01 : SUCCESS in 4m 45s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/afd0aeaa597745b8a6ff995cdaabf19d : SUCCESS in 5m 32s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/a44ac0b269ab46159b9c298dec9b0cd8 : SUCCESS in 14m 29s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/a9cbfbc67a2e455284fa645e69b54401 : SUCCESS in 14m 36s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/88a17ccff7914d94b4e1c60cdfde9622 : SUCCESS in 10m 34s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/75becd4278644e94b5ec996c844a4799 : SUCCESS in 10m 57s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/3cb203a953f942eda6335bd7e73756ea : SUCCESS in 1h 07m 00s\n- puppet-openstack-litmus-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/762fc3698e6340ce983cad824dba524e : SUCCESS in 21m 54s","accounts_in_message":[],"_revision_number":8},{"id":"a537d7b0ee17e507213ea9651bb9eed2510f29ff","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-03 11:50:55.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"93551f863cf13777dab94c2123fb04fa22a45758","revisions":{"5fc6dde325e7795e86e68bf503d991a527ee83fe":{"kind":"REWORK","_number":1,"created":"2022-02-21 16:05:36.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/26/830226/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-trove","ref":"refs/changes/26/830226/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/26/830226/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/26/830226/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/26/830226/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-trove refs/changes/26/830226/1"}}},"commit":{"parents":[{"commit":"538f6d5237c5a47b27a3f969891f3f70206803f8","subject":"Allow customizing separator for api-paste.ini","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/538f6d5237c5a47b27a3f969891f3f70206803f8"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-21 16:05:15.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-21 16:05:15.000000000","tz":540},"subject":"WIP: Use openstack CLI instad of trove CLI","message":"WIP: Use openstack CLI instad of trove CLI\n\nChange-Id: Ib848ce76f842255242ef099c0fc04b06ec5e1706\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/5fc6dde325e7795e86e68bf503d991a527ee83fe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/5fc6dde325e7795e86e68bf503d991a527ee83fe"}]},"branch":"refs/heads/master"},"9bec0806284dd454ae81cbb01ae005413818b228":{"kind":"REWORK","_number":2,"created":"2022-02-21 17:50:03.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/26/830226/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-trove","ref":"refs/changes/26/830226/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/26/830226/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/26/830226/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/26/830226/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-trove refs/changes/26/830226/2"}}},"commit":{"parents":[{"commit":"538f6d5237c5a47b27a3f969891f3f70206803f8","subject":"Allow customizing separator for api-paste.ini","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/538f6d5237c5a47b27a3f969891f3f70206803f8"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-21 16:05:15.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-21 17:49:54.000000000","tz":540},"subject":"WIP: Use openstack CLI instad of trove CLI","message":"WIP: Use openstack CLI instad of trove CLI\n\nChange-Id: Ib848ce76f842255242ef099c0fc04b06ec5e1706\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/9bec0806284dd454ae81cbb01ae005413818b228"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/9bec0806284dd454ae81cbb01ae005413818b228"}]},"branch":"refs/heads/master"},"ff8f80ac41fc2adaf7b24ebef5a02a2c07b2aec4":{"kind":"REWORK","_number":3,"created":"2022-02-22 10:43:53.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/26/830226/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-trove","ref":"refs/changes/26/830226/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/26/830226/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/26/830226/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/26/830226/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-trove refs/changes/26/830226/3"}}},"commit":{"parents":[{"commit":"538f6d5237c5a47b27a3f969891f3f70206803f8","subject":"Allow customizing separator for api-paste.ini","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/538f6d5237c5a47b27a3f969891f3f70206803f8"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-21 16:05:15.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-22 10:42:02.000000000","tz":540},"subject":"Use openstack CLI instad of trove CLI","message":"Use openstack CLI instad of trove CLI\n\nThis change replaces trove CLI by openstack CLI, so that the providers\ncan leverage clouds.yaml generated by puppet-keystone.\n\nChange-Id: Ib848ce76f842255242ef099c0fc04b06ec5e1706\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/ff8f80ac41fc2adaf7b24ebef5a02a2c07b2aec4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/ff8f80ac41fc2adaf7b24ebef5a02a2c07b2aec4"}]},"branch":"refs/heads/master"},"6ebd604258cafad5574b44395c427e0a7f7ce764":{"kind":"REWORK","_number":4,"created":"2022-02-22 12:19:39.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/26/830226/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-trove","ref":"refs/changes/26/830226/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/26/830226/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/26/830226/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/26/830226/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-trove refs/changes/26/830226/4"}}},"commit":{"parents":[{"commit":"538f6d5237c5a47b27a3f969891f3f70206803f8","subject":"Allow customizing separator for api-paste.ini","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/538f6d5237c5a47b27a3f969891f3f70206803f8"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-21 16:05:15.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-22 12:19:29.000000000","tz":540},"subject":"Use openstack CLI instad of trove CLI","message":"Use openstack CLI instad of trove CLI\n\nThis change replaces trove CLI by openstack CLI, so that the providers\ncan leverage clouds.yaml generated by puppet-keystone.\n\nChange-Id: Ib848ce76f842255242ef099c0fc04b06ec5e1706\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/6ebd604258cafad5574b44395c427e0a7f7ce764"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/6ebd604258cafad5574b44395c427e0a7f7ce764"}]},"branch":"refs/heads/master"},"250d1579947c1d203dfacd30cb2e4704d1c38d98":{"kind":"REWORK","_number":5,"created":"2022-02-22 14:06:34.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/26/830226/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-trove","ref":"refs/changes/26/830226/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/26/830226/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/26/830226/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/26/830226/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-trove refs/changes/26/830226/5"}}},"commit":{"parents":[{"commit":"538f6d5237c5a47b27a3f969891f3f70206803f8","subject":"Allow customizing separator for api-paste.ini","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/538f6d5237c5a47b27a3f969891f3f70206803f8"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-21 16:05:15.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-22 14:06:24.000000000","tz":540},"subject":"Use openstack CLI instad of trove CLI","message":"Use openstack CLI instad of trove CLI\n\nThis change replaces trove CLI by openstack CLI, so that the providers\ncan leverage clouds.yaml generated by puppet-keystone.\n\nChange-Id: Ib848ce76f842255242ef099c0fc04b06ec5e1706\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/250d1579947c1d203dfacd30cb2e4704d1c38d98"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/250d1579947c1d203dfacd30cb2e4704d1c38d98"}]},"branch":"refs/heads/master"},"0d402454bc24595198775c2a30f6a8f73aa083a9":{"kind":"REWORK","_number":6,"created":"2022-02-22 14:41:13.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/26/830226/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-trove","ref":"refs/changes/26/830226/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/26/830226/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/26/830226/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/26/830226/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-trove refs/changes/26/830226/6"}}},"commit":{"parents":[{"commit":"538f6d5237c5a47b27a3f969891f3f70206803f8","subject":"Allow customizing separator for api-paste.ini","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/538f6d5237c5a47b27a3f969891f3f70206803f8"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-21 16:05:15.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-22 14:40:59.000000000","tz":540},"subject":"Use openstack CLI instad of trove CLI","message":"Use openstack CLI instad of trove CLI\n\nThis change replaces trove CLI by openstack CLI, so that the providers\ncan leverage clouds.yaml generated by puppet-keystone.\n\nChange-Id: Ib848ce76f842255242ef099c0fc04b06ec5e1706\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/0d402454bc24595198775c2a30f6a8f73aa083a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/0d402454bc24595198775c2a30f6a8f73aa083a9"}]},"branch":"refs/heads/master"},"33f4c31b81bbc26355e6ba9d8d0cc3697c059c5e":{"kind":"REWORK","_number":7,"created":"2022-02-22 14:43:08.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/26/830226/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-trove","ref":"refs/changes/26/830226/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/26/830226/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/26/830226/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/26/830226/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-trove refs/changes/26/830226/7"}}},"commit":{"parents":[{"commit":"538f6d5237c5a47b27a3f969891f3f70206803f8","subject":"Allow customizing separator for api-paste.ini","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/538f6d5237c5a47b27a3f969891f3f70206803f8"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-21 16:05:15.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-22 14:42:59.000000000","tz":540},"subject":"Use openstack CLI instad of trove CLI","message":"Use openstack CLI instad of trove CLI\n\nThis change replaces trove CLI by openstack CLI, so that the providers\ncan leverage clouds.yaml generated by puppet-keystone.\n\nChange-Id: Ib848ce76f842255242ef099c0fc04b06ec5e1706\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/33f4c31b81bbc26355e6ba9d8d0cc3697c059c5e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/33f4c31b81bbc26355e6ba9d8d0cc3697c059c5e"}]},"branch":"refs/heads/master"},"93551f863cf13777dab94c2123fb04fa22a45758":{"kind":"REWORK","_number":8,"created":"2022-02-22 15:32:09.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/26/830226/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-trove","ref":"refs/changes/26/830226/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/26/830226/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/26/830226/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/26/830226/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-trove refs/changes/26/830226/8"}}},"commit":{"parents":[{"commit":"538f6d5237c5a47b27a3f969891f3f70206803f8","subject":"Allow customizing separator for api-paste.ini","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/538f6d5237c5a47b27a3f969891f3f70206803f8"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-21 16:05:15.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-22 15:32:00.000000000","tz":540},"subject":"Use openstack CLI instad of trove CLI","message":"Use openstack CLI instad of trove CLI\n\nThis change replaces trove CLI by openstack CLI, so that the providers\ncan leverage clouds.yaml generated by puppet-keystone.\n\nChange-Id: Ib848ce76f842255242ef099c0fc04b06ec5e1706\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/93551f863cf13777dab94c2123fb04fa22a45758"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/93551f863cf13777dab94c2123fb04fa22a45758"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
