)]}'
{"id":"openstack%2Fpuppet-trove~830620","triplet_id":"openstack%2Fpuppet-trove~master~Icc23d46b18c0bdc0564219d3773f2a3b5d99a829","project":"openstack/puppet-trove","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-03-11 00:46:22.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Icc23d46b18c0bdc0564219d3773f2a3b5d99a829","subject":"Deprecate the use_neutron parameter","status":"MERGED","created":"2022-02-23 14:11:49.000000000","updated":"2022-03-11 00:48:45.000000000","submitted":"2022-03-11 00:46:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"830620","meta_rev_id":"2697ca9b62a3c0ab19dcab60bdf7dcbecf5b2505","_number":830620,"virtual_id_number":830620,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-11 00:46:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2022-03-10 23:39:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2022-03-11 00:48:45.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":1,"date":"2022-03-10 23:39:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-23 16:07:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-24 09:14:23.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2022-03-10 23:39:08.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"CC"},{"updated":"2022-03-10 23:39:30.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"18c191e35f675680e06a6f480fb2d3d9695b729e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-02-23 14:11:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"20b5c6f8a4e1261f78ece518ce57086686234bb6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-02-23 14:15:51.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"cb3774930721ce7833ae6aeaa6e971c028aea2f5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-02-23 14:19:44.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"903ae24b52c0ff4d58dc065ddaf928796392c5ea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-02-23 14:35:14.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"0428c6b4d96e8e06adf32d7f16a27529361caf1f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-02-23 14:36:02.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"26464b78bff647c3177a22b5eade2bc216c27ff5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-02-23 14:57:44.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"bd978d074f09b43ab928fab1e7aeebe35c735f99","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-23 16:07:24.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cde6e6f949ae4f23922d9f15a5f1cfa7 : SUCCESS in 3m 32s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4ed6c1a9af2642959a0495a0b6057e6f : SUCCESS in 4m 03s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/57b686d81fb542d3b596782bb432da8e : SUCCESS in 3m 52s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/0eb6c9451c664d8a879e6103200adda7 : SUCCESS in 11m 36s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/625866b583824278bd670442faedcef8 : SUCCESS in 13m 01s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/eb74582071d94724a8be436d4adbd8e5 : SUCCESS in 8m 45s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5bf296d038974728aac3ac70ce51359c : SUCCESS in 9m 46s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ba6b53cb50d94f66a50cc136730dcf11 : SUCCESS in 9m 32s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/16a5252127de4289b44b217a790b3351 : FAILURE in 33m 48s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/56b34b41216a431c8e3836a97d210d8e : SUCCESS in 1h 04m 37s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/e69b614476784aafbcbcbc2ce049f0fa : SUCCESS in 1h 07m 22s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/936ea59de0f243a388a5b7031fc4b55c : SUCCESS in 14m 27s (non-voting)\n- puppet-openstack-litmus-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/737b0a3a98a64c0cadb873cf4bade5de : SUCCESS in 24m 37s\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/b88e66a9d51848808dd6bc3092f871a7 : FAILURE in 10m 40s (non-voting)\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3402edb2be504727ad933064030b7513 : SUCCESS in 14m 41s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/badbd10fb1444d029f03df3483f9344b : SUCCESS in 17m 57s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4a3aef163b5a48d4b4641d0712c0557b : SUCCESS in 3m 40s","accounts_in_message":[],"_revision_number":6},{"id":"2c581ee07247b3f2aa9743e7a1c7ad53cfeed269","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-02-24 09:14:23.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"b71c399a4b149250a266d0cbe488d705cfd8e6ba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-03-07 23:57:24.000000000","message":"Patch Set 7: Patch Set 6 was rebased","accounts_in_message":[],"_revision_number":7},{"id":"e5a16f7b01303e090b1e7d3f3ed10f55b6dd8e13","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-08 01:14:39.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0da7186be80841e5a6f9cc5c30912f44 : SUCCESS in 4m 52s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0aea54c2051645ecbd645b38df4cb604 : SUCCESS in 4m 49s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4b9d61bd84e24b5c96df557009524f5d : SUCCESS in 4m 50s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/4e956dab3a914434a1b355e18e3bda46 : SUCCESS in 11m 28s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/c01d712cbc0649fd887f065fe6a1a88c : SUCCESS in 11m 40s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9f8b5cf6f0c74ac0a16445e6c5acd9ae : SUCCESS in 9m 33s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cca6f4dd5a274320a9d6780aa30d4738 : SUCCESS in 9m 16s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/95fc1b67fdcd492ca8dea72f49cd5194 : SUCCESS in 9m 19s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c8b5b864c8164410857ff3f135936b41 : FAILURE in 32m 12s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/e9c46cd8350a4e32908f9250cd592d71 : SUCCESS in 1h 05m 09s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a6aef38a5f8c49b1b9b71fc9a5e7d11b : SUCCESS in 57m 10s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fb1a063dd6b64e68bec45fb881b8d8fa : SUCCESS in 16m 15s (non-voting)\n- puppet-openstack-litmus-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/95e70a76fcb9428bb1390826d9bc5fe6 : SUCCESS in 20m 39s\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/96ca2823e1ec45d390d3689d003ecc8c : FAILURE in 12m 28s (non-voting)\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4cf71568739a40f78d267ac17d27a3d7 : SUCCESS in 15m 18s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c759fa80f43c47dfa6cf14593c313eaf : SUCCESS in 16m 11s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/51adb7dd17cf4ead86f7017802b502c0 : SUCCESS in 5m 02s","accounts_in_message":[],"_revision_number":7},{"id":"95719ed86165f2daf23d9ce8f1d7dab6d1eaf47b","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2022-03-10 23:39:30.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"b1663d904a419d43981dc94d35a10448b225ea1d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-10 23:40:46.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"0e1d00fd1ceddd02a22ec969904c6f250da6ecca","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-11 00:46:22.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7fc7611eb9764e74a8155c27a284ea01 : SUCCESS in 4m 50s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3c2203298a9a4eb2a9b4b1783f6915de : SUCCESS in 5m 15s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/61c58be807ca48399b86bc9c77b104ee : SUCCESS in 5m 19s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/4ff9c2869a324063a8f0f78d24ee7220 : SUCCESS in 11m 20s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/6694ce15ac6f444790fef447d7e89e54 : SUCCESS in 11m 36s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/896deed30aee433598dd415d88b305e5 : SUCCESS in 9m 21s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2877a6882e5c4db3a229bf26bc1a35fb : SUCCESS in 9m 56s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/5c0a7c87823143a5898c0db7e1e95653 : SUCCESS in 57m 59s\n- puppet-openstack-litmus-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/477056d5631247fd8b06502cc1521d5c : SUCCESS in 19m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5998bd0f5d8e4913a33ac38881dc7556 : SUCCESS in 5m 37s","accounts_in_message":[],"_revision_number":7},{"id":"afa56077bdad72c6b21222aa6c4d99b348e6f4f8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-11 00:46:22.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7},{"id":"640f58a86b873abfd03df12efdd47871585b7bd5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-11 00:48:45.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a362e51bfb1a44f4b7dba39e2d3b79cc : SUCCESS in 53s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"ddb0f14f95156c483c499fe95982d2e5777b236f","revisions":{"51ab6ecd2fc60e9db1698474399710a409e05a69":{"kind":"REWORK","_number":1,"created":"2022-02-23 14:11:49.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/20/830620/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-trove","ref":"refs/changes/20/830620/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/20/830620/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/20/830620/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/20/830620/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-trove refs/changes/20/830620/1"}}},"commit":{"parents":[{"commit":"ed106767a274b69db62b68729921d692db600d4c","subject":"Use single trove.conf for api, conductor and taskmanager","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/ed106767a274b69db62b68729921d692db600d4c"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-23 14:10:10.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-23 14:11:40.000000000","tz":540},"subject":"Deprecate the use_neutron parameter","message":"Deprecate the use_neutron parameter\n\n... because nova-network was removed a while ago and now neutron should\nbe used always.\n\nChange-Id: Icc23d46b18c0bdc0564219d3773f2a3b5d99a829\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/51ab6ecd2fc60e9db1698474399710a409e05a69"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/51ab6ecd2fc60e9db1698474399710a409e05a69"}]},"branch":"refs/heads/master"},"1c85707195d8c30f706de9c1becac9342c325622":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-02-23 14:15:51.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/20/830620/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-trove","ref":"refs/changes/20/830620/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/20/830620/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/20/830620/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/20/830620/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-trove refs/changes/20/830620/2"}}},"commit":{"parents":[{"commit":"5736b8c484a326d1131f48b3998271004e3133a2","subject":"Use single trove.conf for api, conductor and taskmanager","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/5736b8c484a326d1131f48b3998271004e3133a2"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-23 14:10:10.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-23 14:15:37.000000000","tz":540},"subject":"Deprecate the use_neutron parameter","message":"Deprecate the use_neutron parameter\n\n... because nova-network was removed a while ago and now neutron should\nbe used always.\n\nChange-Id: Icc23d46b18c0bdc0564219d3773f2a3b5d99a829\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/1c85707195d8c30f706de9c1becac9342c325622"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/1c85707195d8c30f706de9c1becac9342c325622"}]},"branch":"refs/heads/master"},"af06a6bd9a905597a4ecf5c23401e941ee060159":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2022-02-23 14:19:44.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/20/830620/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-trove","ref":"refs/changes/20/830620/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/20/830620/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/20/830620/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/20/830620/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-trove refs/changes/20/830620/3"}}},"commit":{"parents":[{"commit":"d0fda45f6659b394aade1f22f7c584fe6c5ff7d3","subject":"Use single trove.conf for api, conductor and taskmanager","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/d0fda45f6659b394aade1f22f7c584fe6c5ff7d3"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-23 14:10:10.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-23 14:19:15.000000000","tz":540},"subject":"Deprecate the use_neutron parameter","message":"Deprecate the use_neutron parameter\n\n... because nova-network was removed a while ago and now neutron should\nbe used always.\n\nChange-Id: Icc23d46b18c0bdc0564219d3773f2a3b5d99a829\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/af06a6bd9a905597a4ecf5c23401e941ee060159"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/af06a6bd9a905597a4ecf5c23401e941ee060159"}]},"branch":"refs/heads/master"},"3d95adbff32d3f505a66a32407f462d92808d363":{"kind":"REWORK","_number":4,"created":"2022-02-23 14:35:14.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/20/830620/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-trove","ref":"refs/changes/20/830620/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/20/830620/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/20/830620/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/20/830620/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-trove refs/changes/20/830620/4"}}},"commit":{"parents":[{"commit":"d0fda45f6659b394aade1f22f7c584fe6c5ff7d3","subject":"Use single trove.conf for api, conductor and taskmanager","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/d0fda45f6659b394aade1f22f7c584fe6c5ff7d3"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-23 14:10:10.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-23 14:35:00.000000000","tz":540},"subject":"Deprecate the use_neutron parameter","message":"Deprecate the use_neutron parameter\n\n... because nova-network was removed a while ago and now neutron should\nbe used always.\n\nChange-Id: Icc23d46b18c0bdc0564219d3773f2a3b5d99a829\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/3d95adbff32d3f505a66a32407f462d92808d363"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/3d95adbff32d3f505a66a32407f462d92808d363"}]},"branch":"refs/heads/master"},"c73a82eb020dbed9cd8deb41b339a9470cb84f06":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2022-02-23 14:36:02.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/20/830620/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-trove","ref":"refs/changes/20/830620/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/20/830620/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/20/830620/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/20/830620/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-trove refs/changes/20/830620/5"}}},"commit":{"parents":[{"commit":"16f8b807eb817dfc2210a8f54b2f1f79406761a5","subject":"Use single trove.conf for api, conductor and taskmanager","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/16f8b807eb817dfc2210a8f54b2f1f79406761a5"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-23 14:10:10.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-23 14:35:51.000000000","tz":540},"subject":"Deprecate the use_neutron parameter","message":"Deprecate the use_neutron parameter\n\n... because nova-network was removed a while ago and now neutron should\nbe used always.\n\nChange-Id: Icc23d46b18c0bdc0564219d3773f2a3b5d99a829\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/c73a82eb020dbed9cd8deb41b339a9470cb84f06"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/c73a82eb020dbed9cd8deb41b339a9470cb84f06"}]},"branch":"refs/heads/master"},"8e07e52ff85c01d479a7e86b1c5269c60de92f4e":{"kind":"REWORK","_number":6,"created":"2022-02-23 14:57:44.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/20/830620/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-trove","ref":"refs/changes/20/830620/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/20/830620/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/20/830620/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/20/830620/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-trove refs/changes/20/830620/6"}}},"commit":{"parents":[{"commit":"16f8b807eb817dfc2210a8f54b2f1f79406761a5","subject":"Use single trove.conf for api, conductor and taskmanager","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/16f8b807eb817dfc2210a8f54b2f1f79406761a5"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-23 14:10:10.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-23 14:57:35.000000000","tz":540},"subject":"Deprecate the use_neutron parameter","message":"Deprecate the use_neutron parameter\n\n... because nova-network was removed a while ago and now neutron should\nbe used always.\n\nChange-Id: Icc23d46b18c0bdc0564219d3773f2a3b5d99a829\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/8e07e52ff85c01d479a7e86b1c5269c60de92f4e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/8e07e52ff85c01d479a7e86b1c5269c60de92f4e"}]},"branch":"refs/heads/master"},"ddb0f14f95156c483c499fe95982d2e5777b236f":{"kind":"NO_CHANGE","_number":7,"created":"2022-03-07 23:57:24.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/20/830620/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-trove","ref":"refs/changes/20/830620/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/20/830620/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/20/830620/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/20/830620/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-trove refs/changes/20/830620/7"}}},"commit":{"parents":[{"commit":"f1c11c68d56e4f1ad5125fa0df04ab91d929cd33","subject":"Use single trove.conf for api, conductor and taskmanager","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/f1c11c68d56e4f1ad5125fa0df04ab91d929cd33"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-23 14:10:10.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-03-07 23:57:24.000000000","tz":0},"subject":"Deprecate the use_neutron parameter","message":"Deprecate the use_neutron parameter\n\n... because nova-network was removed a while ago and now neutron should\nbe used always.\n\nChange-Id: Icc23d46b18c0bdc0564219d3773f2a3b5d99a829\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/ddb0f14f95156c483c499fe95982d2e5777b236f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/ddb0f14f95156c483c499fe95982d2e5777b236f"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
