)]}'
{"id":"openstack%2Fpuppet-trove~834257","triplet_id":"openstack%2Fpuppet-trove~master~Iba0599ff4485668ace1b044e6a922b2dafec6d3b","project":"openstack/puppet-trove","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-03-18 21:50:08.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iba0599ff4485668ace1b044e6a922b2dafec6d3b","subject":"Avoid testing details of oslo::messaging(::*)","status":"MERGED","created":"2022-03-18 03:47:22.000000000","updated":"2022-03-18 21:50:08.000000000","submitted":"2022-03-18 21:50:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"834257","meta_rev_id":"3d998f499d8bdc2ea06ff473fa316566caec317c","_number":834257,"virtual_id_number":834257,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-18 21:50:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2022-03-18 08:11:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2022-03-18 21:26:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"date":"2022-03-18 08:11:32.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2022-03-18 21:26:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-18 04:05:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-18 08:11:28.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2022-03-18 21:26:01.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"12deb6d495ebc092ae707187ca5301797d4281b5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-03-18 03:47:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"655836511503874557b232b3414f852ec647a6d6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-18 04:05:47.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/572b44e357b94b0bb26c9cd38a8efc2e : SUCCESS in 5m 34s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/311ddee1469147dea21c549921221533 : SUCCESS in 5m 22s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/39b992f7156e40fb8ccda97606f6f717 : SUCCESS in 5m 42s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/245d4bcd05f441538dc5e2cd7066cd50 : FAILURE in 15m 03s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/17e1bc84e13741479a3074de5f3befc8 : FAILURE in 14m 55s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d3968257a89a4b189074b1e18a1af960 : FAILURE in 11m 50s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5b1891010ee843439efcb9deff35db8a : FAILURE in 11m 36s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/39861c5a9ccc4e38a31d8b64bc4f0fcb : FAILURE in 11m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bcdba713358202e274f079e86755629a0152a6b4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-03-18 04:30:46.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"52f0b8c76af77d777f3d8ae54bec0fc65b3ee130","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-18 04:44:33.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a29c10567de249668642995bb0307408 : SUCCESS in 4m 47s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5d3d9723cfd042338c0375c8b24a5f94 : SUCCESS in 4m 37s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a6324ffbf268470e9cb9a785d3a2d894 : SUCCESS in 4m 43s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/76c9b229d2274f5da2332017e15a4622 : FAILURE in 12m 04s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/9e7418ecb98d4e25a0d2c520d14ad645 : FAILURE in 11m 48s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/39f95bf1d5dc47c0ac400f063531ee4d : FAILURE in 9m 23s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5f2ece237777421fa1a62737479c7881 : FAILURE in 9m 25s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f5b784b5d18d44f5a4f0797319730f1b : FAILURE in 9m 23s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b75fcfa0265fbc3661622ff2f51bf7b9f3d5f7ee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-03-18 04:50:17.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"c218192fb88e9680ae8d91b8829064006b90d476","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-18 05:06:31.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/78ab128e213d4f9cad5cfd05b59741c6 : SUCCESS in 4m 58s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/88a095078bd54c04a42281c3621e1c10 : SUCCESS in 4m 51s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1e7b5a8267eb414f9c56146bed217f0c : SUCCESS in 5m 02s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/e7874d5257b84621a56314a88e2112b0 : SUCCESS in 11m 14s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/c41a7b2480024e5d872fdb3a56fb36be : SUCCESS in 11m 16s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c40ceaab9e224fbbb42f922441dfe710 : SUCCESS in 9m 20s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/97fd69890bc74fa9b8ba0fbe75c34e97 : SUCCESS in 9m 16s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f3ff780062f747b5ba514241447cee00 : SUCCESS in 9m 35s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"9719eca6bd7ead4af4957c2ce063ab616b1fac90","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-03-18 08:11:28.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"786605fc0b78b04ea594d7c6e14ec269a9dd90c2","tag":"autogenerated:gerrit:deleteVote","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-03-18 08:11:32.000000000","message":"Removed Workflow+1 by Tobias Urdin \u003ctobias.urdin@binero.com\u003e\n","accounts_in_message":[],"_revision_number":3},{"id":"b0e69fadac5232b4eb14b3f703eb6b01bfeda494","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2022-03-18 21:26:01.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"863759a34f84e73f9b4fd87aaa2b9b4f83de61b8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-18 21:27:34.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"7ca9e3569e0c583fa553b1e5521f5acab8fef3b4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-18 21:50:08.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/514205f91d414604bd632826fa9dbee9 : SUCCESS in 4m 32s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a81ec436136f42c2b459bf3b1952628f : SUCCESS in 5m 06s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2505349b36674593847aade7290e0b57 : SUCCESS in 5m 01s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/fca73212f64e479cb215c42bfa83d750 : SUCCESS in 10m 12s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/e2e7844f3730478399677702f023b9e6 : SUCCESS in 10m 14s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/59510a5fd71240b19f9c957b56ba6dae : SUCCESS in 8m 43s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/163c59fe80144b63be6dbd92b297ebe9 : SUCCESS in 9m 37s","accounts_in_message":[],"_revision_number":3},{"id":"3d998f499d8bdc2ea06ff473fa316566caec317c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-18 21:50:08.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"9d0381ab81f10a3edfe2537a6c3a9528f2581e86","revisions":{"146d0ac409b8bd45a9ced98471a9cabd2ad0688d":{"kind":"REWORK","_number":1,"created":"2022-03-18 03:47:22.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/57/834257/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-trove","ref":"refs/changes/57/834257/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/57/834257/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/57/834257/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/57/834257/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-trove refs/changes/57/834257/1"}}},"commit":{"parents":[{"commit":"977c0dc0e537286810828c785b1ef1c404df6fd2","subject":"guestagent: Deprecate the backup_aes_cbc_key parameter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/977c0dc0e537286810828c785b1ef1c404df6fd2"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-03-18 03:46:15.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-03-18 03:46:50.000000000","tz":540},"subject":"Avoid testing details of oslo::messaging(::*)","message":"Avoid testing details of oslo::messaging(::*)\n\n... so that any change in puppet-oslo would not directly break unit\ntests.\n\nChange-Id: Iba0599ff4485668ace1b044e6a922b2dafec6d3b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/146d0ac409b8bd45a9ced98471a9cabd2ad0688d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/146d0ac409b8bd45a9ced98471a9cabd2ad0688d"}]},"branch":"refs/heads/master"},"25eef6d65791b4e680ec20102b1a04943cbb5a13":{"kind":"REWORK","_number":2,"created":"2022-03-18 04:30:46.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/57/834257/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-trove","ref":"refs/changes/57/834257/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/57/834257/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/57/834257/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/57/834257/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-trove refs/changes/57/834257/2"}}},"commit":{"parents":[{"commit":"977c0dc0e537286810828c785b1ef1c404df6fd2","subject":"guestagent: Deprecate the backup_aes_cbc_key parameter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/977c0dc0e537286810828c785b1ef1c404df6fd2"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-03-18 03:46:15.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-03-18 04:30:37.000000000","tz":540},"subject":"Avoid testing details of oslo::messaging(::*)","message":"Avoid testing details of oslo::messaging(::*)\n\n... so that any change in puppet-oslo would not directly break unit\ntests.\n\nChange-Id: Iba0599ff4485668ace1b044e6a922b2dafec6d3b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/25eef6d65791b4e680ec20102b1a04943cbb5a13"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/25eef6d65791b4e680ec20102b1a04943cbb5a13"}]},"branch":"refs/heads/master"},"9d0381ab81f10a3edfe2537a6c3a9528f2581e86":{"kind":"REWORK","_number":3,"created":"2022-03-18 04:50:17.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/57/834257/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-trove","ref":"refs/changes/57/834257/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/57/834257/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/57/834257/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/57/834257/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-trove refs/changes/57/834257/3"}}},"commit":{"parents":[{"commit":"977c0dc0e537286810828c785b1ef1c404df6fd2","subject":"guestagent: Deprecate the backup_aes_cbc_key parameter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/977c0dc0e537286810828c785b1ef1c404df6fd2"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-03-18 03:46:15.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-03-18 04:50:07.000000000","tz":540},"subject":"Avoid testing details of oslo::messaging(::*)","message":"Avoid testing details of oslo::messaging(::*)\n\n... so that any change in puppet-oslo would not directly break unit\ntests.\n\nChange-Id: Iba0599ff4485668ace1b044e6a922b2dafec6d3b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/9d0381ab81f10a3edfe2537a6c3a9528f2581e86"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/9d0381ab81f10a3edfe2537a6c3a9528f2581e86"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
