)]}'
{"id":"openstack%2Fpuppet-trove~842711","triplet_id":"openstack%2Fpuppet-trove~master~Ib9eab7d3c174c11932afb05c66138dfe3cbf5a7c","project":"openstack/puppet-trove","branch":"master","topic":"trove-cleanup","attention_set":{},"removed_from_attention_set":{"9414":{"account":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"last_update":"2022-06-28 14:23:48.000000000","reason":"Change was submitted"},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-06-28 14:23:48.000000000","reason":"Change was submitted"},"16137":{"account":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"last_update":"2022-06-28 13:12:59.000000000","reason":"\u003cGERRIT_ACCOUNT_16137\u003e replied on the change","reason_account":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}},"hashtags":[],"change_id":"Ib9eab7d3c174c11932afb05c66138dfe3cbf5a7c","subject":"Remove use_neutron parameter","status":"MERGED","created":"2022-05-20 13:42:50.000000000","updated":"2022-06-28 14:25:39.000000000","submitted":"2022-06-28 14:23:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"842711-trove-cleanup","meta_rev_id":"881a81c1302583b0458db38b69ad531e54050294","_number":842711,"virtual_id_number":842711,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-06-28 14:23:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2022-06-24 13:33:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2022-06-28 13:12:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"date":"2022-06-28 02:10:00.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2022-06-28 13:12:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"}],"reviewers":{"REVIEWER":[{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-20 14:50:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-24 13:33:35.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2022-06-28 02:10:00.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"state":"REVIEWER"},{"updated":"2022-06-28 02:10:00.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"6af2c5c39855f8384560d14c3d15af784e626236","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-05-20 13:42:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e0cc58c0571a2c02c38d7bd865870064b1db642b","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-05-20 13:43:54.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"e9a4cf873367996c3fffcb23d5670c673e400f59","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-20 14:50:27.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7a4ba2097a4a4b6da1850c4f4da6181e : SUCCESS in 3m 03s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e6ebe1b20ff34b298d94695e2d3ab20c : SUCCESS in 2m 53s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9bcd0eb653d646089fec428b309e8030 : SUCCESS in 3m 20s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/dfda3fc89a1246c9a44dbafac0f500b3 : FAILURE in 9m 53s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ecd6e77adc3e4eafbb95aee4f349cd6f : FAILURE in 7m 15s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/89593f504d8e4afab1eb6cd8f4d78208 : FAILURE in 7m 14s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c27cd769ab3d4d5bbb21f2f7e4b3f621 : FAILURE in 7m 13s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/26d69aa3cc5c433baec89c076bf673a7 : FAILURE in 36m 45s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c9f44371dda84ec6bc09c4e042366c1c : SUCCESS in 1h 01m 54s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a1749c78418a402c8cd3daaccae0b88e : SUCCESS in 12m 54s (non-voting)\n- puppet-openstack-litmus-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/02649f7a3dea40a1abcdbfeebb36ec83 : SUCCESS in 18m 15s\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/5efa38f4db1d420fb8208214aeeffcab : FAILURE in 7m 59s (non-voting)\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9aac2e6dde284e83b1f1eb876bb491b8 : SUCCESS in 13m 27s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2a013a1c5c8f404a8d6695e3627482b5 : SUCCESS in 12m 36s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c8690838e7434adab930b947ebea2f59 : SUCCESS in 3m 09s","accounts_in_message":[],"_revision_number":1},{"id":"1b4701436d85183646b887619f41bc366f9e81c3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-05-21 03:12:28.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e8aee91cc0d173d62cec0874dd89f7ca52f40568","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-05-21 03:13:29.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"9eefd6a1d09af93d3be6ff65d221d7efecfd8a28","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-05-21 03:24:49.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"a005b6724bc3dc9b85ccb7dd6fa31d00349d9627","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-05-21 03:27:22.000000000","message":"Patch Set 3: Workflow-1","accounts_in_message":[],"_revision_number":3},{"id":"07883b1c9c5fa10d40c94164c21ca1ca88c916c1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-21 04:42:24.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3303ff3efdaf4e21af83675b11bb9d99 : SUCCESS in 3m 08s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6a44fd2663dc43f5a7f01f4270eee23c : SUCCESS in 3m 24s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3929a38721554b6ebe0af2dc3f18e7a2 : SUCCESS in 3m 27s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/5bb8db47097e465ca5b468dfdc1507b2 : SUCCESS in 10m 53s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9aeca8a6ba6d4d6299b6b5f304e15993 : SUCCESS in 8m 19s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cde09161916344458379ba60633986b9 : SUCCESS in 8m 24s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/99671c05a8d3465598715313ee67f120 : SUCCESS in 9m 47s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ab4fcf6e1c4646188a5c6d64d2c2b43e : FAILURE in 50m 00s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/56fbc9cf7a7742f7b54b9e38760477c2 : SUCCESS in 1h 10m 02s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a28201ed213541239c4625dfbf1decc0 : SUCCESS in 15m 56s (non-voting)\n- puppet-openstack-litmus-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/3e99f828ca49488bb03f8725a13836b9 : SUCCESS in 17m 19s\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/e22aa27cb4e540279a7d536526b15289 : FAILURE in 8m 41s (non-voting)\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b3705935e5ba466d97dc57cfb5860653 : SUCCESS in 11m 39s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e0e4717fc5994bd793ee3de7f13904e2 : SUCCESS in 16m 04s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/19e0f485606b46b8870c5fdd401fa75c : SUCCESS in 3m 51s","accounts_in_message":[],"_revision_number":3},{"id":"85914483789d16c1a2bf02e22e5d6ce4d35f643d","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-06-24 09:13:30.000000000","message":"Patch Set 3: -Workflow\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"2bad0c46f1c6ffd383fc3bf2f34ec7bab5ccf607","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2022-06-24 13:33:35.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"b9a7beaf96cd8c7ba1da46cd5648d62c50bf0531","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-06-28 13:12:59.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"f3d2954e395fbbb8cd5710e5e7ca06e92b2a703f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-28 13:14:37.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"255bedf4653bcda09cc24d7e834b4bc5c4f5f50e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-28 14:23:48.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1c7fc2a9485644d18e97a20d8bcad07f : SUCCESS in 3m 44s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d16ded23e09449288c04efd1a3af3a61 : SUCCESS in 3m 39s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2621e685b9474e1da9cd2c2c2a1df76e : SUCCESS in 3m 42s\n- puppet-openstack-unit-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/324b43bead2d4ac5bd19d5c32f88a131 : SUCCESS in 12m 23s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4be6ed29cd6042cea410081271e61fd9 : SUCCESS in 8m 15s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8685fe2c386d4ebfa2f1337af1e7df28 : SUCCESS in 9m 01s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/8d2047c1aa384469af218cf95728d9d3 : SUCCESS in 1h 03m 25s\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/973f5eae49be4448a1e0b69b6459b8af : SUCCESS in 17m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dfd9cf94d74b4463a27fac97016ce043 : SUCCESS in 3m 50s","accounts_in_message":[],"_revision_number":3},{"id":"7da838601ae7d2df8c9a00591aa3c82f80f87e33","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-28 14:23:48.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"881a81c1302583b0458db38b69ad531e54050294","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-28 14:25:39.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/38ae9be8cf344838a0801e125f75aa04 : SUCCESS in 1m 10s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"18a4e46704df2cd83f407f6bee57de3922e926a1","revisions":{"fc8bc60238856ccf2692da7eb67ee0b159c425c6":{"kind":"REWORK","_number":1,"created":"2022-05-20 13:42:50.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/11/842711/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-trove","ref":"refs/changes/11/842711/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/11/842711/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/11/842711/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/11/842711/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-trove refs/changes/11/842711/1"}}},"commit":{"parents":[{"commit":"4d87fa2c694adb33607fa33273faa849792351b1","subject":"Clean up deprecated items related to conductor/taskmanager","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/4d87fa2c694adb33607fa33273faa849792351b1"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-05-20 13:36:54.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-05-20 13:36:54.000000000","tz":540},"subject":"Remove use_neutron parameter","message":"Remove use_neutron parameter\n\n... because it was deprecated during Yoga cycle[1] and has had no\neffect since then.\n\n[1] ddb0f14f95156c483c499fe95982d2e5777b236f\n\nChange-Id: Ib9eab7d3c174c11932afb05c66138dfe3cbf5a7c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/fc8bc60238856ccf2692da7eb67ee0b159c425c6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/fc8bc60238856ccf2692da7eb67ee0b159c425c6"}]},"branch":"refs/heads/master"},"ce03b0b4444104a0003c909383ac4d0fef14a90d":{"kind":"REWORK","_number":2,"created":"2022-05-21 03:12:28.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/11/842711/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-trove","ref":"refs/changes/11/842711/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/11/842711/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/11/842711/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/11/842711/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-trove refs/changes/11/842711/2"}}},"commit":{"parents":[{"commit":"4d87fa2c694adb33607fa33273faa849792351b1","subject":"Clean up deprecated items related to conductor/taskmanager","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/4d87fa2c694adb33607fa33273faa849792351b1"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-05-20 13:36:54.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-05-21 03:12:14.000000000","tz":540},"subject":"Remove use_neutron parameter","message":"Remove use_neutron parameter\n\n... because it was deprecated during Yoga cycle[1] and has had no\neffect since then.\n\n[1] ddb0f14f95156c483c499fe95982d2e5777b236f\n\nChange-Id: Ib9eab7d3c174c11932afb05c66138dfe3cbf5a7c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/ce03b0b4444104a0003c909383ac4d0fef14a90d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/ce03b0b4444104a0003c909383ac4d0fef14a90d"}]},"branch":"refs/heads/master"},"18a4e46704df2cd83f407f6bee57de3922e926a1":{"kind":"REWORK","_number":3,"created":"2022-05-21 03:24:49.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/11/842711/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-trove","ref":"refs/changes/11/842711/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/11/842711/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/11/842711/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/11/842711/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-trove refs/changes/11/842711/3"}}},"commit":{"parents":[{"commit":"4d87fa2c694adb33607fa33273faa849792351b1","subject":"Clean up deprecated items related to conductor/taskmanager","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/4d87fa2c694adb33607fa33273faa849792351b1"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-05-20 13:36:54.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-05-21 03:24:20.000000000","tz":540},"subject":"Remove use_neutron parameter","message":"Remove use_neutron parameter\n\n... because it was deprecated during Yoga cycle[1] and has had no\neffect since then.\n\n[1] ddb0f14f95156c483c499fe95982d2e5777b236f\n\nChange-Id: Ib9eab7d3c174c11932afb05c66138dfe3cbf5a7c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/18a4e46704df2cd83f407f6bee57de3922e926a1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/18a4e46704df2cd83f407f6bee57de3922e926a1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[]}
