)]}'
{"id":"openstack%2Fpuppet-trove~842712","triplet_id":"openstack%2Fpuppet-trove~master~I6e7402bacd8ca23abda39d9fb3330cface5da288","project":"openstack/puppet-trove","branch":"master","topic":"trove-cleanup","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-06-28 14:23:50.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I6e7402bacd8ca23abda39d9fb3330cface5da288","subject":"Remove deprecated database options","status":"MERGED","created":"2022-05-20 13:42:50.000000000","updated":"2022-06-28 14:23:50.000000000","submitted":"2022-06-28 14:23:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"842712-trove-cleanup","meta_rev_id":"65b0399c1f303e1abaa770bdb5f99319833d64ed","_number":842712,"virtual_id_number":842712,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-06-28 14:23:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2022-06-21 12:57:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2022-06-23 21:45:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2022-06-23 21:45:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-20 14:41:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-21 12:57:44.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2022-06-23 21:45:15.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"550123052cda0ca6a131c123f5f351b4aea077a8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-05-20 13:42:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f8acf4041581f27cf9f028a87508648fe598335e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-05-20 13:51:23.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"3a69832844e178c97cf76b6d693c62edcf978c46","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-20 14:41:25.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fae175e2217446aba4be20fa240994b6 : FAILURE in 2m 52s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/529cceccbb1743f9bd3b47ee588f9b70 : SUCCESS in 2m 51s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/69f88babf43c43f9bc698575a4ed0e92 : SUCCESS in 3m 06s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/fcd8a41fd87f428f804dba7861133730 : FAILURE in 11m 22s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/de15b9b32b6d47099d6f001669da5501 : FAILURE in 7m 50s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/614738035abc4fcc822d784f87202523 : FAILURE in 8m 07s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/65b0303c932d4a9981d2fddd00c5c384 : FAILURE in 7m 56s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b492f2d5a5224eeb830144530b8c08f8 : FAILURE in 42m 55s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c9056e3240714c9f9671184fca841d63 : FAILURE in 12m 47s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/32660c54afc44c87aa283371e7785789 : FAILURE in 6m 56s (non-voting)\n- puppet-openstack-litmus-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/ca2f999cb7564757bb0ff369203e7e59 : FAILURE in 9m 55s\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/918cab107007436d8c2937a7a9a1fb61 : FAILURE in 8m 25s (non-voting)\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bdc868700c5843c18104f69531f41c6a : FAILURE in 7m 23s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/25fefe449ab34d7883d8622aedfc8de3 : FAILURE in 6m 50s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1584469db3064a92bfe685094c554adc : SUCCESS in 3m 37s","accounts_in_message":[],"_revision_number":2},{"id":"ea4fcc9ca5d52a6f60c35371010ef9c7a940c398","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-05-21 03:12:28.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"b135b0e3f25e9676f2e3edfea648fd732a2104db","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-05-21 03:17:15.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"2d0d7223dfb0fe64c58db2d9788a90ac063080b0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-05-21 03:24:49.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"d795663467f95a0e338e79e002ff2e0ca3ba9db5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-21 04:17:06.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3de30aa676d147cb83ceee406cfae0c0 : SUCCESS in 3m 17s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/414df133aaa446539d733453ce91ea82 : SUCCESS in 3m 00s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/df0ea44e86dc471685c066abe1d360d0 : SUCCESS in 3m 44s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/a6c6008d0092485e8081cf4f291141ae : SUCCESS in 10m 57s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5fd659a19dd54a5aae349acf45e8e939 : SUCCESS in 7m 53s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ae91e60fc1024226aae20a2bdedad147 : SUCCESS in 7m 39s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5730ca3b6d624052b5e5d6e04c86a356 : SUCCESS in 7m 46s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/946dbb3010fc457a921bcf35eb213d78 : FAILURE in 44m 15s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/8a530a5159eb479eb03d0aee640f6510 : FAILURE in 13m 19s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1b498835d88945f988b8ecd60ee72f94 : FAILURE in 6m 51s (non-voting)\n- puppet-openstack-litmus-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/c1ad68a7a1454beb92151fe0960ebb2e : FAILURE in 10m 35s\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/e5057ca907d84103bc4ce2b8b6846d50 : FAILURE in 9m 05s (non-voting)\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/52f6b4caaebd4cbd95e413a50ba66562 : FAILURE in 8m 40s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b08d4033d812430787c0b6386971132c : FAILURE in 7m 28s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4fe630af654341ce9afdb54e4fb5db37 : SUCCESS in 3m 28s","accounts_in_message":[],"_revision_number":5},{"id":"a510d70c6fc60c01b89070c9c44c0fa85845560b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-05-21 04:31:55.000000000","message":"Patch Set 6: Commit message was updated.","accounts_in_message":[],"_revision_number":6},{"id":"aab74ef51ba3522c5d01d4fa2516b93c5f98b242","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-21 05:40:29.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/57fb33a9a9464d5091793461fa21daf0 : SUCCESS in 2m 54s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/edc5e65837b149b6854207e5bcf1407c : SUCCESS in 2m 52s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7f47485d86a64624a7a3af06dbaa18c4 : SUCCESS in 2m 43s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/1472817294c947d39ee101001bbc0797 : SUCCESS in 9m 15s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f2ddbf50b5a6455eb2d07a63e575b824 : SUCCESS in 9m 58s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dbe890600d33474db0510e21e56197ce : SUCCESS in 7m 16s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e488a7f0246c4e54aa6d779866a23af7 : SUCCESS in 9m 38s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/24c1571f579949e3ace129bc132c4af9 : FAILURE in 33m 24s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/aa743c525b0f40bdb8e3bd6f66748c35 : SUCCESS in 1h 01m 54s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8beb547c2e4748599354707de8226894 : SUCCESS in 17m 36s (non-voting)\n- puppet-openstack-litmus-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/61aa934395ce4bfba1810b2fd1235082 : SUCCESS in 17m 01s\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/d475f72e38844156bf882a0e97abb6df : FAILURE in 9m 05s (non-voting)\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/496d2a47d8e34dee8a83e4682d8fedaf : SUCCESS in 12m 11s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5d1990cb63394d889afcb88d10246605 : SUCCESS in 17m 29s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2ebda84be89c48cf99a3fd41e47c35ab : SUCCESS in 3m 18s","accounts_in_message":[],"_revision_number":6},{"id":"125259af0bea39b335a0047fc21dc901ecaae6e1","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-06-21 12:57:44.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"30db2d505217b7f4d51baec9e249ba259cd204c7","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2022-06-23 21:45:15.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"362f4f519f91781888d9a904c5c8bfca78805096","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-06-28 13:13:13.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"d61c731481dbbca371917876cb9ddb054dabb46e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-28 13:14:52.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"734130d4d890fe53a453b03411cc84c44d2942ab","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-28 14:23:50.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/907f0e3f9ac94b2c91071b142e720b19 : SUCCESS in 3m 52s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e8ef6224134f4982b478f9e53898b613 : SUCCESS in 4m 01s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/326cfc34806341ecad7f8c7c9d96206b : SUCCESS in 3m 24s\n- puppet-openstack-unit-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/79cc75ea01ab496c91acf25c922779de : SUCCESS in 11m 03s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c449bf3b2896410e9338435ed5ef79fd : SUCCESS in 7m 57s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3288b6c26fcf408aa5c6dd27da7940a4 : SUCCESS in 8m 09s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/4d4c69a3be8b4c57bb9db09426d28bda : SUCCESS in 58m 24s\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/67e9e050c5764a11b365626183824a7b : SUCCESS in 15m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ac6ee4301fcf42b7b266445a08f80704 : SUCCESS in 5m 28s","accounts_in_message":[],"_revision_number":6},{"id":"65b0399c1f303e1abaa770bdb5f99319833d64ed","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-28 14:23:50.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"7d5e85200aae972efece21b7779a1e5065490a34","revisions":{"740af9c637310eb6ddccfb9ff98767a00f45c63f":{"kind":"REWORK","_number":1,"created":"2022-05-20 13:42:50.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/12/842712/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-trove","ref":"refs/changes/12/842712/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/12/842712/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/12/842712/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/12/842712/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-trove refs/changes/12/842712/1"}}},"commit":{"parents":[{"commit":"fc8bc60238856ccf2692da7eb67ee0b159c425c6","subject":"Remove use_neutron parameter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/fc8bc60238856ccf2692da7eb67ee0b159c425c6"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-05-20 13:41:09.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-05-20 13:41:09.000000000","tz":540},"subject":"Remove deprecated database options","message":"Remove deprecated database options\n\n... because these options were deprecated during Yoga cycle[1].\n\n[1] ddb0f14f95156c483c499fe95982d2e5777b236f\n\nChange-Id: I6e7402bacd8ca23abda39d9fb3330cface5da288\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/740af9c637310eb6ddccfb9ff98767a00f45c63f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/740af9c637310eb6ddccfb9ff98767a00f45c63f"}]},"branch":"refs/heads/master"},"571200d961f455725155b22d87f88a55830a1b08":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-05-20 13:51:23.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/12/842712/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-trove","ref":"refs/changes/12/842712/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/12/842712/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/12/842712/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/12/842712/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-trove refs/changes/12/842712/2"}}},"commit":{"parents":[{"commit":"fc8bc60238856ccf2692da7eb67ee0b159c425c6","subject":"Remove use_neutron parameter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/fc8bc60238856ccf2692da7eb67ee0b159c425c6"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-05-20 13:41:09.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-05-20 13:47:07.000000000","tz":540},"subject":"Remove deprecated database options","message":"Remove deprecated database options\n\n... because these options were deprecated during Yoga cycle[1].\n\n[1] a1f85ce1ef33f2d1f8c9bfe89e84d4f8a32248c0\n\nChange-Id: I6e7402bacd8ca23abda39d9fb3330cface5da288\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/571200d961f455725155b22d87f88a55830a1b08"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/571200d961f455725155b22d87f88a55830a1b08"}]},"branch":"refs/heads/master"},"e6f56b7e1facb1c86dfae40bbf69e545a5cc7db2":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2022-05-21 03:12:28.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/12/842712/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-trove","ref":"refs/changes/12/842712/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/12/842712/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/12/842712/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/12/842712/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-trove refs/changes/12/842712/3"}}},"commit":{"parents":[{"commit":"ce03b0b4444104a0003c909383ac4d0fef14a90d","subject":"Remove use_neutron parameter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/ce03b0b4444104a0003c909383ac4d0fef14a90d"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-05-20 13:41:09.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-05-21 03:12:19.000000000","tz":540},"subject":"Remove deprecated database options","message":"Remove deprecated database options\n\n... because these options were deprecated during Yoga cycle[1].\n\n[1] a1f85ce1ef33f2d1f8c9bfe89e84d4f8a32248c0\n\nChange-Id: I6e7402bacd8ca23abda39d9fb3330cface5da288\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/e6f56b7e1facb1c86dfae40bbf69e545a5cc7db2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/e6f56b7e1facb1c86dfae40bbf69e545a5cc7db2"}]},"branch":"refs/heads/master"},"e9302ca22c6d483ffd120e53014f3984f134471b":{"kind":"REWORK","_number":4,"created":"2022-05-21 03:17:15.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/12/842712/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-trove","ref":"refs/changes/12/842712/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/12/842712/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/12/842712/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/12/842712/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-trove refs/changes/12/842712/4"}}},"commit":{"parents":[{"commit":"ce03b0b4444104a0003c909383ac4d0fef14a90d","subject":"Remove use_neutron parameter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/ce03b0b4444104a0003c909383ac4d0fef14a90d"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-05-20 13:41:09.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-05-21 03:17:03.000000000","tz":540},"subject":"Remove deprecated database options","message":"Remove deprecated database options\n\n... because these options were deprecated during Yoga cycle[1].\n\n[1] a1f85ce1ef33f2d1f8c9bfe89e84d4f8a32248c0\n\nChange-Id: I6e7402bacd8ca23abda39d9fb3330cface5da288\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/e9302ca22c6d483ffd120e53014f3984f134471b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/e9302ca22c6d483ffd120e53014f3984f134471b"}]},"branch":"refs/heads/master"},"eb27e906d242a0d31807236258b447571c177097":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2022-05-21 03:24:49.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/12/842712/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-trove","ref":"refs/changes/12/842712/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/12/842712/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/12/842712/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/12/842712/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-trove refs/changes/12/842712/5"}}},"commit":{"parents":[{"commit":"18a4e46704df2cd83f407f6bee57de3922e926a1","subject":"Remove use_neutron parameter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/18a4e46704df2cd83f407f6bee57de3922e926a1"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-05-20 13:41:09.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-05-21 03:24:23.000000000","tz":540},"subject":"Remove deprecated database options","message":"Remove deprecated database options\n\n... because these options were deprecated during Yoga cycle[1].\n\n[1] a1f85ce1ef33f2d1f8c9bfe89e84d4f8a32248c0\n\nChange-Id: I6e7402bacd8ca23abda39d9fb3330cface5da288\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/eb27e906d242a0d31807236258b447571c177097"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/eb27e906d242a0d31807236258b447571c177097"}]},"branch":"refs/heads/master"},"7d5e85200aae972efece21b7779a1e5065490a34":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2022-05-21 04:31:55.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/12/842712/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-trove","ref":"refs/changes/12/842712/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/12/842712/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/12/842712/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/12/842712/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-trove refs/changes/12/842712/6"}}},"commit":{"parents":[{"commit":"18a4e46704df2cd83f407f6bee57de3922e926a1","subject":"Remove use_neutron parameter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/18a4e46704df2cd83f407f6bee57de3922e926a1"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-05-20 13:41:09.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-05-21 04:31:55.000000000","tz":0},"subject":"Remove deprecated database options","message":"Remove deprecated database options\n\n... because these options were deprecated during Yoga cycle[1].\n\n[1] a1f85ce1ef33f2d1f8c9bfe89e84d4f8a32248c0\n\nDepends-on: https://review.opendev.org/842828\nChange-Id: I6e7402bacd8ca23abda39d9fb3330cface5da288\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/7d5e85200aae972efece21b7779a1e5065490a34"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/7d5e85200aae972efece21b7779a1e5065490a34"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[]}
