)]}'
{"id":"openstack%2Fpuppet-trove~849193","triplet_id":"openstack%2Fpuppet-trove~master~I66c1ccedc2ef1a73b5ee3f053b843e97be0d59b4","project":"openstack/puppet-trove","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-07-25 16:03:11.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I66c1ccedc2ef1a73b5ee3f053b843e97be0d59b4","subject":"Remove resource types for per-daemon config files","status":"MERGED","created":"2022-07-09 13:52:38.000000000","updated":"2022-07-25 16:04:31.000000000","submitted":"2022-07-25 16:03:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"849193","meta_rev_id":"40604be10dcdd97702dfdb81fbba70d778464473","_number":849193,"virtual_id_number":849193,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-25 16:03:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2022-07-25 14:50:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2022-07-15 18:51:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":1,"date":"2022-07-25 14:50:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-09 14:28:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-11 22:02:25.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2022-07-15 18:51:31.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"c3cd44b287024f161d967ab0e425c33abf590d49","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-09 13:52:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7837ac57fa8817db7d96fc97b201089b9a2b3788","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-09 14:28:43.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/13686c2b6dfd4a578b01220450aa555b\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2ec0982bd5004c358e8602bd32b97bc5 : SUCCESS in 2m 54s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a71d793fd1c64732aa05a81ef5255bef : SUCCESS in 2m 34s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cf108f85a089486fac5cad77e95daf6f : SUCCESS in 2m 37s\n- puppet-openstack-unit-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/928664c5c09149c5a3f3be5baea3f683 : FAILURE in 8m 19s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4c1a20253fa2498285023dbaf15fbcf5 : FAILURE in 6m 21s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b6c9a243e49b426fa98f91f392f9dcda : FAILURE in 6m 28s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e810f3e34457442bbca04f67fe388686 : FAILURE in 6m 45s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/260040bf47394fdd9af8bda87c1202a1 : FAILURE in 34m 20s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c8c12dd7edf54ffa8db13b3baa3adeb6 : FAILURE in 9m 29s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7c256f5de20f4d858ef5b5ea072beb5c : FAILURE in 6m 43s (non-voting)\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/c5e3bb67358b49c7af6e8e59e441d406 : FAILURE in 8m 11s\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b73352bae37249d79ba90126c9a358d1 : FAILURE in 6m 27s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e4972aa8360749f58f620dfed7eae102 : FAILURE in 13m 53s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/37f86247d4c446c1ba6847b986e17bb3 : SUCCESS in 2m 57s","accounts_in_message":[],"_revision_number":1},{"id":"12b34e12febb19e6db6e7862f4b67730f8fe52be","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-09 14:32:37.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a0bc27620eb8f4c94a79ac32d6be74d4788f84ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-09 15:41:02.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8c143769d91b432990502fd00d6d8c16\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6cf017b53f4741ee98ea12c33b285500 : SUCCESS in 2m 47s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0391be96ce8e413f9c6febb5c5c830a1 : SUCCESS in 2m 38s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cb45b6ef86de492d8023be5efbcc743b : SUCCESS in 3m 09s\n- puppet-openstack-unit-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/e09001e93fde487d8d2d4ea8860b68f4 : SUCCESS in 10m 24s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/27c3e42de038443faf7ec0cf3996111e : SUCCESS in 6m 15s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/48071e6d89724a2eb29a96a89f5d10ca : SUCCESS in 7m 53s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3123dcbf1af74371b0c04f67c1c5dcd7 : SUCCESS in 6m 36s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0368517e60524914b2f880749ebf198c : FAILURE in 41m 16s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/ffc7b1b50c124ceba78e3764eaf795f2 : SUCCESS in 1h 07m 15s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b0d3c1c477cf478991678f7357573c46 : SUCCESS in 17m 13s (non-voting)\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/b07cc59e46d54e1b89a6838784e66bc2 : SUCCESS in 15m 34s\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4a1a1d9eb2a74a3485cb61b8e43fe6d7 : SUCCESS in 13m 16s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/04f98efd692e4dce868e7c31de216937 : SUCCESS in 15m 44s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8fec4e6a69b548e98cb450a559950393 : SUCCESS in 4m 05s","accounts_in_message":[],"_revision_number":2},{"id":"0e7cb7eaa405c2722a7387413a700b2e43bfc98d","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2022-07-11 22:02:25.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"95d7e332288ac5aefc6951b02b6b04db61c8aa42","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-15 09:06:10.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"f2539d9f9113488a7d92e4180989eb31897e3821","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-15 10:11:08.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e6c40c258d7a4c879fbf9e9ded5a3716\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d5e7a30ffc1f4a77a8fb219e2c9167d2 : SUCCESS in 3m 39s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7a3ba0980eda408bb6ea82edc0b318b3 : SUCCESS in 3m 21s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/af87956c221a4db9827cdd55ccac79df : SUCCESS in 3m 38s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/71a2ff7e002041ff9676820bd4ade64c : SUCCESS in 15m 38s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/22c1f223b0ba41208aaef8697df3bbdc : SUCCESS in 11m 00s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/21d40e6b62ad47a38a6bc2c5115174e2 : SUCCESS in 10m 53s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4740b47bc3cf41bfb6b5989a866711e3 : SUCCESS in 11m 14s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f253c860790b47f196a7230a53c31228 : FAILURE in 40m 34s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/2bd2425405de477eba82bb2b7ad140b8 : SUCCESS in 1h 02m 44s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/181b569a581a4f8c97c89261d888c16a : SUCCESS in 17m 31s (non-voting)\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/276b11ca29e147fa9eef052a60bbb86b : SUCCESS in 23m 12s\n- puppet-openstack-litmus-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dbfba7c04187420484c529736ba77260 : SUCCESS in 15m 46s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c500498b50b44f9a8d650556ccdc48bb : SUCCESS in 17m 32s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9d8c4c8cee5547d2bf8bb82bd2cfe6b8 : SUCCESS in 4m 18s","accounts_in_message":[],"_revision_number":3},{"id":"4f5f5f04192c9ed82a00366682bf919dc65c7517","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-07-15 18:51:31.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"ffd21a936017409c7b0dfd9c7ce268690c797845","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2022-07-25 14:50:13.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"392f1a2e0c8df4e04aa344dcb19c4b4d1fdf9214","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-25 14:51:26.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"0e2f5cd87dc113c0191287e275d161516bfec29b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-25 16:03:10.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9fc6611756394bd1badbad6279960fc3\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aea0041e9e3a47c0a64bdf2ba699505a : SUCCESS in 3m 35s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f826bfccc8cd4aa6ba9128298cd3355f : SUCCESS in 3m 11s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5ba7345dc6804b6cb197af478c09adc6 : SUCCESS in 3m 39s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/b74f5385f6a64e3a824d9e1e028ecffa : SUCCESS in 13m 03s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0047a7ca31c54d1eacec81aaa4e0325a : SUCCESS in 10m 08s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/abfef4de4fab418ab70d4d58630bf2ad : SUCCESS in 7m 42s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/973ad77d052346d38e4538dbaf4649c2 : SUCCESS in 1h 07m 42s\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/e511aadf41874e47a7cc20a2c429b7aa : SUCCESS in 22m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e07f995071c24872b4704ae9296f1649 : SUCCESS in 3m 57s","accounts_in_message":[],"_revision_number":3},{"id":"931503d5e08abf795427d6f0d177df513c53d2bc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-25 16:03:11.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"40604be10dcdd97702dfdb81fbba70d778464473","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-25 16:04:31.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9c39d60b9020471881cea7675e341cd4\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/577ec61b4adc4f53bcb10653f2b52aba : SUCCESS in 52s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"4d59d16be52af313d006550a9e8b5cb612fdd3b3","revisions":{"5c49d1887dded6a2fc0fa235f5796fe0b8b0aaf1":{"kind":"REWORK","_number":1,"created":"2022-07-09 13:52:38.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/93/849193/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-trove","ref":"refs/changes/93/849193/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/93/849193/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/93/849193/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/93/849193/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-trove refs/changes/93/849193/1"}}},"commit":{"parents":[{"commit":"8b7be7308835d892dac366f950a9f41a4eed14a1","subject":"Merge \"api: Remove redundant default values\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/8b7be7308835d892dac366f950a9f41a4eed14a1"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-09 13:49:04.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-09 13:52:28.000000000","tz":540},"subject":"Remove resource types for per-daemon config files","message":"Remove resource types for per-daemon config files\n\nSince yoga relase, this module no longer uses separate config files for\nindividual daemons but use the single trove.conf [1].\n\nThis change removes the remaining resource types for the separate\n.conf files, which should have been removed by the cleanup patch[2].\n\n[1] f1c11c68d56e4f1ad5125fa0df04ab91d929cd33\n[2] 4d87fa2c694adb33607fa33273faa849792351b1\n\nChange-Id: I66c1ccedc2ef1a73b5ee3f053b843e97be0d59b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/5c49d1887dded6a2fc0fa235f5796fe0b8b0aaf1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/5c49d1887dded6a2fc0fa235f5796fe0b8b0aaf1"}]},"branch":"refs/heads/master"},"340972170d04356e8b6444d669ed68e26350d28f":{"kind":"REWORK","_number":2,"created":"2022-07-09 14:32:37.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/93/849193/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-trove","ref":"refs/changes/93/849193/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/93/849193/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/93/849193/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/93/849193/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-trove refs/changes/93/849193/2"}}},"commit":{"parents":[{"commit":"8b7be7308835d892dac366f950a9f41a4eed14a1","subject":"Merge \"api: Remove redundant default values\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/8b7be7308835d892dac366f950a9f41a4eed14a1"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-09 13:49:04.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-09 14:32:28.000000000","tz":540},"subject":"Remove resource types for per-daemon config files","message":"Remove resource types for per-daemon config files\n\nSince yoga relase, this module no longer uses separate config files for\nindividual daemons but use the single trove.conf [1].\n\nThis change removes the remaining resource types for the separate\n.conf files, which should have been removed by the cleanup patch[2].\n\n[1] f1c11c68d56e4f1ad5125fa0df04ab91d929cd33\n[2] 4d87fa2c694adb33607fa33273faa849792351b1\n\nChange-Id: I66c1ccedc2ef1a73b5ee3f053b843e97be0d59b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/340972170d04356e8b6444d669ed68e26350d28f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/340972170d04356e8b6444d669ed68e26350d28f"}]},"branch":"refs/heads/master"},"4d59d16be52af313d006550a9e8b5cb612fdd3b3":{"kind":"REWORK","_number":3,"created":"2022-07-15 09:06:10.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/93/849193/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-trove","ref":"refs/changes/93/849193/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/93/849193/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/93/849193/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/93/849193/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-trove refs/changes/93/849193/3"}}},"commit":{"parents":[{"commit":"f2896b839fee10017b2bf07a820d308729fa7747","subject":"Fix wrong test description","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/f2896b839fee10017b2bf07a820d308729fa7747"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-09 13:49:04.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-15 09:05:59.000000000","tz":540},"subject":"Remove resource types for per-daemon config files","message":"Remove resource types for per-daemon config files\n\nSince yoga relase, this module no longer uses separate config files for\nindividual daemons but use the single trove.conf [1].\n\nThis change removes the remaining resource types for the separate\n.conf files, which should have been removed by the cleanup patch[2].\n\n[1] f1c11c68d56e4f1ad5125fa0df04ab91d929cd33\n[2] 4d87fa2c694adb33607fa33273faa849792351b1\n\nChange-Id: I66c1ccedc2ef1a73b5ee3f053b843e97be0d59b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/4d59d16be52af313d006550a9e8b5cb612fdd3b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/4d59d16be52af313d006550a9e8b5cb612fdd3b3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[]}
