)]}'
{"id":"openstack%2Fpuppet-trove~859736","triplet_id":"openstack%2Fpuppet-trove~master~I1e7f54ae462e72866d03a2ea7f00e656e68db513","project":"openstack/puppet-trove","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-09-29 21:40:46.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I1e7f54ae462e72866d03a2ea7f00e656e68db513","subject":"Remove deprecated trove::guestagent::control_exchange","status":"MERGED","created":"2022-09-28 23:55:01.000000000","updated":"2022-09-29 21:41:43.000000000","submitted":"2022-09-29 21:40:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"859736","meta_rev_id":"f432cde722183476448a41ac3823b9db295affb7","_number":859736,"virtual_id_number":859736,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-09-29 21:40:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2022-09-29 20:07:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2022-09-29 20:31:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2022-09-29 20:31:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-29 00:46:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-09-29 20:07:06.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2022-09-29 20:31:22.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"de378d40820fa6a795047a093ff550f67d427732","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-09-28 23:55:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f8c721bd1e407edafdadafb7b706bfaa6870653b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-29 00:46:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e147c0d0ce8e4474aadeaf068f4b68d4\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5f0a4a0848c04c01b0be5971841d1271 : SUCCESS in 3m 02s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8287104dfb574d5e876baaa6e4e33607 : SUCCESS in 2m 55s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3f8a4f0ef250463698e329806f486ecc : SUCCESS in 3m 04s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/8852fc8e5e9542629b31b75e8fffcd48 : SUCCESS in 10m 58s\n- puppet-openstack-unit-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c52d455203064182a8813504e817a219 : SUCCESS in 7m 34s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/80a4cd52925e49cca84bf119b8d9e75d : SUCCESS in 7m 33s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/12806c153897410f830ff6ee1e2b3d0b : SUCCESS in 7m 37s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/24b3c53d2e244c0f91c22faa638f0205 : SUCCESS in 39m 46s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/7f2555f3d38549b3aef75b303d0f3a18 : SUCCESS in 49m 34s\n- puppet-openstack-litmus-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1b3b1d83e4144b06a17067f04a2d4c97 : SUCCESS in 13m 33s (non-voting)\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/cf0170cb0657456e8123487b47a447b3 : SUCCESS in 19m 00s\n- puppet-openstack-litmus-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ecdac17c1bad4e92a1760badb1acdfe5 : SUCCESS in 13m 05s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8d00a8881028402589b3b8ad9e1a2eb8 : SUCCESS in 13m 25s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/94bbb931cf0740ad9288cbff1c629917 : SUCCESS in 3m 28s","accounts_in_message":[],"_revision_number":1},{"id":"2ee2696ae1e23e7a2688606cc67d98e0eebf6c86","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2022-09-29 20:07:06.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"a189155ef63cc3c44752b3e86c6fa78dfcc97492","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-09-29 20:31:22.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8a71542178bfa1b47fd47342e0b4d6c0d286b7ba","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-29 20:32:26.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cba8fe9a4fb32e0d7f8e1c0d1ba79343be72a975","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-29 21:40:46.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/851e627b48414b2aa8ae41c98b947ba5\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/21cf8ab5f6e645b6938bc863a0a8af32 : SUCCESS in 3m 16s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/68638060fada43ecb2eba38984378f0b : SUCCESS in 2m 59s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/61319f0fb4ad4dfab6f78f2e6173c700 : SUCCESS in 3m 38s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/702ecb7aeae04189ade977a51dc26465 : SUCCESS in 12m 45s\n- puppet-openstack-unit-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/41c7b70dabfa49e8b96698e45c0d3464 : SUCCESS in 6m 23s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f2d32f1e6475432199e782d21575822b : SUCCESS in 8m 49s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/963fbdc9ceb0475d84b48621ca65e3b7 : SUCCESS in 1h 04m 12s\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/d498ce9d479f4bb584a35d1b9b19279c : SUCCESS in 23m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/604bd2c4da4a4690a45be5c189e59b12 : SUCCESS in 4m 31s","accounts_in_message":[],"_revision_number":1},{"id":"288be1ad202df0e5cda4601df35092ec51bb3622","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-29 21:40:46.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"f432cde722183476448a41ac3823b9db295affb7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-29 21:41:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9462dd55566e4e83b300a5497de302fc\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9242a22a6f1b4c26962302510ad43c61 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d76efc7787bdf8a840f41453edf3b2aa3386efc2","revisions":{"d76efc7787bdf8a840f41453edf3b2aa3386efc2":{"kind":"REWORK","_number":1,"created":"2022-09-28 23:55:01.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/36/859736/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-trove","ref":"refs/changes/36/859736/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/36/859736/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/36/859736/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/36/859736/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-trove refs/changes/36/859736/1"}}},"commit":{"parents":[{"commit":"28089130676cabff3ae8f04a49970c5f09a43410","subject":"Enable memcached in acceptance tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/28089130676cabff3ae8f04a49970c5f09a43410"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-09-28 23:52:46.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-09-28 23:54:05.000000000","tz":540},"subject":"Remove deprecated trove::guestagent::control_exchange","message":"Remove deprecated trove::guestagent::control_exchange\n\nThe parameter was deprecated long ago[1] and has had no effect.\n\n[1] 7cb103c17ff121de8f852988f356386fcf4229f1\n\nChange-Id: I1e7f54ae462e72866d03a2ea7f00e656e68db513\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/d76efc7787bdf8a840f41453edf3b2aa3386efc2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/d76efc7787bdf8a840f41453edf3b2aa3386efc2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[]}
