)]}'
{"id":"openstack%2Fpuppet-trove~863337","triplet_id":"openstack%2Fpuppet-trove~stable%2Fwallaby~If5299ede0323e0c19bf327031c19811bb836ed9f","project":"openstack/puppet-trove","branch":"stable/wallaby","topic":"lp/1966382-stable/wallaby","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-11-04 01:25:10.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"If5299ede0323e0c19bf327031c19811bb836ed9f","subject":"setuptools: Disable auto discovery","status":"MERGED","created":"2022-11-02 15:14:48.000000000","updated":"2022-11-04 01:30:12.000000000","submitted":"2022-11-04 01:30:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"863337-lp/1966382-stable/wallaby","cherry_pick_of_change":863239,"cherry_pick_of_patch_set":2,"meta_rev_id":"95fa4ea603fed4932986faa273f91a1946b6d9a4","_number":863337,"virtual_id_number":863337,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-11-04 01:30:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2022-11-04 01:25:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2022-11-02 16:24:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2022-11-04 01:25:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-02 15:14:48.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"state":"CC"},{"updated":"2022-11-02 15:19:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-11-02 15:50:52.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"state":"REVIEWER"},{"updated":"2022-11-02 15:51:45.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"state":"REMOVED"},{"updated":"2022-11-02 16:24:03.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"3641306520345cefaae3342ca8da055196fe6a0a","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-11-02 15:14:48.000000000","message":"Patch Set 1: Cherry Picked from branch stable/xena.\n\nThe following files contain Git conflicts:\n* setup.py","accounts_in_message":[],"_revision_number":1},{"id":"536968c1e19d29fb7f1456c5305f711a7de3c0a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-02 15:19:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c43ab280cb78440dbf74a29c0163504c\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c1abd241868d4c38a2a6d0bd654c78a0 : SUCCESS in 2m 55s","accounts_in_message":[],"_revision_number":1},{"id":"9425dd4130f3542dd9f45352db0a41cbcbac5360","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-11-02 15:50:52.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c242e44a56fd60e80baa6a3cad9efb0e454dedac","tag":"autogenerated:gerrit:deleteReviewer","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-11-02 15:51:45.000000000","message":"Removed reviewer \u003cGERRIT_ACCOUNT_32926\u003e.","accounts_in_message":[{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"}],"_revision_number":2},{"id":"fd5bebbdb130dd6c4fbd0230214d2212f75a32df","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-11-02 15:51:47.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"c1b67ff413eaf40600439446550514353e3be711","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-02 15:58:09.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fce63b7724a54f76b73f52f9d867d260\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/159dcd86c5a246bf9291e5bf64f6a746 : SUCCESS in 2m 47s","accounts_in_message":[],"_revision_number":2},{"id":"063bae41b4e413b4542487804cace44b85aa7d3d","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2022-11-02 16:24:03.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"8df1a177853c34b0b0318649c49131dacc89fa25","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-11-04 01:25:10.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"885400d72e1bbaae579a4d81ad1df738093e1ac1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-04 01:25:33.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"0eab97781809f16eb244f76d0953c0e1f9bbb920","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-04 01:30:12.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/23466e4f7ac64c8fad4c100dc202060e\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7927d141fdc24bfc9913f0c222d3501d : SUCCESS in 3m 57s","accounts_in_message":[],"_revision_number":2},{"id":"95fa4ea603fed4932986faa273f91a1946b6d9a4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-04 01:30:12.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"1c180f650868f704e57414d7246ed160757dda95","revisions":{"3b158025db2ea26ffa39e0f6daf6b18a17eb6eb2":{"kind":"REWORK","_number":1,"created":"2022-11-02 15:14:48.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/37/863337/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-trove","ref":"refs/changes/37/863337/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/37/863337/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/37/863337/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/37/863337/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-trove refs/changes/37/863337/1"}}},"commit":{"parents":[{"commit":"8360c2e08e9a62637131e86f5240535266eddb2f","subject":"Set fixture paths for unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/8360c2e08e9a62637131e86f5240535266eddb2f"}]}],"author":{"name":"Jiri Podivin","email":"jpodivin@redhat.com","date":"2022-03-29 08:42:49.000000000","tz":120},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-11-02 15:14:48.000000000","tz":0},"subject":"setuptools: Disable auto discovery","message":"setuptools: Disable auto discovery\n\nThe latest release of setuptools 61.0 made a breaking change[1] and\nbecause of this change \u0027pip install\u0027 fails with the following error.\n\n~~~\nerror: Multiple top-level packages discovered in a flat-layout:\n[\u0027lib\u0027, \u0027spec\u0027, \u0027manifests\u0027, \u0027releasenotes\u0027].\n~~~\n\nUsers that don\u0027t set \u0027packages\u0027, \u0027py_modules\u0027, or configuration\u0027\nare still likely to observe the auto-discovery behavior, which may\nhalt the build if the project contains multiple directories and/or\nmultiple Python files directly under the project root.\n\nTo disable auto discovery, one can do below in setup.py\n\n~~~\nsetuptools.setup(..,packages\u003d[],..)\n~~~\n\nor\n\n~~~\nsetuptools.setup(..,py_modules\u003d[],..)\n~~~\n\n[1] https://github.com/pypa/setuptools/issues/3197\n\nNote setup.py is not used to install puppet modules. However it is used\nto generate a release note, thus should be fixed.\n\nSigned-off-by: Jiri Podivin \u003cjpodivin@redhat.com\u003e\nChange-Id: If5299ede0323e0c19bf327031c19811bb836ed9f\n(cherry picked from commit de17c05a96a10982e1162a47cfbfa29577103037)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/3b158025db2ea26ffa39e0f6daf6b18a17eb6eb2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/3b158025db2ea26ffa39e0f6daf6b18a17eb6eb2"}]},"branch":"refs/heads/stable/wallaby"},"1c180f650868f704e57414d7246ed160757dda95":{"kind":"REWORK","_number":2,"created":"2022-11-02 15:50:52.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/37/863337/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-trove","ref":"refs/changes/37/863337/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/37/863337/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/37/863337/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/37/863337/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-trove refs/changes/37/863337/2"}}},"commit":{"parents":[{"commit":"8360c2e08e9a62637131e86f5240535266eddb2f","subject":"Set fixture paths for unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/8360c2e08e9a62637131e86f5240535266eddb2f"}]}],"author":{"name":"Jiri Podivin","email":"jpodivin@redhat.com","date":"2022-03-29 08:42:49.000000000","tz":120},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-11-02 15:50:49.000000000","tz":540},"subject":"setuptools: Disable auto discovery","message":"setuptools: Disable auto discovery\n\nThe latest release of setuptools 61.0 made a breaking change[1] and\nbecause of this change \u0027pip install\u0027 fails with the following error.\n\n~~~\nerror: Multiple top-level packages discovered in a flat-layout:\n[\u0027lib\u0027, \u0027spec\u0027, \u0027manifests\u0027, \u0027releasenotes\u0027].\n~~~\n\nUsers that don\u0027t set \u0027packages\u0027, \u0027py_modules\u0027, or configuration\u0027\nare still likely to observe the auto-discovery behavior, which may\nhalt the build if the project contains multiple directories and/or\nmultiple Python files directly under the project root.\n\nTo disable auto discovery, one can do below in setup.py\n\n~~~\nsetuptools.setup(..,packages\u003d[],..)\n~~~\n\nor\n\n~~~\nsetuptools.setup(..,py_modules\u003d[],..)\n~~~\n\n[1] https://github.com/pypa/setuptools/issues/3197\n\nNote setup.py is not used to install puppet modules. However it is used\nto generate a release note, thus should be fixed.\n\nConflicts:\n\tsetup.py\n\nSigned-off-by: Jiri Podivin \u003cjpodivin@redhat.com\u003e\nChange-Id: If5299ede0323e0c19bf327031c19811bb836ed9f\n(cherry picked from commit de17c05a96a10982e1162a47cfbfa29577103037)\n(cherry picked from commit 5b6ad03310c300daf2ecfb9b904c3d66dcd9692b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/1c180f650868f704e57414d7246ed160757dda95"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/1c180f650868f704e57414d7246ed160757dda95"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}]}],"submit_requirements":[]}
