)]}'
{"id":"openstack%2Fpuppet-trove~874369","triplet_id":"openstack%2Fpuppet-trove~master~Ic7b1d555cf2090f7278980edf226ddf318be489f","project":"openstack/puppet-trove","branch":"master","topic":"ga-opts","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2023-02-21 20:04:57.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic7b1d555cf2090f7278980edf226ddf318be489f","subject":"Support mysql/mariadb options","status":"MERGED","created":"2023-02-20 04:58:41.000000000","updated":"2023-02-21 20:05:52.000000000","submitted":"2023-02-21 20:04:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"874369-ga-opts","meta_rev_id":"1bd50c3764fd6fa005a856a101560dfe71253576","_number":874369,"virtual_id_number":874369,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-21 20:04:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2023-02-20 16:48:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2023-02-21 18:59:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2023-02-21 18:59:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-20 06:22:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-20 16:48:44.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2023-02-21 18:59:29.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"7c95c5723aeced0bfd9368e1d26d29c2c544a5e6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-02-20 04:58:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"52b06ee13c86a0fde2560d85d129abae351a7ed5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-02-20 05:09:21.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"885f902a6b8303b157f5a76826596c346dda8557","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-20 06:22:05.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/28b31d80b3394f4891a3ec3de92ceb99\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6217ac4f6a3f49cb9f75d99eead246d1 : FAILURE in 2m 55s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f49d403d9a0d411c9d29ef5bb34bc9cb : SUCCESS in 2m 51s\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/953c9ce097c4453ca95a5089af850c96 : FAILURE in 9m 31s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1ccf33a923114f0d879463b26cb5cd5d : FAILURE in 6m 16s\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ad8ecb4826384996ae8097cff807ac71 : FAILURE in 6m 44s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c712ee56708c4be1a3e692c09397cf91 : FAILURE in 28m 15s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/199da2977c254b4894285441584893a4 : SUCCESS in 1h 03m 34s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/a48af18c666942b3a44a8feb9ab8fd79 : SUCCESS in 18m 47s\n- puppet-openstack-litmus-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9860971842864fa4aea43f68c0c0bd05 : SUCCESS in 13m 19s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4250d39326524fdd851ea1c6c3c87a9e : SUCCESS in 15m 31s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/80c3532165ed4e8887a435bd741ac4b5 : SUCCESS in 3m 11s","accounts_in_message":[],"_revision_number":2},{"id":"1968e43ab97434dea4ea816f60bf0391475defbb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-02-20 06:27:43.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"80a499810a33016bb4eccd8984f6a4f4968e8bf7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-02-20 06:45:21.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"0502a663bca16f59853026190ffc1e3529fb84e8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-02-20 06:50:47.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"3f9df3a6aa2ac133bc879b1314b8bb00c5bd1653","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-20 08:06:05.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/bc793f2cfd334f80967cca217ed07b35\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/819bd25dcc5141d0a6999eee4df9e5fd : SUCCESS in 2m 24s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/15215d3ff1264980ba7f30c981c76310 : SUCCESS in 2m 40s\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/48dd7c444cf649d49c78e6ca8e7539fb : FAILURE in 11m 46s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/eceab5f6b2db420ca54728a2e30b8218 : FAILURE in 6m 05s\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/997999c4c2c04319a4eb7cfbe82e8c9e : FAILURE in 5m 26s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/436e2dc7145d48c4bbd525a327610662 : FAILURE in 32m 22s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/5131230229f440fc9c736295ee4f1468 : SUCCESS in 1h 09m 52s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/300bc4dd3cb946bb88df9ecd56cf3f06 : SUCCESS in 14m 36s\n- puppet-openstack-litmus-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a07ecf2aca5e48cfa9aa501b53b59e1d : SUCCESS in 13m 38s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d0cf8a66c167445baa6fbec8f55d2486 : SUCCESS in 16m 11s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1ddde53d0b4d4a2ea09ec83bbf1c2e65 : SUCCESS in 2m 54s","accounts_in_message":[],"_revision_number":5},{"id":"762d6bfa62e533db737324f2adf5a0f0b73d6931","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-02-20 09:49:35.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"2a7f939fed38086400c8240eb9975d56f7b672a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-20 11:01:15.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/69cb6e7bbc314c09be2b8c6dbc5bfbdb\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4e899af07fbd4d4dae609f511ebf053c : SUCCESS in 3m 08s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a78ed740e45a48c4a1ab7bc4d790d71f : SUCCESS in 2m 37s\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/513f744cc4d54951ba31c1ff971deec7 : FAILURE in 11m 37s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b7deb91e66904098a95a47d567b89eb6 : FAILURE in 8m 39s\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bf44560b41db456995d4f8c4b50a2387 : FAILURE in 9m 03s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/05f72542f5654f5cadb6e46a32b7da4e : FAILURE in 24m 59s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/79e04b9faadf4b109a0f5f21a2522880 : SUCCESS in 1h 03m 35s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/8ec6ab0f9b5f4c06934ed2ad96c6f5c9 : SUCCESS in 17m 22s\n- puppet-openstack-litmus-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b43e9e6f4ed34c82aa8ca7ae28698a38 : SUCCESS in 14m 21s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9cd0e255ee5b42eb8f93ea029ce7c1a8 : FAILURE in 11m 40s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e473d215237b473db367e389326a891c : SUCCESS in 2m 54s","accounts_in_message":[],"_revision_number":6},{"id":"d7ce68a288e329ea69d92f029744d5549760db1e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-02-20 11:01:50.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"946182a78a54a38597b996e3e080b4b920e686a7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-02-20 11:03:04.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"8868c44e816e98b5af32986c55787526b5b5359c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-20 12:13:05.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d1432f51f9c94abca1f504976b63f14a\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/73cb2d925d854d7a996b2a6621fa16d1 : SUCCESS in 2m 39s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c7f7fc7cd33a406a9b835b9477b05f44 : SUCCESS in 3m 08s\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/cff9edd822aa429ab1babcaf816aab62 : SUCCESS in 10m 28s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cd89a02c1b2d47309463fa1d917779c0 : SUCCESS in 6m 01s\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1ff06811afc64a46b09c90328e7ed9c9 : SUCCESS in 5m 53s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/93917a77037e40f1bed701378d1ef2d7 : FAILURE in 27m 40s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/03ccf95c85ac4148bc27e75d580fb9c5 : SUCCESS in 1h 03m 12s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/627c77ccf3694f42aaf81275b4e01971 : SUCCESS in 21m 11s\n- puppet-openstack-litmus-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c008b8ed44b7478684c3c8143fbd6751 : SUCCESS in 12m 39s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/461deea8e04049378564c37cdf0b653f : SUCCESS in 12m 21s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b9c5c20cb5844d4dabc54ad1e3c38fa5 : SUCCESS in 3m 43s","accounts_in_message":[],"_revision_number":8},{"id":"58905458acde0ed9905b92db3075eafe6fe696e6","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2023-02-20 16:48:44.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"bc19cd28c2601d7e405e0831e690ea1cbdbc58a1","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2023-02-21 18:59:29.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"2aecdfc2d772e5dccd2cd900511e4c535052d950","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-21 19:00:27.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"15692aad5a9eb7ed4ffbdb5866ab74e2392103f7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-21 20:04:56.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fe20e9b1bed640538e2382951a376e74\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bf199097245646938961d0396ce3e4ac : SUCCESS in 3m 05s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f8fc261f7e454369ac13f6cb33f61deb : SUCCESS in 2m 49s\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/5c5ff3f05aae45fc8f54b59099eb2858 : SUCCESS in 9m 45s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a0ad6d413e0f4782ab1af072bee07464 : SUCCESS in 7m 06s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/de122e07872241db99a441f23ae7db76 : SUCCESS in 1h 03m 10s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/7754c9ddabb1481996c5e971e5d609c0 : SUCCESS in 18m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/99c71c688c0741459cfc7691cf3e0e26 : SUCCESS in 3m 12s","accounts_in_message":[],"_revision_number":8},{"id":"df651a333c35186b1a418702c45e20e8d4ca11f6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-21 20:04:57.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":8},{"id":"1bd50c3764fd6fa005a856a101560dfe71253576","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-21 20:05:52.000000000","message":"Patch Set 8:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/455b0f31cf3a4d7e9c3f7d28b6bf848d\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d2da850ee3344451980e5ae82cf86f09 : SUCCESS in 37s","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"12131677e9c06c8d25fe31a7666493e27157099f","revisions":{"692c69ee3a02d988c8df66c081587b08363577b5":{"kind":"REWORK","_number":1,"created":"2023-02-20 04:58:41.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/69/874369/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-trove","ref":"refs/changes/69/874369/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/69/874369/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/69/874369/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/69/874369/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-trove refs/changes/69/874369/1"}}},"commit":{"parents":[{"commit":"0ca65d22c5d54837bca96c6358e49fabcceb5cfe","subject":"Merge \"Replace mocha by rspec-mocks\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/0ca65d22c5d54837bca96c6358e49fabcceb5cfe"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-02-20 04:56:40.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-02-20 04:58:26.000000000","tz":540},"subject":"Support [mysql] options","message":"Support [mysql] options\n\nThis change introduces the new class to manage [mysql] options.\n\nChange-Id: Ic7b1d555cf2090f7278980edf226ddf318be489f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/692c69ee3a02d988c8df66c081587b08363577b5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/692c69ee3a02d988c8df66c081587b08363577b5"}]},"branch":"refs/heads/master"},"b6fe02a3e2b236a2061b3ff382b39f2cd9306e17":{"kind":"REWORK","_number":2,"created":"2023-02-20 05:09:21.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/69/874369/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-trove","ref":"refs/changes/69/874369/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/69/874369/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/69/874369/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/69/874369/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-trove refs/changes/69/874369/2"}}},"commit":{"parents":[{"commit":"0ca65d22c5d54837bca96c6358e49fabcceb5cfe","subject":"Merge \"Replace mocha by rspec-mocks\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/0ca65d22c5d54837bca96c6358e49fabcceb5cfe"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-02-20 04:56:40.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-02-20 05:07:13.000000000","tz":540},"subject":"Support [mysql] options","message":"Support [mysql] options\n\nThis change introduces the new class to manage [mysql] options, so that\nusers can configure behavior of the guest agent to launch and manage\nmysql instances.\n\nThis also introduces support for options to manage container registry\nused to pull mysql images.\n\nChange-Id: Ic7b1d555cf2090f7278980edf226ddf318be489f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/b6fe02a3e2b236a2061b3ff382b39f2cd9306e17"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/b6fe02a3e2b236a2061b3ff382b39f2cd9306e17"}]},"branch":"refs/heads/master"},"5c1ab88b074672e11a02a732bc64cc18c913a5ef":{"kind":"REWORK","_number":3,"created":"2023-02-20 06:27:43.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/69/874369/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-trove","ref":"refs/changes/69/874369/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/69/874369/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/69/874369/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/69/874369/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-trove refs/changes/69/874369/3"}}},"commit":{"parents":[{"commit":"0ca65d22c5d54837bca96c6358e49fabcceb5cfe","subject":"Merge \"Replace mocha by rspec-mocks\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/0ca65d22c5d54837bca96c6358e49fabcceb5cfe"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-02-20 04:56:40.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-02-20 06:27:30.000000000","tz":540},"subject":"Support [mysql] options","message":"Support [mysql] options\n\nThis change introduces the new class to manage [mysql] options, so that\nusers can configure behavior of the guest agent to launch and manage\nmysql instances.\n\nThis also introduces support for options to manage container registry\nused to pull mysql images.\n\nChange-Id: Ic7b1d555cf2090f7278980edf226ddf318be489f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/5c1ab88b074672e11a02a732bc64cc18c913a5ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/5c1ab88b074672e11a02a732bc64cc18c913a5ef"}]},"branch":"refs/heads/master"},"c36710c9c25cb75cfacfb2ae103dcf994f5604fe":{"kind":"REWORK","_number":4,"created":"2023-02-20 06:45:21.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/69/874369/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-trove","ref":"refs/changes/69/874369/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/69/874369/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/69/874369/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/69/874369/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-trove refs/changes/69/874369/4"}}},"commit":{"parents":[{"commit":"0ca65d22c5d54837bca96c6358e49fabcceb5cfe","subject":"Merge \"Replace mocha by rspec-mocks\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/0ca65d22c5d54837bca96c6358e49fabcceb5cfe"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-02-20 04:56:40.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-02-20 06:45:12.000000000","tz":540},"subject":"Support [mysql] options","message":"Support [mysql] options\n\nThis change introduces the new class to manage [mysql] options, so that\nusers can configure behavior of the guest agent to launch and manage\nmysql instances.\n\nThis also introduces support for options to manage container registry\nused to pull mysql images.\n\nChange-Id: Ic7b1d555cf2090f7278980edf226ddf318be489f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/c36710c9c25cb75cfacfb2ae103dcf994f5604fe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/c36710c9c25cb75cfacfb2ae103dcf994f5604fe"}]},"branch":"refs/heads/master"},"79d78cac1b33dc75f63b60d4d0e77bc84eeedfcb":{"kind":"REWORK","_number":5,"created":"2023-02-20 06:50:47.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/69/874369/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-trove","ref":"refs/changes/69/874369/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/69/874369/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/69/874369/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/69/874369/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-trove refs/changes/69/874369/5"}}},"commit":{"parents":[{"commit":"0ca65d22c5d54837bca96c6358e49fabcceb5cfe","subject":"Merge \"Replace mocha by rspec-mocks\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/0ca65d22c5d54837bca96c6358e49fabcceb5cfe"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-02-20 04:56:40.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-02-20 06:50:38.000000000","tz":540},"subject":"Support mysql/mariadb options","message":"Support mysql/mariadb options\n\nThis change introduces the new class to manage [mysql] options and\n[mariadb] options so that users can configure behavior of the guest\nagent to launch and manage mysql instances and mariadb instances.\n\nThis also introduces support for options to manage container registry\nused to pull container images.\n\nChange-Id: Ic7b1d555cf2090f7278980edf226ddf318be489f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/79d78cac1b33dc75f63b60d4d0e77bc84eeedfcb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/79d78cac1b33dc75f63b60d4d0e77bc84eeedfcb"}]},"branch":"refs/heads/master"},"29e0726c3e5e302266071251b84791a61926ea09":{"kind":"REWORK","_number":6,"created":"2023-02-20 09:49:35.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/69/874369/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-trove","ref":"refs/changes/69/874369/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/69/874369/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/69/874369/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/69/874369/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-trove refs/changes/69/874369/6"}}},"commit":{"parents":[{"commit":"0ca65d22c5d54837bca96c6358e49fabcceb5cfe","subject":"Merge \"Replace mocha by rspec-mocks\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/0ca65d22c5d54837bca96c6358e49fabcceb5cfe"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-02-20 04:56:40.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-02-20 09:49:26.000000000","tz":540},"subject":"Support mysql/mariadb options","message":"Support mysql/mariadb options\n\nThis change introduces the new class to manage [mysql] options and\n[mariadb] options so that users can configure behavior of the guest\nagent to launch and manage mysql instances and mariadb instances.\n\nThis also introduces support for options to manage container registry\nused to pull container images.\n\nChange-Id: Ic7b1d555cf2090f7278980edf226ddf318be489f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/29e0726c3e5e302266071251b84791a61926ea09"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/29e0726c3e5e302266071251b84791a61926ea09"}]},"branch":"refs/heads/master"},"9478ee8fc00e046ce16fc1eef2d9cb7b736955c6":{"kind":"REWORK","_number":7,"created":"2023-02-20 11:01:50.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/69/874369/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-trove","ref":"refs/changes/69/874369/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/69/874369/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/69/874369/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/69/874369/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-trove refs/changes/69/874369/7"}}},"commit":{"parents":[{"commit":"0ca65d22c5d54837bca96c6358e49fabcceb5cfe","subject":"Merge \"Replace mocha by rspec-mocks\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/0ca65d22c5d54837bca96c6358e49fabcceb5cfe"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-02-20 04:56:40.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-02-20 11:01:42.000000000","tz":540},"subject":"Support mysql/mariadb options","message":"Support mysql/mariadb options\n\nThis change introduces the new class to manage [mysql] options and\n[mariadb] options so that users can configure behavior of the guest\nagent to launch and manage mysql instances and mariadb instances.\n\nThis also introduces support for options to manage container registry\nused to pull container images.\n\nChange-Id: Ic7b1d555cf2090f7278980edf226ddf318be489f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/9478ee8fc00e046ce16fc1eef2d9cb7b736955c6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/9478ee8fc00e046ce16fc1eef2d9cb7b736955c6"}]},"branch":"refs/heads/master"},"12131677e9c06c8d25fe31a7666493e27157099f":{"kind":"REWORK","_number":8,"created":"2023-02-20 11:03:04.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/69/874369/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-trove","ref":"refs/changes/69/874369/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/69/874369/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/69/874369/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/69/874369/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-trove refs/changes/69/874369/8"}}},"commit":{"parents":[{"commit":"0ca65d22c5d54837bca96c6358e49fabcceb5cfe","subject":"Merge \"Replace mocha by rspec-mocks\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/0ca65d22c5d54837bca96c6358e49fabcceb5cfe"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-02-20 04:56:40.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-02-20 11:02:56.000000000","tz":540},"subject":"Support mysql/mariadb options","message":"Support mysql/mariadb options\n\nThis change introduces the new class to manage [mysql] options and\n[mariadb] options so that users can configure behavior of the guest\nagent to launch and manage mysql instances and mariadb instances.\n\nThis also introduces support for options to manage container registry\nused to pull container images.\n\nChange-Id: Ic7b1d555cf2090f7278980edf226ddf318be489f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/12131677e9c06c8d25fe31a7666493e27157099f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/12131677e9c06c8d25fe31a7666493e27157099f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[]}
