)]}'
{"id":"openstack%2Fpuppet-trove~901110","triplet_id":"openstack%2Fpuppet-trove~master~I198e2cc089a12e1ef260d78665f274db9149f9c9","project":"openstack/puppet-trove","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2023-11-17 18:16:05.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I198e2cc089a12e1ef260d78665f274db9149f9c9","subject":"Deprecate usage of authtoken parameters in providers","status":"MERGED","created":"2023-11-16 06:45:58.000000000","updated":"2023-11-17 18:17:10.000000000","submitted":"2023-11-17 18:16:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"901110","meta_rev_id":"77347504433d22f0dd6e69abb7b73f2ac2006aa4","_number":901110,"virtual_id_number":901110,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-11-17 18:16:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2023-11-17 14:07:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2023-11-17 15:14:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2023-11-17 15:14:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-16 06:49:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-17 14:07:16.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2023-11-17 15:14:08.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"a5833a163b420482801a52f3289a43e5658bfd33","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-11-16 06:45:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5abef31b8b74cbfaf5b1a15d6b563f63ee4cb619","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-16 06:49:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9d7fed19e8734fb9a3083675165dd483\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6d07ef9593ae4e8fba3d1705218a349e : SUCCESS in 2m 48s","accounts_in_message":[],"_revision_number":1},{"id":"97728d7de13f8f1ace9382e6145563c98e8064e7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-11-16 06:54:04.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"182d39c86f3815dd7b8bcdfd43476560ddf205b3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-16 08:05:52.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/84271261244149fdabf312e731802ea0\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/18876267458848ce8a3168abcd4e7022 : SUCCESS in 2m 22s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/64de5affdc8f45029b440c14b2aefc45 : SUCCESS in 2m 30s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/d426a0ae7d584bccbe3118d60cd630e3 : SUCCESS in 5m 45s (non-voting)\n- puppet-openstack-unit-7.26-centos-9-stream https://zuul.opendev.org/t/openstack/build/254004c41285400db3bea94847c120a3 : SUCCESS in 8m 54s\n- puppet-openstack-unit-7.26-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cb1703d4e73446048ee825ab3146d3fb : SUCCESS in 7m 28s\n- puppet-openstack-unit-8.2-centos-9-stream https://zuul.opendev.org/t/openstack/build/a43bb590e3494bc8852a44c2511ab676 : SUCCESS in 9m 18s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/36fe8eadc7f44861bf235ee42a42314d : SUCCESS in 6m 37s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/298cf370aa3643c9889445950718a469 : SUCCESS in 45m 05s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/7939b4b379824264a2a531e5b9648c22 : SUCCESS in 1h 04m 15s\n- puppet-openstack-litmus-7.26-centos-9-stream https://zuul.opendev.org/t/openstack/build/8cc3461408974c8199111656e41c29db : SUCCESS in 24m 49s\n- puppet-openstack-litmus-7.26-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a10ee31c509e409da62817d77b1f93c7 : SUCCESS in 17m 33s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6e16c5a702614ea4b2c5da2be4eb1f3d : SUCCESS in 19m 10s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/00d0fa38b9df46349e90a768cdb02c8d : SUCCESS in 3m 38s","accounts_in_message":[],"_revision_number":2},{"id":"c5dc4143a63ef3fa6a55585637ac078c9361d25f","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2023-11-17 14:07:16.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"bd3d049aaa4620efe23acc9ab488910d39e0fb19","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2023-11-17 15:14:08.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"ef64c06d404afc8673a93e8deefb7e34baca8fa2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-17 15:15:10.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d660274490ef7efc7ca451d4f08a304b6167a437","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-17 18:16:05.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8f174100627c472e87bfe8aa7d42126c\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c86f14d8ed57427fb9e72eebd66d2193 : SUCCESS in 2m 48s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fac15201527641a591f7f9dbaedb8285 : SUCCESS in 2m 47s\n- puppet-openstack-unit-7.26-centos-9-stream https://zuul.opendev.org/t/openstack/build/182fbde3ff37480f95ccf58309877dfb : SUCCESS in 11m 12s\n- puppet-openstack-unit-7.26-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/67e7b4502be54baaa3bbe7bce4dc572a : SUCCESS in 7m 03s\n- puppet-openstack-unit-8.2-centos-9-stream https://zuul.opendev.org/t/openstack/build/d92af38fdc7b4655a9e4b99506d858d4 : SUCCESS in 10m 49s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/b6a263e6c7de41b7af724d890b27b336 : SUCCESS in 1h 06m 09s\n- puppet-openstack-litmus-7.26-centos-9-stream https://zuul.opendev.org/t/openstack/build/2580d4c93e074188b966319494c1a9cd : SUCCESS in 24m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/18b5b909e60240ebb184e96880ca38db : SUCCESS in 3m 15s","accounts_in_message":[],"_revision_number":2},{"id":"dabd88b2031028469dc7522b738f69e3e4880c0f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-17 18:16:05.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"77347504433d22f0dd6e69abb7b73f2ac2006aa4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-17 18:17:10.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cdd8ff0a803e462b8187e04a22adf118\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5a2a2f723ac94943945c65724187b0ee : SUCCESS in 39s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"f99ea85cd380d03b88162005e961293a471b7656","revisions":{"c60f6d61ee882ce7a2a8acd9880bb330bbd93d77":{"kind":"REWORK","_number":1,"created":"2023-11-16 06:45:58.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/10/901110/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-trove","ref":"refs/changes/10/901110/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/10/901110/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/10/901110/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/10/901110/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-trove refs/changes/10/901110/1"}}},"commit":{"parents":[{"commit":"428129c2e1867804861830257de6db78541dfb27","subject":"Remove usage of deleted manifest_dir","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/428129c2e1867804861830257de6db78541dfb27"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-16 06:45:34.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-16 06:45:34.000000000","tz":540},"subject":"Deprecate usage of authtoken parameters in providers","message":"Deprecate usage of authtoken parameters in providers\n\nWhen SRBAC is completely implemented according to the current spec,\neach service user no longer has full admin privilege but has access\nlimited to the actual requirement. This means that we are no longer\nable to use service users to create resources.\n\nThis change is a prep-work to get rid of usage of the service user,\ncurrently loaded from authtoken parameters in the .conf file. Now\neach providers loads clouds.yaml file placed in /etc/openstack/puppet.\n\nChange-Id: I198e2cc089a12e1ef260d78665f274db9149f9c9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/c60f6d61ee882ce7a2a8acd9880bb330bbd93d77"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/c60f6d61ee882ce7a2a8acd9880bb330bbd93d77"}]},"branch":"refs/heads/master"},"f99ea85cd380d03b88162005e961293a471b7656":{"kind":"REWORK","_number":2,"created":"2023-11-16 06:54:04.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/10/901110/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-trove","ref":"refs/changes/10/901110/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/10/901110/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/10/901110/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/10/901110/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-trove refs/changes/10/901110/2"}}},"commit":{"parents":[{"commit":"428129c2e1867804861830257de6db78541dfb27","subject":"Remove usage of deleted manifest_dir","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/428129c2e1867804861830257de6db78541dfb27"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-16 06:45:34.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-16 06:53:54.000000000","tz":540},"subject":"Deprecate usage of authtoken parameters in providers","message":"Deprecate usage of authtoken parameters in providers\n\nWhen SRBAC is completely implemented according to the current spec,\neach service user no longer has full admin privilege but has access\nlimited to the actual requirement. This means that we are no longer\nable to use service users to create resources.\n\nThis change is a prep-work to get rid of usage of the service user,\ncurrently loaded from authtoken parameters in the .conf file. Now\neach providers loads clouds.yaml file placed in /etc/openstack/puppet.\n\nChange-Id: I198e2cc089a12e1ef260d78665f274db9149f9c9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/f99ea85cd380d03b88162005e961293a471b7656"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/f99ea85cd380d03b88162005e961293a471b7656"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
