)]}'
{"id":"openstack%2Fpuppet-trove~907645","triplet_id":"openstack%2Fpuppet-trove~master~Ic4f8b9dc285efb4cd983cc6ba8a9eeaedd188456","project":"openstack/puppet-trove","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-02-05 22:15:48.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic4f8b9dc285efb4cd983cc6ba8a9eeaedd188456","subject":"Add support for caching options","status":"MERGED","created":"2024-02-03 07:39:19.000000000","updated":"2024-02-05 22:16:42.000000000","submitted":"2024-02-05 22:15:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"907645","meta_rev_id":"4aa8affe5ba95f474397d70a6996d59e285720bf","_number":907645,"virtual_id_number":907645,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-02-05 22:15:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2024-02-05 17:29:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2024-02-05 21:05:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2024-02-05 21:05:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-03 08:56:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-05 17:29:34.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2024-02-05 21:05:43.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"b89181433b387915b1162ff7e0d9666f77580a23","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-03 07:39:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3eb5c5019a199dfb5bcf24155ed02eab38be0d01","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-03 08:56:35.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/dc87a89f73c1410bbc1e5e81574e7dc9\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cf89c2cff7e44e49bb7835c8e430373a : FAILURE in 2m 32s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/458c036e21af40fb8dc44a7636ee9579 : SUCCESS in 2m 38s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/1d333aa8ea0c46ba82c285178536e53e : SUCCESS in 5m 48s (non-voting)\n- puppet-openstack-unit-7.27-centos-9-stream https://zuul.opendev.org/t/openstack/build/664c998eb8214f3dbb8c231b6c66352f : FAILURE in 8m 12s\n- puppet-openstack-unit-7.27-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f4fc29e5ec884c20baec0804a726f43f : FAILURE in 7m 56s\n- puppet-openstack-unit-8.3-centos-9-stream https://zuul.opendev.org/t/openstack/build/75060cb4106347c8b27972b393ec5799 : FAILURE in 10m 07s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/04c9a5e4359f4748b7a2d65b4e1f8bb9 : FAILURE in 6m 54s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bfbba309b69f4408a3c3abe70d66b372 : SUCCESS in 53m 17s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/5a584250f42e4eea8e4c49f1ed92d56f : SUCCESS in 1h 11m 00s\n- puppet-openstack-litmus-7.27-centos-9-stream https://zuul.opendev.org/t/openstack/build/452a8263760144c28b532c9d1e9962eb : SUCCESS in 22m 12s\n- puppet-openstack-litmus-7.27-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8077f5aa98e44b40b318693184ef52a5 : SUCCESS in 18m 29s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d142065f1015449dba6c24aed5890a47 : SUCCESS in 20m 31s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f78ebb18b65347ad9b42d770aef7019a : SUCCESS in 2m 48s","accounts_in_message":[],"_revision_number":1},{"id":"26f9ed234465f44fb79f8abd5b27ea4198b1037f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-03 14:36:44.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"079396869364cc11aa037b1b225bf39506ed3542","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-03 15:51:08.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/aa9fc4c3da0d4f978096fc5b0a282d74\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/418f8b6b607144808d839a1e2016549d : SUCCESS in 3m 32s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1278c051763a4527a5db5edc25c7f1ed : SUCCESS in 2m 17s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/efe888420dba402ebd4c34a3f9fa7f84 : SUCCESS in 5m 34s (non-voting)\n- puppet-openstack-unit-7.27-centos-9-stream https://zuul.opendev.org/t/openstack/build/45810eae73a343528345110406674864 : FAILURE in 9m 59s\n- puppet-openstack-unit-7.27-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a6a07a1e5a2b49949e06755cba0d8658 : FAILURE in 7m 00s\n- puppet-openstack-unit-8.3-centos-9-stream https://zuul.opendev.org/t/openstack/build/0581e871a0ad4dc191a8ca97ec04efe0 : FAILURE in 10m 09s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/359cead25eed4b8da90cd603fe125bbc : FAILURE in 6m 47s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/33c359864d0f4216960900b650691b36 : SUCCESS in 53m 05s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/8e052240bd80479d9c0ef8902942308f : SUCCESS in 1h 05m 03s\n- puppet-openstack-litmus-7.27-centos-9-stream https://zuul.opendev.org/t/openstack/build/c7f0faf7a29f48938ac9bd1c062c54a8 : SUCCESS in 24m 07s\n- puppet-openstack-litmus-7.27-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9195103b186140cf8e47b257dc5d53b2 : SUCCESS in 16m 41s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4ab522bda3f24b299ea4f074fb4b5c80 : SUCCESS in 19m 22s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cb4e7f3ad01446778695a803315ceee9 : SUCCESS in 2m 56s","accounts_in_message":[],"_revision_number":2},{"id":"ff41a0b537877ccc6c51451cbc74b746e7aaefe4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-03 15:56:16.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"f25d2e9cde1bd20c578d263ea87fe79fa0c4dbd3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-03 16:37:41.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"60a64c0b023e14efa1ce588bb230cb131ffa8b3f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-03 17:23:36.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4d083e21645845ba9ae80ed77367a732\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/51b9acb6b49640b5a8abd5b97dc7dfcf : SUCCESS in 2m 21s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/aee6050bee954335a9a21ae65c31f451 : SUCCESS in 2m 25s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/e9824e38aa7c4e0091bce217462718b0 : SUCCESS in 4m 39s (non-voting)\n- puppet-openstack-unit-7.27-centos-9-stream https://zuul.opendev.org/t/openstack/build/fa27c3b15da443a593db27f750114816 : SUCCESS in 8m 20s\n- puppet-openstack-unit-7.27-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d298623ed6dd4116bc7aac2828aec8f2 : SUCCESS in 6m 15s\n- puppet-openstack-unit-8.3-centos-9-stream https://zuul.opendev.org/t/openstack/build/16a6ce01108a455088d6c9258ef4d69d : SUCCESS in 8m 09s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/10df46086b504ff4b45d62dfb3547e5b : SUCCESS in 5m 52s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b45fbd1fc4b747d0aa3b18a7ea2b6fb1 : SUCCESS in 39m 09s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/8727df61f0764141a11ac8da940945d0 : SUCCESS in 44m 16s\n- puppet-openstack-litmus-7.27-centos-9-stream https://zuul.opendev.org/t/openstack/build/3976c9c1c4d14337b9e12bdf208d3335 : SUCCESS in 19m 44s\n- puppet-openstack-litmus-7.27-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d2b2c3e866724162b1cec8c9cd778208 : SUCCESS in 16m 14s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/be51ee2e33a74e46a48709dbef22c541 : SUCCESS in 18m 31s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e6970a4868bf410c957a640883a23ca4 : SUCCESS in 2m 53s","accounts_in_message":[],"_revision_number":4},{"id":"f90e50c14987505a61919dbf0f84ac28acb95774","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2024-02-05 17:29:34.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"a7794d903b7e5bba98f961ef12ec24d99eb099a9","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2024-02-05 21:05:43.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"f290286567fad3ad3c86826b4acf570bb1211b9e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-05 21:06:57.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"512249b0ab31c0352129ff6ea294a0dbc7cc4802","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-05 22:15:47.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cf1c93df50d146f9b25307831e4fe34c\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cb04683121cd458c887831039794db3d : SUCCESS in 2m 31s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ca1ba7c27b8740dfa2d6c457b5faa611 : SUCCESS in 3m 22s\n- puppet-openstack-unit-7.27-centos-9-stream https://zuul.opendev.org/t/openstack/build/a60299520f2e42eeb32a53fe5d2ea80d : SUCCESS in 10m 06s\n- puppet-openstack-unit-7.27-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6b4f1def950d41c6a3ac032c1afb56af : SUCCESS in 6m 02s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/6861234084ac4d02a34fae433b3b693b : SUCCESS in 1h 07m 37s\n- puppet-openstack-litmus-7.27-centos-9-stream https://zuul.opendev.org/t/openstack/build/de3043921f1041db9a8ea64b26737aac : SUCCESS in 23m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a2b99a9180fb4aa59cd9036997c4130e : SUCCESS in 2m 58s","accounts_in_message":[],"_revision_number":4},{"id":"943795a2c87312e0d2ba3aac4c47aa83cc66874d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-05 22:15:48.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"4aa8affe5ba95f474397d70a6996d59e285720bf","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-05 22:16:42.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/79f9dd348dc6496d988564570cb1b563\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/75e18c585c0948639fcfa273d7158605 : SUCCESS in 36s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"65214a004aba12ad8b8b75104a2c06c3f7fb8555","revisions":{"a3c74168748ddf6424cb5eb10c50da734ea7c92c":{"kind":"REWORK","_number":1,"created":"2024-02-03 07:39:19.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/45/907645/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-trove","ref":"refs/changes/45/907645/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/45/907645/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/45/907645/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/45/907645/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-trove refs/changes/45/907645/1"}}},"commit":{"parents":[{"commit":"0a832b7a0665ce01bf95e4da3e646286160d6b0f","subject":"Add support for volume options","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/0a832b7a0665ce01bf95e4da3e646286160d6b0f"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-03 07:38:35.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-03 07:38:35.000000000","tz":540},"subject":"Add support for caching options","message":"Add support for caching options\n\nTrove supports caching using the oslo.cache library. This add support\nfor customizing caching options.\n\nChange-Id: Ic4f8b9dc285efb4cd983cc6ba8a9eeaedd188456\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/a3c74168748ddf6424cb5eb10c50da734ea7c92c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/a3c74168748ddf6424cb5eb10c50da734ea7c92c"}]},"branch":"refs/heads/master"},"5244b3b6a85aa08469fb4af211d676c0cebba689":{"kind":"REWORK","_number":2,"created":"2024-02-03 14:36:44.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/45/907645/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-trove","ref":"refs/changes/45/907645/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/45/907645/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/45/907645/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/45/907645/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-trove refs/changes/45/907645/2"}}},"commit":{"parents":[{"commit":"0a832b7a0665ce01bf95e4da3e646286160d6b0f","subject":"Add support for volume options","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/0a832b7a0665ce01bf95e4da3e646286160d6b0f"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-03 07:38:35.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-03 14:36:44.000000000","tz":0},"subject":"Add support for caching options","message":"Add support for caching options\n\nTrove supports caching using the oslo.cache library. This add support\nfor customizing caching options.\n\nChange-Id: Ic4f8b9dc285efb4cd983cc6ba8a9eeaedd188456\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/5244b3b6a85aa08469fb4af211d676c0cebba689"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/5244b3b6a85aa08469fb4af211d676c0cebba689"}]},"branch":"refs/heads/master"},"4943b6cfab0bc579ce79c61c91bdf660cd6067f5":{"kind":"REWORK","_number":3,"created":"2024-02-03 15:56:16.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/45/907645/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-trove","ref":"refs/changes/45/907645/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/45/907645/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/45/907645/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/45/907645/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-trove refs/changes/45/907645/3"}}},"commit":{"parents":[{"commit":"0a832b7a0665ce01bf95e4da3e646286160d6b0f","subject":"Add support for volume options","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/0a832b7a0665ce01bf95e4da3e646286160d6b0f"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-03 07:38:35.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-03 15:56:05.000000000","tz":540},"subject":"Add support for caching options","message":"Add support for caching options\n\nTrove supports caching using the oslo.cache library. This add support\nfor customizing caching options.\n\nChange-Id: Ic4f8b9dc285efb4cd983cc6ba8a9eeaedd188456\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/4943b6cfab0bc579ce79c61c91bdf660cd6067f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/4943b6cfab0bc579ce79c61c91bdf660cd6067f5"}]},"branch":"refs/heads/master"},"65214a004aba12ad8b8b75104a2c06c3f7fb8555":{"kind":"REWORK","_number":4,"created":"2024-02-03 16:37:41.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/45/907645/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-trove","ref":"refs/changes/45/907645/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/45/907645/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/45/907645/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/45/907645/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-trove refs/changes/45/907645/4"}}},"commit":{"parents":[{"commit":"0a832b7a0665ce01bf95e4da3e646286160d6b0f","subject":"Add support for volume options","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/0a832b7a0665ce01bf95e4da3e646286160d6b0f"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-03 07:38:35.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-03 16:37:31.000000000","tz":540},"subject":"Add support for caching options","message":"Add support for caching options\n\nTrove supports caching using the oslo.cache library. This add support\nfor customizing caching options.\n\nChange-Id: Ic4f8b9dc285efb4cd983cc6ba8a9eeaedd188456\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/65214a004aba12ad8b8b75104a2c06c3f7fb8555"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/65214a004aba12ad8b8b75104a2c06c3f7fb8555"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
