)]}'
{"id":"openstack%2Fpuppet-trove~914048","triplet_id":"openstack%2Fpuppet-trove~master~I9c5e44d4a8bd73e85efa1a8f254a228ad312d677","project":"openstack/puppet-trove","branch":"master","topic":"bug/1485485","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-03-26 12:08:20.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I9c5e44d4a8bd73e85efa1a8f254a228ad312d677","subject":"guestagent: Support volume operation options","status":"MERGED","created":"2024-03-25 01:34:14.000000000","updated":"2024-03-26 12:09:22.000000000","submitted":"2024-03-26 12:08:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"914048-bug/1485485","meta_rev_id":"7deefb1b50ef2f8b673c1f57d493ffe31c7c55d7","_number":914048,"virtual_id_number":914048,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-03-26 12:08:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2024-03-25 14:58:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2024-03-26 11:12:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2024-03-26 11:12:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-03-25 02:51:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-03-25 14:58:00.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2024-03-26 11:12:38.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"f7203f909236c9678e06d626d2dab139451e8a0a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-03-25 01:34:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5e341acb7e93bfb656410715e2541a2badb5fce4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-03-25 01:40:22.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e2310e09b0fa59e28bf24ff5e67cf04173b1b434","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-25 02:51:35.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/743bb034422943618ea715d2d524256d\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3128f73ddf2a4280be84b73f9a467ab7 : SUCCESS in 2m 24s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7774e2c8621440118a66ea1bdbb269b3 : SUCCESS in 2m 27s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/4cd9e9e08d484f6ba18f1dbe312fc3d3 : SUCCESS in 5m 22s (non-voting)\n- puppet-openstack-unit-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/2005560af4a0464c98fdced781619a98 : SUCCESS in 9m 12s\n- puppet-openstack-unit-7.29-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bc0edcb4ef534acdb8b331de6383e2b0 : SUCCESS in 6m 43s\n- puppet-openstack-unit-8.5-centos-9-stream https://zuul.opendev.org/t/openstack/build/cce6dcb718394ea2a590636c3e6eaf82 : SUCCESS in 8m 29s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a7cb6666d4fa436f8bf7cfc23d57bc84 : SUCCESS in 6m 34s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0a1caf5148db42578b4d43477fa8bf1a : SUCCESS in 38m 43s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/27597193b9ae45c68214ddaeefec0ba7 : SUCCESS in 59m 15s\n- puppet-openstack-litmus-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/43a38e832808408e873f8f89ad7b7efa : SUCCESS in 22m 24s\n- puppet-openstack-litmus-7.29-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c3a1b6fe76614c848b8a2fd2fa903e28 : SUCCESS in 16m 44s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/befdc586d3fa4938a1bf24234fae3f95 : SUCCESS in 18m 50s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2e756341508b446b958d8c95058a9316 : SUCCESS in 2m 25s","accounts_in_message":[],"_revision_number":2},{"id":"f094c4c1060a73def32a630b9a751a26c7e9df50","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2024-03-25 14:58:00.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"bd072637ff73028252981fc4a371881c2dcbdd0e","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2024-03-26 11:12:38.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"57d29d3b26e4201c6c51c99e388235d9a626ebfe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-26 11:13:28.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"1c5da3691d8b1911a6ee85425ce2bfe8a6be44b2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-26 12:08:20.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e7d5977c16134517b7e41cce54b7684c\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/05f91cd60d6943cc9544f7162161e31f : SUCCESS in 2m 55s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b4c1024ba96546c68621838ae37dcba6 : SUCCESS in 2m 47s\n- puppet-openstack-unit-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/a9937f1ac1394d01b075f4b2faec9f01 : SUCCESS in 9m 14s\n- puppet-openstack-unit-7.29-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/26e95d16ff244749850544b041e85eac : SUCCESS in 6m 50s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/7ced8416951f4606b6c6c1080b8db1f7 : SUCCESS in 54m 47s\n- puppet-openstack-litmus-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/8940ea572fde41af83f27b14fb0e0f4a : SUCCESS in 25m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cf772b8fd548492187b74ed01acaea22 : SUCCESS in 2m 39s","accounts_in_message":[],"_revision_number":2},{"id":"e84f5f2164d7f644f11b883cd99b7f8d8b60394c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-26 12:08:20.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"7deefb1b50ef2f8b673c1f57d493ffe31c7c55d7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-26 12:09:22.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bbe0ca1966f74e329c4bf4db8d6afcd1\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/015f1c52c93643d2a9b7dd93ac5a62d3 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"5108129ad064eca2be8e925587c0de9dc08d619d","revisions":{"e80653192053562c2367ab8ebd860b0165366a3d":{"kind":"REWORK","_number":1,"created":"2024-03-25 01:34:14.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/48/914048/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-trove","ref":"refs/changes/48/914048/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/48/914048/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/48/914048/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/48/914048/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-trove refs/changes/48/914048/1"}}},"commit":{"parents":[{"commit":"225592993c634b156af54d755bda41b9d8c132cc","subject":"Merge \"Expose rabbit_transient_quorum_queue\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/225592993c634b156af54d755bda41b9d8c132cc"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-03-25 01:33:21.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-03-25 01:33:21.000000000","tz":540},"subject":"guestagent: Support volume operation options","message":"guestagent: Support volume operation options\n\nCloses-Bug: #1485485\nChange-Id: I9c5e44d4a8bd73e85efa1a8f254a228ad312d677\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/e80653192053562c2367ab8ebd860b0165366a3d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/e80653192053562c2367ab8ebd860b0165366a3d"}]},"branch":"refs/heads/master"},"5108129ad064eca2be8e925587c0de9dc08d619d":{"kind":"REWORK","_number":2,"created":"2024-03-25 01:40:22.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/48/914048/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-trove","ref":"refs/changes/48/914048/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/48/914048/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/48/914048/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/48/914048/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-trove refs/changes/48/914048/2"}}},"commit":{"parents":[{"commit":"225592993c634b156af54d755bda41b9d8c132cc","subject":"Merge \"Expose rabbit_transient_quorum_queue\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/225592993c634b156af54d755bda41b9d8c132cc"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-03-25 01:33:21.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-03-25 01:40:12.000000000","tz":540},"subject":"guestagent: Support volume operation options","message":"guestagent: Support volume operation options\n\nCloses-Bug: #1485485\nChange-Id: I9c5e44d4a8bd73e85efa1a8f254a228ad312d677\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/5108129ad064eca2be8e925587c0de9dc08d619d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/5108129ad064eca2be8e925587c0de9dc08d619d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
