)]}'
{"id":"openstack%2Fpuppet-trove~914052","triplet_id":"openstack%2Fpuppet-trove~master~I4016752d94da35ea257f9c76079a33797151640b","project":"openstack/puppet-trove","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-03-25 16:22:13.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4016752d94da35ea257f9c76079a33797151640b","subject":"Prepare to rename trove::api::service_credentials","status":"MERGED","created":"2024-03-25 04:06:59.000000000","updated":"2024-03-25 16:24:04.000000000","submitted":"2024-03-25 16:22:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"914052","meta_rev_id":"145e1a91e0d1bc35070dc2e7a3e375c8916d5170","_number":914052,"virtual_id_number":914052,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-03-25 16:22:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2024-03-25 09:07:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2024-03-25 14:58:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2024-03-25 14:58:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-03-25 04:34:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-03-25 09:07:40.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2024-03-25 14:58:23.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"4324421aed09be3866f744d4a3afac9141d5bf95","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-03-25 04:06:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f8486257cd459710a82ef70865f8a5873fc60571","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-25 04:34:57.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d75e0a74969f4a0fb463734227bc7b3a\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/44de11f40ce54ad0b5d5f847c0818ce6 : SUCCESS in 3m 18s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0054ef3ba5f14af09bedee7ad27b2f5d : FAILURE in 2m 29s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/08a9d6fe86fd4712a2f98fc6f95e8741 : FAILURE in 4m 12s (non-voting)\n- puppet-openstack-unit-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/a62c7790a4494b70bee9e7a19e152d24 : FAILURE in 7m 12s\n- puppet-openstack-unit-7.29-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fd20dd0b1c31496f8bb3e2fdd08e0f2e : FAILURE in 5m 50s\n- puppet-openstack-unit-8.5-centos-9-stream https://zuul.opendev.org/t/openstack/build/59e3326f0d78400ba7da36f60010488b : FAILURE in 9m 06s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c43a5baee9a44232889009ec90e150a9 : FAILURE in 8m 01s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/beebf3e02ea54f429e090803111c9ce9 : FAILURE in 5m 29s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/9a177cd1393342eea5d5d026b22c55eb : FAILURE in 9m 29s\n- puppet-openstack-litmus-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/86c58cc080b44584a0e592e47a4afe95 : FAILURE in 10m 10s\n- puppet-openstack-litmus-7.29-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/68d3cffe38d14ad9aa67b9b5670c0941 : FAILURE in 5m 31s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e6d24adc26e640efbdb6bc7c9dfb72a3 : FAILURE in 7m 42s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8bafa43af92f4c088b2eec7f0258c477 : SUCCESS in 2m 16s","accounts_in_message":[],"_revision_number":1},{"id":"00baa4f2d74972229d0a3dbc6fe0bab9ca67e704","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-03-25 05:11:14.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"579715b3676c6c9199da13056374b7dfb3fa0aa6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-25 06:13:26.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/98345c9ae48c4b7d80a1e5b56ae23536\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6e18109230ea4e28b938f6ccdc2ac67f : SUCCESS in 2m 20s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/525799028d3b4c898a49250cef9392af : SUCCESS in 2m 39s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/7275da7d34784cb8a7cd921cda5cbc44 : SUCCESS in 4m 18s (non-voting)\n- puppet-openstack-unit-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/ebfeb096bbde4ee6960527c052236198 : SUCCESS in 8m 21s\n- puppet-openstack-unit-7.29-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5fedd8f42fed4d8bb96293af14e1be26 : SUCCESS in 5m 58s\n- puppet-openstack-unit-8.5-centos-9-stream https://zuul.opendev.org/t/openstack/build/15df826ccdf84c9fbc03b370fc85a0d7 : SUCCESS in 10m 45s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/32c927b0840f4b8a8bf6fc5000d586af : SUCCESS in 8m 03s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a5562116075b4ebd89f68b9d90935213 : SUCCESS in 47m 12s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/369dcd24e68f48eca44f8e525d0d8d1e : SUCCESS in 56m 53s\n- puppet-openstack-litmus-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/3c3b348ab1f64e938ddeb6a7c074328d : SUCCESS in 26m 20s\n- puppet-openstack-litmus-7.29-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5e0f82bb99ac4f37b6843c570ec7cde4 : SUCCESS in 17m 53s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3f282c86f3a94ddc81925ccbf6c154e0 : SUCCESS in 18m 04s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1a35bc9dcaa749289ce045f1261d687e : SUCCESS in 3m 15s","accounts_in_message":[],"_revision_number":2},{"id":"2a8cc0bb01abd4deb8535fc5678f7d84db22ae14","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2024-03-25 09:07:40.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"6061df660a607894e7044d3eaa144b01ea6fd77a","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2024-03-25 14:58:23.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"295c6e658a39ea4b071bec11063a93fd49208af5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-25 14:59:15.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"137bb26ffe91a4587f321951d4bcd2db9fa558f7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-25 16:22:13.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7e0db5c513a14e5787db644ac32286cf\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/518cbb3df1a04e3ab85d62eb1ac29955 : SUCCESS in 2m 39s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/25fc31587a3746969d8fe4ef60efca85 : SUCCESS in 2m 23s\n- puppet-openstack-unit-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/f5fbd8a618564b78ab155e7cd90e6a7b : SUCCESS in 10m 47s\n- puppet-openstack-unit-7.29-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c0311d53ee1240eea45dd89fe3f4b2ac : SUCCESS in 5m 36s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a208639fa2f643ed8d155d3738a19de1 : SUCCESS in 53m 30s\n- puppet-openstack-litmus-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/cb1a51dbaf0c4f938bfab04ec2872114 : SUCCESS in 21m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/78412356b51e43ca99073b92dfe7ff89 : SUCCESS in 3m 36s","accounts_in_message":[],"_revision_number":2},{"id":"d0a8787adb00bb200bba2d0f08204f13c57e4efd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-25 16:22:13.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"145e1a91e0d1bc35070dc2e7a3e375c8916d5170","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-25 16:24:04.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/81d2ebc8b729494099b2f694115d25ee\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7884677bf061440fb8af34e5b39734d3 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"22ef63f8f957ba3adcaada6cc41f5a8f49858ee7","revisions":{"fd799980f04f332a1cfc59c4416222ef773f6301":{"kind":"REWORK","_number":1,"created":"2024-03-25 04:06:59.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/52/914052/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-trove","ref":"refs/changes/52/914052/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/52/914052/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/52/914052/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/52/914052/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-trove refs/changes/52/914052/1"}}},"commit":{"parents":[{"commit":"225592993c634b156af54d755bda41b9d8c132cc","subject":"Merge \"Expose rabbit_transient_quorum_queue\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/225592993c634b156af54d755bda41b9d8c132cc"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-03-25 04:05:54.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-03-25 04:05:54.000000000","tz":540},"subject":"Prepare to rename trove::api::service_credentials","message":"Prepare to rename trove::api::service_credentials\n\nThis class is used by all services but is still kept under the api\ndirectory.\n\nChange-Id: I4016752d94da35ea257f9c76079a33797151640b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/fd799980f04f332a1cfc59c4416222ef773f6301"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/fd799980f04f332a1cfc59c4416222ef773f6301"}]},"branch":"refs/heads/master"},"22ef63f8f957ba3adcaada6cc41f5a8f49858ee7":{"kind":"REWORK","_number":2,"created":"2024-03-25 05:11:14.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/52/914052/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-trove","ref":"refs/changes/52/914052/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/52/914052/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/52/914052/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/52/914052/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-trove refs/changes/52/914052/2"}}},"commit":{"parents":[{"commit":"225592993c634b156af54d755bda41b9d8c132cc","subject":"Merge \"Expose rabbit_transient_quorum_queue\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/225592993c634b156af54d755bda41b9d8c132cc"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-03-25 04:05:54.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-03-25 05:10:57.000000000","tz":540},"subject":"Prepare to rename trove::api::service_credentials","message":"Prepare to rename trove::api::service_credentials\n\nThis class is used by all services but is still kept under the api\ndirectory.\n\nChange-Id: I4016752d94da35ea257f9c76079a33797151640b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/22ef63f8f957ba3adcaada6cc41f5a8f49858ee7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/22ef63f8f957ba3adcaada6cc41f5a8f49858ee7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
