)]}'
{"id":"openstack%2Fpuppet-trove~942043","triplet_id":"openstack%2Fpuppet-trove~master~If524b5e73a48a5fb25ac44efdd7754720581eadb","project":"openstack/puppet-trove","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-02-19 16:19:30.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If524b5e73a48a5fb25ac44efdd7754720581eadb","subject":"Deprecate support for [oslo_messaging_rabbit] heartbeat_in_pthread","status":"MERGED","created":"2025-02-18 02:04:45.000000000","updated":"2025-02-19 16:20:58.000000000","submitted":"2025-02-19 16:19:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"942043","meta_rev_id":"1708ad6178b56601fbac857bfa6b3f0fce272256","_number":942043,"virtual_id_number":942043,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-02-19 16:19:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2025-02-18 15:10:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2025-02-19 14:18:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2025-02-19 14:18:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-18 05:11:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-02-18 15:10:01.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2025-02-19 14:18:22.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"ae6409ac6baeff1d39dc9f0510f81e5d3c613d2a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-02-18 02:04:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"954a721e304dd04fcde565377c6276a36637469a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-02-18 02:35:24.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"312598a6c95d8e971cffddcc80884412bb72f295","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-02-18 02:57:41.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"5d7ad3d33289680411d35617d503b561e7d63e9e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-18 05:11:35.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/16d04cacba7f417fa510aee5c1c61940\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/74bf6176507f4e67a6b9a2a291264cb7 : SUCCESS in 2m 47s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/53463168f2ff41a393635ba5b58b1c57 : SUCCESS in 2m 44s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/b6619452ac094058941ee14c18ac275d : SUCCESS in 3m 21s (non-voting)\n- puppet-openstack-unit-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/87bd198866494f4eb81f22ef0f417c75 : SUCCESS in 11m 07s\n- puppet-openstack-unit-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c5b67a28a68946baa062042f0a70ae7f : SUCCESS in 6m 08s\n- puppet-openstack-unit-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/180038697c1a48d0b15054bc2f26865b : SUCCESS in 8m 20s\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fae822a3ce5b4e038434cfc16a3ae0c7 : SUCCESS in 5m 56s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/237559adad554adba57c495551d92fc1 : SUCCESS in 44m 20s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/778075dd2d424b4eb8b486da88868478 : SUCCESS in 1h 16m 34s\n- puppet-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/fa6cfe0d75f74072bf5b9d775a828064 : SUCCESS in 1h 10m 34s\n- puppet-openstack-litmus-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/b4b3a136568142a19a0e0ab0e48f8286 : SUCCESS in 45m 50s\n- puppet-openstack-litmus-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1159c2d8629045d8912c6472fd8ea740 : FAILURE in 36m 17s (non-voting)\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/47c475221e3641d2829e94bab37b1f0c : SUCCESS in 40m 17s\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e48fa05927014b5aa35eb9cca8b6f89f : FAILURE in 40m 30s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/62f5acc970a948a2b78db696f7a8058c : SUCCESS in 2m 15s","accounts_in_message":[],"_revision_number":3},{"id":"cba4447e9178bb40375e48e04b2b25e809b423d2","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2025-02-18 15:10:01.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"77a9eafb5a030ed2a56eb623b4b7bf63f3bf149a","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-02-19 14:18:22.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"e9a0cb156ceb2c9c7eeba6a301f15d50831c6d8a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-19 14:19:39.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"af64895a0867459b4b7d018adf09fe04671408be","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-19 16:19:30.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b82b68bfdbac42ceb9dbefa92d066bd4\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9d93f674a09f466994b3e69231ce7bf7 : SUCCESS in 2m 54s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c55804fbd56a46c8af268e0c220922de : SUCCESS in 2m 51s\n- puppet-openstack-unit-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/c7513a7709ed4dafa232fae43eb2b3fb : SUCCESS in 16m 16s\n- puppet-openstack-unit-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/10fa0dde79764e6786a7a7e693f896bf : SUCCESS in 10m 23s\n- puppet-openstack-unit-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/216ba25779fa4ab7af24a204ec8fa70a : SUCCESS in 9m 31s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/0401d5f034004c30b1d549fa6c631486 : SUCCESS in 1h 15m 34s\n- puppet-openstack-litmus-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/565a6236b13842cbaa1c12727a829561 : SUCCESS in 44m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/873b47a286504bd1b2b8aa5dba022b71 : SUCCESS in 2m 44s","accounts_in_message":[],"_revision_number":3},{"id":"9c8cb369f551f650cb3e696774570d70ceaf5773","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-19 16:19:30.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"1708ad6178b56601fbac857bfa6b3f0fce272256","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-19 16:20:58.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/24487f665e5a47439605905c3f7975aa\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c97046ab41ed4bc7afac8035ecba2cc0 : SUCCESS in 52s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"16575e8ea4cab7e4fa49938cff4b69fc09a2f902","revisions":{"0edf009282a0ee2cbb8a8a276dfb365e070ea65f":{"kind":"REWORK","_number":1,"created":"2025-02-18 02:04:45.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/43/942043/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-trove","ref":"refs/changes/43/942043/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/43/942043/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/43/942043/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/43/942043/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-trove refs/changes/43/942043/1"}}},"commit":{"parents":[{"commit":"ddf51fbc41b04ffc00de5f7a0ea606629ea26800","subject":"Merge \"Mark puppet 8 supported\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/ddf51fbc41b04ffc00de5f7a0ea606629ea26800"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-02-18 02:04:32.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-02-18 02:04:32.000000000","tz":540},"subject":"Deprecate support for rabbit_heartbeat_in_pthread","message":"Deprecate support for rabbit_heartbeat_in_pthread\n\n... because the option has been deprecated.\n\nDepends-on: https://review.opendev.org/925778\nChange-Id: If524b5e73a48a5fb25ac44efdd7754720581eadb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/0edf009282a0ee2cbb8a8a276dfb365e070ea65f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/0edf009282a0ee2cbb8a8a276dfb365e070ea65f"}]},"branch":"refs/heads/master"},"9ec597b958b48fb2331ec85e99b3af6d0aba0d88":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2025-02-18 02:35:24.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/43/942043/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-trove","ref":"refs/changes/43/942043/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/43/942043/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/43/942043/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/43/942043/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-trove refs/changes/43/942043/2"}}},"commit":{"parents":[{"commit":"ddf51fbc41b04ffc00de5f7a0ea606629ea26800","subject":"Merge \"Mark puppet 8 supported\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/ddf51fbc41b04ffc00de5f7a0ea606629ea26800"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-02-18 02:04:32.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-02-18 02:35:14.000000000","tz":540},"subject":"Deprecate support for rabbit_heartbeat_in_pthread","message":"Deprecate support for rabbit_heartbeat_in_pthread\n\n... because the option has been deprecated.\n\nNote that the deprecation warning may be added in puppet-oslo so is not\nimplemented in individual module.\n\nDepends-on: https://review.opendev.org/925778\nChange-Id: If524b5e73a48a5fb25ac44efdd7754720581eadb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/9ec597b958b48fb2331ec85e99b3af6d0aba0d88"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/9ec597b958b48fb2331ec85e99b3af6d0aba0d88"}]},"branch":"refs/heads/master"},"16575e8ea4cab7e4fa49938cff4b69fc09a2f902":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2025-02-18 02:57:41.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/43/942043/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-trove","ref":"refs/changes/43/942043/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/43/942043/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/43/942043/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/43/942043/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-trove refs/changes/43/942043/3"}}},"commit":{"parents":[{"commit":"ddf51fbc41b04ffc00de5f7a0ea606629ea26800","subject":"Merge \"Mark puppet 8 supported\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/ddf51fbc41b04ffc00de5f7a0ea606629ea26800"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-02-18 02:04:32.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-02-18 02:57:29.000000000","tz":540},"subject":"Deprecate support for [oslo_messaging_rabbit] heartbeat_in_pthread","message":"Deprecate support for [oslo_messaging_rabbit] heartbeat_in_pthread\n\n... because the option has been deprecated.\n\nNote that the deprecation warning may be added in puppet-oslo so is not\nimplemented in individual module.\n\nDepends-on: https://review.opendev.org/925778\nChange-Id: If524b5e73a48a5fb25ac44efdd7754720581eadb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/16575e8ea4cab7e4fa49938cff4b69fc09a2f902"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/16575e8ea4cab7e4fa49938cff4b69fc09a2f902"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
