)]}'
{"id":"openstack%2Fpuppet-trove~944850","triplet_id":"openstack%2Fpuppet-trove~master~I1d56b5af3f975012ab371f331643a73d5c84ace0","project":"openstack/puppet-trove","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-03-26 12:18:06.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},"14985":{"account":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"last_update":"2025-03-26 22:42:18.000000000","reason":"removed on reply"},"6476":{"account":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"last_update":"2025-03-26 23:09:44.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I1d56b5af3f975012ab371f331643a73d5c84ace0","subject":"Add a rabbit_transient_queues_ttl and amqp_auto_delete params","status":"MERGED","created":"2025-03-18 09:53:26.000000000","updated":"2025-03-26 23:10:42.000000000","submitted":"2025-03-26 23:09:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"944850","meta_rev_id":"9eb5e2263bdb39c632fa8d1fa38d5ab882f744b4","_number":944850,"virtual_id_number":944850,"owner":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-03-26 23:09:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2025-03-26 12:18:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2025-03-26 22:42:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":1,"date":"2025-03-26 22:42:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-03-18 11:00:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-03-19 13:21:16.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2025-03-21 12:22:33.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"}],"messages":[{"id":"f91193ab8259c6b4d4833d902723f0717ab7daf1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2025-03-18 09:53:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3649c05e1f84f11125e906a9e2b8a6c8fe1d0a86","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-18 11:00:33.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a5110d654eb34611bf094fe901d12176\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b37291322a9c421bb11f629191c3fed6 : FAILURE in 1m 27s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d547e937cc5f43c1b9fae083a7ac8e66 : SUCCESS in 1m 54s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/e3c68b98e24545488656c1ff437aa81c : SUCCESS in 3m 01s (non-voting)\n- puppet-openstack-unit-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/4bf348ef6aa045109790f23587a02f7a : SUCCESS in 8m 48s\n- puppet-openstack-unit-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/69a4c1bb690f4ff39e021b33eb8d6cd7 : SUCCESS in 5m 15s\n- puppet-openstack-unit-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/a4fa9008fe294f398aa85c7edbd453bd : SUCCESS in 8m 23s\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9b784abf5f2b4b2395dca05e7f6a0ddf : SUCCESS in 5m 31s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/91d1f8fb6056468280abef8a7df622db : FAILURE in 25m 26s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/df20cbe7669a4e31b5ab7ab85044bd9a : SUCCESS in 1h 01m 16s\n- puppet-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/092a0aa2d266431ebf16bb5c62a3fe7b : SUCCESS in 45m 59s\n- puppet-openstack-litmus-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/951f3e717f184872af341cd291181616 : SUCCESS in 29m 01s\n- puppet-openstack-litmus-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/82c864df1307418784575105b6e9f1f1 : FAILURE in 37m 15s (non-voting)\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/2104512cd7484e559f7c18dfea53effe : SUCCESS in 27m 32s\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b26dc33a12764978a336ee5e3ee41aba : FAILURE in 36m 42s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1ce9efbc12f44077978c165db6a01c4b : SUCCESS in 2m 19s","accounts_in_message":[],"_revision_number":1},{"id":"5a85870752fb9b590daf30125bc06398150d54ea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2025-03-19 07:22:36.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"13cbcb755a098f92f5e432181866371f393561dd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-19 08:12:00.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ae4ccc2ffb5a4e30abe2c1e1d810fd98\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a6abfbcb5ad74d3b94ef215b8b1ec094 : SUCCESS in 5m 11s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3bfdd45e9dee48a08f47149108d94402 : SUCCESS in 3m 32s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/636851f0b46c4cfab1e9cc0dd3c462c1 : SUCCESS in 5m 22s (non-voting)\n- puppet-openstack-unit-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/ef91dcf7cf7a4462832fbf6652349799 : SUCCESS in 4m 33s\n- puppet-openstack-unit-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5e032d52c51846be8c061f0de76b9a57 : SUCCESS in 2m 40s\n- puppet-openstack-unit-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/56c8bdefe14d48e1a9ddbdecce0aab91 : SUCCESS in 6m 39s\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8f80cb0ff03e4585ba284c21531092c8 : SUCCESS in 3m 06s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/44eaaf6ca75e401c826036b4b533c3bc : SUCCESS in 42m 43s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/1ce5f6e6f27d4db490de0d9dc34d4085 : SUCCESS in 26m 25s\n- puppet-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/fdbe9ce7943b4e80b8e1c4b7cf64b11d : SUCCESS in 48m 01s\n- puppet-openstack-litmus-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/68ddc27e72f4407cb944df5923b76180 : SUCCESS in 27m 34s\n- puppet-openstack-litmus-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2fef335bf2c547af995f048238cbbe4f : FAILURE in 37m 53s (non-voting)\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/22c8bdc9451c43e78b319e22b8ef5041 : SUCCESS in 25m 30s\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f5e429fbc90f4104a5f1f2fe32e88e63 : FAILURE in 38m 00s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6d62ee966f5845aca317cb8219460d5b : SUCCESS in 2m 28s","accounts_in_message":[],"_revision_number":2},{"id":"42429585c8cfed78540ea9849d11148fc40ecef5","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2025-03-19 13:21:16.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"a3e2de746a58374b4dea24d847799da321c806f9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2025-03-21 09:20:06.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"42e750cc17a7475ea70575b4937294881069beda","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-21 10:34:15.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/156f260f3f0a4a758ef76df05389e2ee\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/12e707d8a1374bd0bde67a1ec35e6396 : SUCCESS in 3m 31s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1f2513e50e4b4541afacf31170a7b8c7 : SUCCESS in 2m 58s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/f48c4cc4bbe44eae8edb1dc437c00a56 : SUCCESS in 4m 45s (non-voting)\n- puppet-openstack-unit-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/21c76e25b7aa4aff8e28a575d6129917 : SUCCESS in 7m 11s\n- puppet-openstack-unit-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/625ae23341db46f1aa2cd4c8e6f4eb34 : SUCCESS in 5m 06s\n- puppet-openstack-unit-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/cd7c5ad771c14c3d8ac55047203d911a : SUCCESS in 8m 55s\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9657d3b253e2444cabd59fbe01176cfd : SUCCESS in 5m 21s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/052614bce25c4d9495cface73edeecf2 : FAILURE in 26m 32s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/dfc761dc82f3409ab7e519a8884c6568 : SUCCESS in 1h 05m 49s\n- puppet-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/94b1c9cb6e424291848880fe362e9032 : SUCCESS in 55m 52s\n- puppet-openstack-litmus-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/e930b2d4e73544c283545c1304aa24e1 : SUCCESS in 25m 06s\n- puppet-openstack-litmus-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/460eb61b2e154572b0969d37ea166b2d : FAILURE in 39m 42s (non-voting)\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/012b63cb88604e558f8f0091a712a88d : SUCCESS in 20m 34s\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2214f79e6e7e454d8ddde599a98b6a13 : FAILURE in 36m 49s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6e0238384a7d4ae5b51104a9b7deae10 : SUCCESS in 3m 06s","accounts_in_message":[],"_revision_number":3},{"id":"522c12c7ac331f8c619750d4c9ecf748f477b225","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-03-21 12:22:33.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"941f35d4d604deb44fa01722080c852b79cb1cbd","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2025-03-26 09:38:33.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"d49fae7f6bb5431e26009e9b46fe063fdb3f1d80","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2025-03-26 09:38:40.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"19a0a73ea402e9d3cc5ccdd5ead329d4c90ea009","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-26 10:28:39.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f4b3c1ebc1d94c07b2355b6f053926ab\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b8c946a7b8c6483ebf8a5b42b86de0d0 : SUCCESS in 2m 51s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ff552fcf02894974a444c9b970f85e26 : SUCCESS in 2m 28s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/ae1b1012144a4bbcb0a932903aa8f4f7 : SUCCESS in 3m 40s (non-voting)\n- puppet-openstack-unit-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/1a1ac3d589e94a3b8bd3d081c16e629b : SUCCESS in 5m 27s\n- puppet-openstack-unit-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9738e710f6e74f80b3879e6c923b8cd7 : SUCCESS in 3m 32s\n- puppet-openstack-unit-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/2de88ba704e2423fb1f9437eae350450 : SUCCESS in 4m 50s\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ba679bad55494148833e386a5787990b : SUCCESS in 4m 06s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d526483fe5b54cd988bf1a58c0a52083 : SUCCESS in 38m 55s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/403480926d4b44c1a1dac8179c08c863 : SUCCESS in 48m 47s\n- puppet-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/e4186153171d4de48c86aadc91c934e3 : SUCCESS in 42m 52s\n- puppet-openstack-litmus-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/97aae7a6aaf544958568ad7a225e9339 : SUCCESS in 18m 14s\n- puppet-openstack-litmus-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/49ee7bf8967f4428ac03a367f9f7f080 : FAILURE in 39m 42s (non-voting)\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/9b29687efcca4789a329259ab1de1b2c : SUCCESS in 22m 07s\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4ac1b8084dcd46c8b12c5d0fe5ed33f3 : FAILURE in 36m 52s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ec1191fa5818496bb64ae62246af6dc4 : SUCCESS in 3m 15s","accounts_in_message":[],"_revision_number":4},{"id":"2cfc68f8eb3f601ed09cc16f51076e207aaea006","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-03-26 12:18:06.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"77c340ac867b3a0130ebd81de39af15c9e0c4e47","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2025-03-26 22:42:18.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"0a50c1070680a44274a65c85d4f6275cc5b14d95","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-26 22:43:11.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"daff689e1a0e7b5af7885b5a30412f5be16a3919","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-26 23:09:44.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0dfd7e7bc3c24bac95594b7d31d6527c\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9e0cb57dea874d0dbe4b571da78a60a4 : SUCCESS in 2m 13s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cf4e7be7a52048ada5ca4578e0095fb5 : SUCCESS in 2m 13s\n- puppet-openstack-unit-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/85ef95497e5448aab22b3d64a703e356 : SUCCESS in 6m 17s\n- puppet-openstack-unit-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/40f9e7f8f29542e3bb0623a6c72094e6 : SUCCESS in 4m 00s\n- puppet-openstack-unit-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/eb0d7c84c7ae4c188c749cb7810d619f : SUCCESS in 6m 07s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c2291b379d204bc8bf8efb8aade9a922 : SUCCESS in 26m 05s\n- puppet-openstack-litmus-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/03401e35bd8146ca8aa963eaafebb7d8 : SUCCESS in 22m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/03136f98626e4715b35fa1d391404f07 : SUCCESS in 2m 11s","accounts_in_message":[],"_revision_number":4},{"id":"8ba61123df69b4a635f9263b58700806b28182db","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-26 23:09:44.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"9eb5e2263bdb39c632fa8d1fa38d5ab882f744b4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-26 23:10:42.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/13a0e0c64f4047f6bdb104c6ead5410a\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0e68e25cb797478bbfce9799d105d6e7 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"f51ac5cc4b3fe1f60dea15568cc9bc5253d01ba0","revisions":{"d46f79d64b000568431c0f3877e2c41d3017ff59":{"kind":"REWORK","_number":1,"created":"2025-03-18 09:53:26.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/50/944850/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-trove","ref":"refs/changes/50/944850/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/50/944850/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/50/944850/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/50/944850/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-trove refs/changes/50/944850/1"}}},"commit":{"parents":[{"commit":"5e074680dfb856f4e7fb814a8999a69d49771e11","subject":"Fix assertion of heartbeat_in_pthread parameter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/5e074680dfb856f4e7fb814a8999a69d49771e11"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2025-03-18 09:52:38.000000000","tz":60},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2025-03-18 09:52:38.000000000","tz":60},"subject":"Add a rabbit_transient_queues_ttl and amqp_auto_delete params","message":"Add a rabbit_transient_queues_ttl and amqp_auto_delete params\n\nSince transient_queues_ttl is also managed by oslo::messaging::rabbit,\nthere\u0027s no way to use trove_config to set it, so this patch is mandatory\nfor one to set rabbit_transient_queues_ttl in trove.conf.\n\nChange-Id: I1d56b5af3f975012ab371f331643a73d5c84ace0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/d46f79d64b000568431c0f3877e2c41d3017ff59"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/d46f79d64b000568431c0f3877e2c41d3017ff59"}]},"branch":"refs/heads/master"},"9eda10c439a21002d49bc0a53cb1626333ca9d93":{"kind":"REWORK","_number":2,"created":"2025-03-19 07:22:36.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/50/944850/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-trove","ref":"refs/changes/50/944850/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/50/944850/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/50/944850/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/50/944850/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-trove refs/changes/50/944850/2"}}},"commit":{"parents":[{"commit":"5e074680dfb856f4e7fb814a8999a69d49771e11","subject":"Fix assertion of heartbeat_in_pthread parameter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/5e074680dfb856f4e7fb814a8999a69d49771e11"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2025-03-18 09:52:38.000000000","tz":60},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2025-03-19 07:22:28.000000000","tz":60},"subject":"Add a rabbit_transient_queues_ttl and amqp_auto_delete params","message":"Add a rabbit_transient_queues_ttl and amqp_auto_delete params\n\nSince transient_queues_ttl is also managed by oslo::messaging::rabbit,\nthere\u0027s no way to use trove_config to set it, so this patch is mandatory\nfor one to set rabbit_transient_queues_ttl in trove.conf.\n\nChange-Id: I1d56b5af3f975012ab371f331643a73d5c84ace0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/9eda10c439a21002d49bc0a53cb1626333ca9d93"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/9eda10c439a21002d49bc0a53cb1626333ca9d93"}]},"branch":"refs/heads/master"},"fc134ad25f1fb618a9d82ef6c140b382cb733c34":{"kind":"REWORK","_number":3,"created":"2025-03-21 09:20:06.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/50/944850/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-trove","ref":"refs/changes/50/944850/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/50/944850/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/50/944850/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/50/944850/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-trove refs/changes/50/944850/3"}}},"commit":{"parents":[{"commit":"f1fd684d812ff9a1824950e6b2c6e9136d2fe039","subject":"Fix outdated example","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/f1fd684d812ff9a1824950e6b2c6e9136d2fe039"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2025-03-18 09:52:38.000000000","tz":60},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2025-03-21 09:19:52.000000000","tz":60},"subject":"Add a rabbit_transient_queues_ttl and amqp_auto_delete params","message":"Add a rabbit_transient_queues_ttl and amqp_auto_delete params\n\nSince transient_queues_ttl is also managed by oslo::messaging::rabbit,\nthere\u0027s no way to use trove_config to set it, so this patch is mandatory\nfor one to set rabbit_transient_queues_ttl in trove.conf.\n\nChange-Id: I1d56b5af3f975012ab371f331643a73d5c84ace0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/fc134ad25f1fb618a9d82ef6c140b382cb733c34"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/fc134ad25f1fb618a9d82ef6c140b382cb733c34"}]},"branch":"refs/heads/master"},"f51ac5cc4b3fe1f60dea15568cc9bc5253d01ba0":{"kind":"REWORK","_number":4,"created":"2025-03-26 09:38:40.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/50/944850/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-trove","ref":"refs/changes/50/944850/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/50/944850/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/50/944850/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/50/944850/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-trove refs/changes/50/944850/4"}}},"commit":{"parents":[{"commit":"f1fd684d812ff9a1824950e6b2c6e9136d2fe039","subject":"Fix outdated example","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/f1fd684d812ff9a1824950e6b2c6e9136d2fe039"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2025-03-18 09:52:38.000000000","tz":60},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2025-03-26 09:38:18.000000000","tz":60},"subject":"Add a rabbit_transient_queues_ttl and amqp_auto_delete params","message":"Add a rabbit_transient_queues_ttl and amqp_auto_delete params\n\nSince transient_queues_ttl is also managed by oslo::messaging::rabbit,\nthere\u0027s no way to use trove_config to set it, so this patch is mandatory\nfor one to set rabbit_transient_queues_ttl in trove.conf.\n\nChange-Id: I1d56b5af3f975012ab371f331643a73d5c84ace0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/f51ac5cc4b3fe1f60dea15568cc9bc5253d01ba0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/f51ac5cc4b3fe1f60dea15568cc9bc5253d01ba0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
