)]}'
{"id":"openstack%2Fpuppet-trove~992635","triplet_id":"openstack%2Fpuppet-trove~master~Iaf706e7ee75f1cc7d320d3a828fbb16711935754","project":"openstack/puppet-trove","branch":"master","topic":"debian","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-06-11 08:00:07.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},"14985":{"account":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"last_update":"2026-06-23 11:47:19.000000000","reason":"Change was submitted"},"6476":{"account":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"last_update":"2026-06-23 08:11:48.000000000","reason":"\u003cGERRIT_ACCOUNT_6476\u003e replied on the change","reason_account":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"}}},"hashtags":[],"change_id":"Iaf706e7ee75f1cc7d320d3a828fbb16711935754","subject":"Debian uwsgi: correct scheduling.","status":"MERGED","created":"2026-06-10 12:38:27.000000000","updated":"2026-06-23 11:47:19.000000000","submitted":"2026-06-23 11:47:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"992635-debian","meta_rev_id":"99f520165f9b56e33683a03175fa8c0f58163536","_number":992635,"virtual_id_number":992635,"owner":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-06-23 11:47:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2026-06-11 08:00:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2026-06-22 09:04:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-10 14:12:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-11 01:01:39.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2026-06-22 08:00:50.000000000","updated_by":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"82f64051f6cafd95219fbace98c8477cbfa0e6fd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2026-06-10 12:38:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"15b8006f5e158d239045ec09b99b83f0a4d7d728","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-10 14:12:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/829a0b8ddbc2471da904df6f54de7aab\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d1142de5b6b74f0f918cf93cb82f564d : SUCCESS in 3m 16s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1381c944f62e4b74a4b3787b1a43d6a1 : SUCCESS in 3m 00s\n- puppet-openstack-unit-8.10-centos-10-stream https://zuul.opendev.org/t/openstack/build/deb86f8ae0754e2695b7b04322eefdf0 : SUCCESS in 7m 30s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2ffee6c254664a48ba1e6799e9df0130 : SUCCESS in 6m 20s\n- puppet-openstack-unit-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b376954fd5614db389facb3c049dbdd7 : SUCCESS in 6m 15s (non-voting)\n- puppet-openstack-integration-8-scenario003-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/47c2d019d2204cca94cc29239b83f5b5 : FAILURE in 1h 21m 08s (non-voting)\n- puppet-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/d59278f9d0f74b7cab6f8a0391aa202f : SUCCESS in 34m 52s\n- openvox-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/12b869fb8a7343688ca8c4633cb9cf46 : SUCCESS in 1h 01m 20s (non-voting)\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/2ab2cce0a7a8470593e6d0bec90729d2 : SUCCESS in 21m 00s\n- puppet-openstack-litmus-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dbebb223dd1649b4a53c986163ad880f : FAILURE in 39m 54s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d9cb2edde4f9422f95cc3d1bb6552636 : FAILURE in 41m 23s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7c0674df55b04155aba579abeb0ae025 : SUCCESS in 4m 39s","accounts_in_message":[],"_revision_number":1},{"id":"89f520c1ad0d7a0cf057b0864689afdf572b8687","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-06-11 01:01:39.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a136a223bf2ae399eb979bf4a8a7a1ea82645896","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2026-06-11 07:17:01.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f5ed623706053f0071e88f9faf9d034e830cbffb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2026-06-11 07:17:12.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"dc5b0a032aa83fb03ca012e7c2532b9d92e03986","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-06-11 08:00:07.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"3a4b36e22b6bd0bd7f11827e5b30bb050b85ed03","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-11 08:41:49.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c4cb9e91563a42fa948687d8ca4996fa\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/26996b76bb684182b03055f5edb58c7d : SUCCESS in 3m 10s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e5ef632bff964b7d8fd06a9fce55ab08 : SUCCESS in 2m 57s\n- puppet-openstack-unit-8.10-centos-10-stream https://zuul.opendev.org/t/openstack/build/f0a700b4eb614183b2b79c2f0b3567cb : SUCCESS in 4m 17s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6706c39b02e949e5b5ed563b1743cd14 : SUCCESS in 5m 51s\n- puppet-openstack-unit-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/35e57d35974342b480c6dd8f286769f4 : SUCCESS in 5m 50s (non-voting)\n- puppet-openstack-integration-8-scenario003-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6e8e0140ef074b4ab19c413ba88d8239 : FAILURE in 1h 18m 32s (non-voting)\n- puppet-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/795e3b93ebbf408197a2b1e1b8a7141a : SUCCESS in 48m 02s\n- openvox-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/43e861d5ff504be1832af095c5692999 : SUCCESS in 59m 37s (non-voting)\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/8072ac48e73b4f4e87244616fd71a599 : SUCCESS in 27m 14s\n- puppet-openstack-litmus-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5a1afdbfe87940d4bf5a5bc966f21576 : FAILURE in 39m 15s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e419cb940d5c401d8afa4c98ef4ca9a8 : FAILURE in 38m 21s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0f33f9f39b274977854a866352aa88d4 : SUCCESS in 3m 12s","accounts_in_message":[],"_revision_number":2},{"id":"3efa6f5f1bd2c69462336c5d3c3297684a95658a","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-06-22 09:04:00.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"69cd2968296e14fb3024abc4bfa2276d03cf7eed","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-22 09:06:38.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"47e219f794e53c624c63d7867904667ee50e77ed","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-22 09:38:53.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e784232aad94469b9d6908e0c56782b0\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5bceca69e7db42b4a897e2aa526876c2 : SUCCESS in 3m 22s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/54079729704c4aa4bac67a62b468b134 : SUCCESS in 3m 47s\n- puppet-openstack-unit-8.10-centos-10-stream https://zuul.opendev.org/t/openstack/build/67c58529aee143d38243dea49a00eeb2 : SUCCESS in 5m 49s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ca648086ad9740c4a2857853bf224772 : SUCCESS in 7m 36s\n- puppet-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/ec21cbd3ad224597866e72c1c6c396fb : FAILURE in 7m 40s\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/01959bd40b9545a393eea43672e2023c : SUCCESS in 27m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3c5896c1487b4ad4a48623cf2f00b450 : SUCCESS in 3m 00s","accounts_in_message":[],"_revision_number":2},{"id":"a77067e6776fe03f4938db39f0da905b1eff857e","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2026-06-23 08:11:48.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7d9e0db466486758d12ea4d098c1879f2ab37aca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-23 10:39:42.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0d9fa75993f34c008ced5173ed67ade1\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/31724cd823cf4075b15d71e6f25e3a9d : SUCCESS in 1m 49s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e04a420f5cda46c480a914dcaa253a59 : SUCCESS in 3m 08s\n- puppet-openstack-unit-8.10-centos-10-stream https://zuul.opendev.org/t/openstack/build/9bdb44911f84489580bedef77b84a61b : SUCCESS in 5m 10s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/52fa441db3b24e5bb679245c35d615cb : SUCCESS in 5m 04s\n- puppet-openstack-unit-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a4336183b2654530aec32e6cd58cf60a : SUCCESS in 5m 49s (non-voting)\n- puppet-openstack-integration-8-scenario003-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6ba17f216274448abb5b00c1d52a3a62 : FAILURE in 1h 14m 47s (non-voting)\n- puppet-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a11c9941ca974c2b8ace30346ee9a4d8 : SUCCESS in 1h 00m 58s\n- openvox-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/9d310ebd38ca402a9e3caace6575e3dc : SUCCESS in 59m 36s (non-voting)\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/aca157461ca34f6f8f553b40cec13ee4 : SUCCESS in 27m 14s\n- puppet-openstack-litmus-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2bdb65e00366452e9ccf65fb4d476f80 : FAILURE in 40m 27s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/34c6f17608fd4cda8ece9cc2ee3e9ea1 : FAILURE in 40m 23s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c39e2403102f47ff9cd041780573030c : SUCCESS in 4m 21s","accounts_in_message":[],"_revision_number":2},{"id":"f455cfae07b179b3cbbe74fa0d010c33cb8c60c0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-23 10:40:36.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"72ece9805181bdb2f2490729f8413aaedbe9f91e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-23 11:47:19.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/34dd407c4c0c480695a7ba54761bfa70\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ea3ea48c86314f25a41ad2f087454ff3 : SUCCESS in 2m 59s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/af741e34e3d944c69be1b3ccea4f1e9a : SUCCESS in 3m 00s\n- puppet-openstack-unit-8.10-centos-10-stream https://zuul.opendev.org/t/openstack/build/e762b8ceb4b74d1a989e9e6664db0950 : SUCCESS in 5m 29s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c281506400764c25b302294e40c13fa8 : SUCCESS in 8m 15s\n- puppet-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/4fd7635f70fd4e6e99f9af368279924d : SUCCESS in 48m 36s\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/710845cab2b54016b99d40de34fad6cd : SUCCESS in 26m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f55ac12f2b82414bb58085a657868b38 : SUCCESS in 3m 40s","accounts_in_message":[],"_revision_number":2},{"id":"99f520165f9b56e33683a03175fa8c0f58163536","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-23 11:47:19.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"f30b4c2089818abd006793608ae772f58007ef45","revisions":{"9dd46b6533e3c4930457a2c1105e3634bbd302aa":{"kind":"REWORK","_number":1,"created":"2026-06-10 12:38:27.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/35/992635/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-trove","ref":"refs/changes/35/992635/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/35/992635/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/35/992635/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/35/992635/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-trove refs/changes/35/992635/1"}}},"commit":{"parents":[{"commit":"04e7754442e06ac47ef81e11c1cc2da31ba56ab4","subject":"Add Debian uwsgi support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/04e7754442e06ac47ef81e11c1cc2da31ba56ab4"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2026-06-10 12:37:00.000000000","tz":120},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2026-06-10 12:38:16.000000000","tz":120},"subject":"Debian uwsgi: correct scheduling.","message":"Debian uwsgi: correct scheduling.\n\nSigned-off-by: Thomas Goirand \u003czigo@debian.org\u003e\nChange-Id: Iaf706e7ee75f1cc7d320d3a828fbb16711935754\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/9dd46b6533e3c4930457a2c1105e3634bbd302aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/9dd46b6533e3c4930457a2c1105e3634bbd302aa"}]},"branch":"refs/heads/master"},"f30b4c2089818abd006793608ae772f58007ef45":{"kind":"REWORK","_number":2,"created":"2026-06-11 07:17:12.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/35/992635/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-trove","ref":"refs/changes/35/992635/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/35/992635/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/35/992635/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-trove refs/changes/35/992635/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-trove refs/changes/35/992635/2"}}},"commit":{"parents":[{"commit":"04e7754442e06ac47ef81e11c1cc2da31ba56ab4","subject":"Add Debian uwsgi support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/04e7754442e06ac47ef81e11c1cc2da31ba56ab4"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2026-06-10 12:37:00.000000000","tz":120},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2026-06-11 07:17:04.000000000","tz":120},"subject":"Debian uwsgi: correct scheduling.","message":"Debian uwsgi: correct scheduling.\n\nSigned-off-by: Thomas Goirand \u003czigo@debian.org\u003e\nChange-Id: Iaf706e7ee75f1cc7d320d3a828fbb16711935754\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/f30b4c2089818abd006793608ae772f58007ef45"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-trove/commit/f30b4c2089818abd006793608ae772f58007ef45"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
